)]}'
{"id":"openstack%2Frequirements~920248","triplet_id":"openstack%2Frequirements~master~I8be43e13460de01e67a8df862e57bda35a728f48","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8be43e13460de01e67a8df862e57bda35a728f48","subject":"update constraint for ceilometermiddleware to new release 3.4.0","status":"MERGED","created":"2024-05-23 08:05:36.000000000","updated":"2024-05-23 19:11:18.000000000","submitted":"2024-05-23 19:11:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920248-new-release","meta_rev_id":"006042a7f3de7f84a2d11153e9e2550fbe7eb108","_number":920248,"virtual_id_number":920248,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-23 19:11:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-05-23 12:11:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-05-23 15:41:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-05-23 15:41:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-23 09:36:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-23 12:11:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-05-23 15:41:42.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"5e2fbf7043f0a3f9412b46047923e0846648cc71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-05-23 08:05:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5324e8141de063f4375094fbfb5d99dcebae47d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 09:36:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4cfa6205c4a4a31ab58f857f5053d78\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b2861937a9b4a3ea2908c06379db892 : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9296a6fa089d485c9fc5bf2b8876e2b6 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e741f5ef94844bbdb856853c512a5999 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/835d2ae12b4f4caa84d61c36e028b3b5 : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8015c93426564f1d956b57bdacd75e78 : FAILURE in 5m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/269df57d310d446eba488c1a40b30975 : SUCCESS in 2m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1625cea6dd954ab2a00926ef09b87531 : SUCCESS in 3m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/29eb49ac11ce4968beac3f3317c57f3f : SUCCESS in 8m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1b1552a7dace42b4ade7f910fcffe766 : SUCCESS in 21m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/375abf2638444c1986a5fe0c5b37deac : SUCCESS in 10m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f27eb198c606458aaabdccd929029c19 : SUCCESS in 15m 26s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/a5c9b3b9e4d54617aa9cda8eac2031a7 : SUCCESS in 5m 17s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/976797d6a19b472e9cd142d29c6279d2 : SUCCESS in 4m 12s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/d43c7cc4dea942f3a1f5055ee70e3662 : SUCCESS in 5m 13s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/4f49e9dfb5e24494be2a61186b1dd74f : SUCCESS in 11m 56s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/71cc85411b7c49248317e741b2b91bd3 : SUCCESS in 4m 35s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/757af82a402f41d391efcec22138bcae : SUCCESS in 8m 09s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/af40bc05d03448f296be9891f5c91067 : SUCCESS in 7m 42s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/d63ceac9cf66497f85411ec4de024407 : SUCCESS in 9m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9b07e3d9ce474331b9043d94459e4286 : SUCCESS in 7m 47s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/cba8d9c0949c40078183bfdf3c1a8b9b : SUCCESS in 12m 57s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/e5d8ea4aba9d42f0b6e6fad0a1588fa9 : SUCCESS in 11m 10s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/a5332a004a6f46d9adb418238a48a160 : SUCCESS in 3m 34s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/51edfab01ecf4fdc9adddfa0897e4bff : SUCCESS in 8m 54s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/bfd7f6a08dac458dae27e9976088762c : SUCCESS in 4m 38s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/10036d767e3e4b729895f6c5bbf1bb82 : SUCCESS in 4m 19s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/2daa3ed13d0546cb8f4fd0d01f07833e : SUCCESS in 12m 29s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/c239a64174ab4accaf56efb4daa27bc2 : SUCCESS in 23m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/38e80ee4350149e9b2116102186273ef : SUCCESS in 7m 16s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/2d31d07bbb4349e3a27d654b154821d8 : SUCCESS in 12m 49s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/6f85310c5c5747d6b223e98f3de3fe90 : SUCCESS in 4m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/83ee0c4307f34cf1a7c780b74e013be4 : SUCCESS in 5m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/57095e79796e4cfd9659f89d09e256dd : SUCCESS in 17m 33s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/94c718ea1f0446c190dd85462d128555 : SUCCESS in 4m 32s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/7253093ae53c47bdb8b0578217ae889f : SUCCESS in 7m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/49e7667879674b0ba236a1633b049bd3 : SUCCESS in 6m 18s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/01898ec48f414838a7f4c8e1360ab962 : SUCCESS in 14m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d4c469cf6804d20b7e1f10418a30c15 : SUCCESS in 1h 11m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/341e0ae2488a450cacbb456c59c64806 : SUCCESS in 1h 20m 51s","accounts_in_message":[],"_revision_number":1},{"id":"9eea75df57f7b60ea82af20e78b7e9a887a94d7b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-05-23 12:11:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"41a5e72e0f68205d17fc04f01ff9c47c5e770579","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-23 15:41:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"78cd393ab9c9a32c309d7faaff0d08775b6f33d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 15:42:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e904f06eca315738fbeb8e223f5bfc271b970bec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 19:11:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be635f4d953f45149d04752cc7680b1b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/668783606dc842d198ce1916d6df8bbc : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bdc593e70fc4718affcdafd81e6d9c4 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6ad1ed7b450419c89fb38126fab34a8 : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/739f1f689ddd4e0881fa91368b17043c : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dbd3c865ed3448f6be6116794793bc14 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1ac8df4a092a4bc5b320a758e3862c84 : SUCCESS in 3m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0ee7079856ca4766ae04f8b24acb905b : SUCCESS in 8m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ceec022e17f648368111a654d7bfc13a : SUCCESS in 22m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/10bc692ba7e64575a27ac0f6bb24c575 : SUCCESS in 10m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5b2e5cab60a640fd91b0a524f4919011 : SUCCESS in 25m 18s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/4b0375d5a5644af98d515efda8330511 : SUCCESS in 13m 27s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/7c52a10ca3c7430d898834fbac6d4528 : SUCCESS in 5m 09s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/c5e17a66cc514c37bf39d71323e6f754 : SUCCESS in 5m 24s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/12aaa7f021614a2d971069f0eaa16f51 : SUCCESS in 8m 10s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/83f931c6d32e43158a05ded0d297c74c : SUCCESS in 12m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/841bf7567f674f2d9d1652f63e0d21c0 : SUCCESS in 9m 24s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/d9b5f063bf824b28b7e7494485f9d639 : SUCCESS in 14m 21s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/72a6b5c1ef6e421baecb2bec34c8cc40 : SUCCESS in 12m 39s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/7af241ce46ba43f78180d1006e077f86 : SUCCESS in 3m 53s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/cf891ba05c5f4118912fb3093f29f24c : SUCCESS in 10m 27s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/d9ce5ad599bf45c9a753d07a3c754ad8 : SUCCESS in 24m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2ea6720244f840f8b842f0735c4a15af : SUCCESS in 6m 35s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/aefa9ca6401244b1991b6cb096883299 : SUCCESS in 14m 49s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/8691a329044a4e29b4af332c2850f301 : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f1e32948498245c8b7654c0fed2d73f4 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/352e7589020349a9a505a9827cad3914 : SUCCESS in 23m 07s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/3ac841a65bbb46e4bcb6837873face4e : SUCCESS in 4m 47s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/74b582691f0e4486a8a879ab713495b0 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c91d81e8ed554e458a41e1cf05545eb7 : SUCCESS in 7m 44s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/a9bdde802a1b4a9688fec034af87f3f0 : SUCCESS in 15m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a6a5bd14e354a378531ccd8b8d9b766 : SUCCESS in 1h 09m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4790636ed1e4adea126efa2290b9a68 : SUCCESS in 1h 24m 26s","accounts_in_message":[],"_revision_number":1},{"id":"006042a7f3de7f84a2d11153e9e2550fbe7eb108","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 19:11:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3fc5ffdee5ba70a1e2cc239c77a20c5eeb066ddd","revisions":{"3fc5ffdee5ba70a1e2cc239c77a20c5eeb066ddd":{"kind":"REWORK","_number":1,"created":"2024-05-23 08:05:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/920248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/920248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/920248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/920248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/920248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/920248/1"}}},"commit":{"parents":[{"commit":"8aaccc37bf096cf6ea19f0a4a1a67ed91611cf1e","subject":"Merge \"update constraint for python-glanceclient to new release 4.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8aaccc37bf096cf6ea19f0a4a1a67ed91611cf1e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-23 08:05:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-23 08:05:35.000000000","tz":0},"subject":"update constraint for ceilometermiddleware to new release 3.4.0","message":"update constraint for ceilometermiddleware to new release 3.4.0\n\nmeta: version: 3.4.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I9006d10bcbcf60cb0636b8876a67a52e1e56a1d4\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I8be43e13460de01e67a8df862e57bda35a728f48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fc5ffdee5ba70a1e2cc239c77a20c5eeb066ddd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fc5ffdee5ba70a1e2cc239c77a20c5eeb066ddd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
