)]}'
{"id":"openstack%2Frequirements~920253","triplet_id":"openstack%2Frequirements~master~I24f4565fa4a087a03a1f7bc6495ec106441b1c0d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I24f4565fa4a087a03a1f7bc6495ec106441b1c0d","subject":"update constraint for python-cloudkittyclient to new release 5.1.0","status":"MERGED","created":"2024-05-23 08:07:40.000000000","updated":"2024-05-23 19:11:04.000000000","submitted":"2024-05-23 19:11:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920253-new-release","meta_rev_id":"a0dc5b09185dc71b9287ced7d5a45dbe2ade8688","_number":920253,"virtual_id_number":920253,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-23 19:11:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-05-23 12:10:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-05-23 15:40:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-05-23 15:40:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-23 09:56:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-23 12:10:32.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-05-23 15:40:02.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"a04b5c6402928dcb390ba55cec4285e14feb8a36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-05-23 08:07:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3daaa7f77516936a696be83697ad25968bb7689","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 09:56:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62c0b1e0a3b6489bb981aac3cf0df957\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6856e4abe7845189aa186b993fd2f98 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d6310f6d3a405db21d6c464f053680 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84372d23bc6f4b6083a4bc430c0423f4 : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61de62d37f7249729c3542fb5183813d : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f78ef617f5e4f0c8cd17a6f3b7bdbff : FAILURE in 5m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af2f7cdb2f54469aafdf8bd6a237a5b2 : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/24ed33f4df4942b0acc0501e9c62aaaa : SUCCESS in 2m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6a0645c71308426a832fec8bb940c99c : SUCCESS in 9m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3cd830a169e74741b859a51fbc648766 : SUCCESS in 23m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/35ddacc83aca4bdd8dec71debf5e9b0b : SUCCESS in 12m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/32d11d4b5d08434fb18d68940250c937 : SUCCESS in 21m 55s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/398263c2e6e84ccfa02b326c7d2825c2 : SUCCESS in 4m 54s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/8a95ff10e3e24d8aa5f1f7ac5ff16c64 : SUCCESS in 5m 22s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/9b8a19a380a2439bab092946f400666b : SUCCESS in 4m 21s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/d4231ed258c148bc95b99e66865b9ee7 : SUCCESS in 12m 03s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/2768d3b34e1648ac9f2b244515d05c6c : SUCCESS in 4m 31s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/19e7ef276a6c4ae29d92cd5bae004e7c : SUCCESS in 5m 17s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/cd85e8be7c1d420aa774621c4ba93844 : SUCCESS in 8m 28s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/c644f8249bba4e998d7c38bbebf72bf6 : SUCCESS in 12m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/76dfe08ec7a446529ed1f5715462d49e : SUCCESS in 7m 24s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/b10905ae02fa4318934b4d7dd339280e : SUCCESS in 13m 21s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/f80422a0e1a74b77888eda15fc692d4d : SUCCESS in 10m 57s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/eef9c7e0f78d429fa488d8ad98a55fc7 : SUCCESS in 4m 46s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/42fc79f15ec94fcd81e8d7a3f2cb481c : SUCCESS in 8m 55s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/636524132e9140ffbb98a64bf8fb3ff8 : SUCCESS in 5m 33s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/6c240081d5a847bd97a6e2dda21f8728 : SUCCESS in 4m 54s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/7b38cc1fd2114bf0b1ee6af3f40636be : SUCCESS in 10m 55s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/40af924a709941f8be5d7e2493841a87 : SUCCESS in 22m 48s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5f157a0a59b04a0293d7830b0e807da6 : SUCCESS in 7m 49s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/719d29cbd9a446fc98ffbf83e1100f7e : SUCCESS in 15m 37s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/3d8b90c368574a4b909f32444f9618c2 : SUCCESS in 4m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f820bd85549b45958b2d929ce8acc7d5 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0d0481d777394c3a9222af2de2d0ce85 : SUCCESS in 24m 06s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/1a529254ae8444e8b0ed0a56e64feeed : SUCCESS in 4m 46s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/a58b49e6bb8c4c67a289d514621895e6 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/58dc6ed570844f10a21ec80a61f18130 : SUCCESS in 5m 48s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/3b96b9d6c6a048e88efeafe80f0717aa : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b335a125af148a0a18fcfd6c0ce1907 : SUCCESS in 49m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44acc8dc8f4b4d5586c9cbd6bca89595 : SUCCESS in 1h 40m 44s","accounts_in_message":[],"_revision_number":1},{"id":"9d9946bdc695ecf011e68bf2717b3d98c38c1bc1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-05-23 12:10:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b1248be0390cf7e2910ddc2e287a912679029ea8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-23 15:40:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cb40993f7be12b8a7f9b7f06ad6e571d12aeeecd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 15:41:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf114fb9b60396f14660f8b6d0a55f147f24f6e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 19:11:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54124a570a604720ab6d4283a7e89bc4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/444db4567b8d407db480350d1e5c1bc0 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e31ca6b61f0640f6ab12f74f43d453de : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac500c523acc4f939944f7d481913d34 : SUCCESS in 6m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/99b1fa30a27c4cb0b4194c9f5fe7094f : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a8761d8fd5345cda24b23157b4a8cb2 : SUCCESS in 2m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e35189cef156441384b924a223eb0daf : SUCCESS in 4m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bbc90e8988c141eba08a1a13c7dce81b : SUCCESS in 9m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2476c6c7e39948ef909097adba3a87c3 : SUCCESS in 20m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f4517df2e52a42efaca3702df7609d66 : SUCCESS in 9m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2dbdba63d7ce4806a8a6ac1d3f499c28 : SUCCESS in 24m 03s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/0807d6e22da14bc7a0223d0142340659 : SUCCESS in 11m 33s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/5817908195b4477dbf4602df9d5b15f9 : SUCCESS in 5m 51s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/428f3fe8dad14567b4f962ad2d08941f : SUCCESS in 4m 57s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/98bd1ab5d0a743038e5a46198aa84123 : SUCCESS in 7m 43s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/614183a199fb4036904ea76ae9fe0c29 : SUCCESS in 12m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8df7ec8ba38e45ebacaf7ec4db744e46 : SUCCESS in 7m 12s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/daf2f3e4e86942848321f5a33321a1f2 : SUCCESS in 13m 06s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/13cb1d0cb6554a90871659bdadc4e839 : SUCCESS in 12m 22s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/612676b0579f45d1802227f0e48a54ab : SUCCESS in 4m 41s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/9da03a3c61464ce3b5b3d32468022768 : SUCCESS in 8m 52s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/73ff62bfd1f14cefbd8d2520c6b6a750 : SUCCESS in 25m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b63943f38279445e80c4ef7382da54ad : SUCCESS in 9m 51s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/8ca09c6f22de47409e1a789955a2d65d : SUCCESS in 13m 44s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/5f7ff95b811a4bc49726bc82bd464c7d : SUCCESS in 5m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8b4ce01472d94e83ad6c862db8e34ed5 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4e54637505bf4367946811dc30cb27b6 : SUCCESS in 25m 28s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/c79c8bc41a254620af0bc95db0291d50 : SUCCESS in 4m 23s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/1bc32ef196944575b299e6646e70d3de : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f0b42aa5b854445d9163c70ebb34808f : SUCCESS in 5m 44s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/c2518b9b8b6c421991956312ae8e22fe : SUCCESS in 13m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/024b9068315b43969f16584fac60a4ac : SUCCESS in 58m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/45016738bb924257ad48eed89bed42c6 : SUCCESS in 1h 29m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a0dc5b09185dc71b9287ced7d5a45dbe2ade8688","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 19:11:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22263899f9703f1fcdb09ec8f268592fc2b98a15","revisions":{"22263899f9703f1fcdb09ec8f268592fc2b98a15":{"kind":"REWORK","_number":1,"created":"2024-05-23 08:07:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/920253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/920253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/920253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/920253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/920253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/920253/1"}}},"commit":{"parents":[{"commit":"8aaccc37bf096cf6ea19f0a4a1a67ed91611cf1e","subject":"Merge \"update constraint for python-glanceclient to new release 4.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8aaccc37bf096cf6ea19f0a4a1a67ed91611cf1e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-23 08:07:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-23 08:07:38.000000000","tz":0},"subject":"update constraint for python-cloudkittyclient to new release 5.1.0","message":"update constraint for python-cloudkittyclient to new release 5.1.0\n\nmeta: version: 5.1.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I293f71412a3d3c04bd3a6ddb70fef113d48d200c\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I24f4565fa4a087a03a1f7bc6495ec106441b1c0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22263899f9703f1fcdb09ec8f268592fc2b98a15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22263899f9703f1fcdb09ec8f268592fc2b98a15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
