)]}'
{"id":"openstack%2Frequirements~920255","triplet_id":"openstack%2Frequirements~master~If612d3eb3fe17617d6200933d50a55ddfb572315","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If612d3eb3fe17617d6200933d50a55ddfb572315","subject":"update constraint for python-barbicanclient to new release 6.0.0","status":"MERGED","created":"2024-05-23 08:14:39.000000000","updated":"2024-05-23 19:11:01.000000000","submitted":"2024-05-23 19:11:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920255-new-release","meta_rev_id":"dcce6b6f9cef2f2c13b120ea39b49e0347487924","_number":920255,"virtual_id_number":920255,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-23 19:11:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-05-23 12:10:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-05-23 15:39:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-05-23 15:39:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-23 10:09:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-23 12:10:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-05-23 15:39:35.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"dbc6afba7e5f864d41c017ca1b56c142ba9bf6e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-05-23 08:14:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ca7d07d8cc1e2350f87e9b8246f86d883659f6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 10:09:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/257f0fd1f0c949c68315a655a1732e18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cac3742501834f0e8db712193b646f80 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0ec26296afa40e490ae36420251cde6 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fbe5ead28454cc6aa0c9b6fd4a248b1 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/505f40774f1b44539cb9a4021df7386c : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/96c5d525c932491ca948bd5e3773e0e3 : FAILURE in 6m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/200fcccd42ee4369b76c2867fb131e2d : SUCCESS in 3m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c49ab723365a46598f1bfa9b644b3d56 : SUCCESS in 3m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/029c552da2a148a8bc1d643107e27e5c : SUCCESS in 8m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dcecc4c432974254a5e3452094bb0b2b : SUCCESS in 21m 47s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6b2219a5fbb549a7addd1ec1ae5ce825 : SUCCESS in 8m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4000a53f14e04797a1d5e64681462ebe : SUCCESS in 22m 13s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/aed874ac7d714ec6a703f04e402f6f74 : SUCCESS in 4m 30s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/e870447f694944bf8e6a4730fc11175e : SUCCESS in 5m 55s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/f3cae75e7c3d403b8d448ea8e8141deb : SUCCESS in 4m 07s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/257175be7c434528ba2aa24976ac2077 : SUCCESS in 11m 33s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/53ded9a7d81446cb9ddf38016a6b2ebf : SUCCESS in 5m 19s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/6d84ae6310c54fc3b592260e2c1cce20 : SUCCESS in 4m 37s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/49ed308a4d6f4c549992d82b7d209f86 : SUCCESS in 7m 17s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/259c116a113b4620b07c61f7a1c50fa7 : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5cef282ea6ab431daaed24b544efa0b2 : SUCCESS in 8m 01s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/d04dfcaf6d4e40cd9ca9bb273a0f983a : SUCCESS in 12m 37s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/dae8a163ec264b4193ebba47c7d0bee8 : SUCCESS in 10m 47s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/429c202c06f64dd4ad12533d69b0f80e : SUCCESS in 4m 17s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/8c997b65b5a14333a6881d899ddc74b5 : SUCCESS in 8m 00s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/6874447d171542b1b96fdbda1b83d2fe : SUCCESS in 4m 26s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/2db342047f904e209f240b8efdcfdab8 : SUCCESS in 4m 52s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/44ddf3d181a945be8717aaf0605a8363 : SUCCESS in 9m 46s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/5c9bac54e4e3403d830d709183bbe9db : SUCCESS in 28m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4059ba5651bd40108994dedf3e60df08 : SUCCESS in 6m 52s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/eb2162d2dc964727b908ddf44b1b355c : SUCCESS in 11m 35s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/0298e107cea741f9bbff4b134b714125 : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2a42b2cc47564996b3f399c8d41524ee : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c7ecb5788f0f46db93a6f723e35db403 : SUCCESS in 22m 03s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/c5553a623487447697388cc9e9fbda31 : SUCCESS in 4m 24s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/5cedf46b5d4f4dd4ba6a2be96ed0e9b2 : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bb3064a6a0314896bf97d87980d6a779 : SUCCESS in 5m 58s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/e048d731e7ef46afa0d40feb2c024b1e : SUCCESS in 12m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a4e96b796a141348c4cea0e667d5ccc : SUCCESS in 56m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec3a621e94504b76b5fb05090f298e59 : SUCCESS in 1h 46m 57s","accounts_in_message":[],"_revision_number":1},{"id":"61f4eeb75275b3108910a786dcc5fda59b5e8dc8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-05-23 12:10:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a260e5d348b8cc3616ed567583f5b7c8ccb6908f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-23 15:39:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"32be5e14463d5c60697b078d6106da26db569552","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 15:40:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0420f6ae3d8eb31d5e4d8559d33569d43e1b4b3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 19:11:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64cf301e8ffc45fe814b4ccd4ddf49f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80b636444d4940e8baf0aa9386ae63b8 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8039250a75e34516abcbdecaa20d4ab9 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2063b9f668bd4dbe890967b6c72a9e7f : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a6a6214b55442eca9749fd1ca5fd095 : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/92ad717f671947839c73f11952b354c6 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f27abb418ae4feabd42f7fc900dd4e2 : SUCCESS in 3m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3e2d2c630a2f46179873739e290d1152 : SUCCESS in 11m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2a1508a1b98c40a692a26010419751d3 : SUCCESS in 22m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7aa393fb26f445e1ab7aa8c23b9ca090 : SUCCESS in 12m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0392720d4cf5447b8d2f2e216574e6e9 : SUCCESS in 23m 16s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/0f85811da7e5498b887bfa6d885c8dca : SUCCESS in 12m 52s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/2d97a37d8fff485695ef90ae101bb5bb : SUCCESS in 5m 45s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/49232ee645f04a6da0116bb3f5aa8b90 : SUCCESS in 7m 06s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/915391f83230402ca7a2f2af1945d45b : SUCCESS in 8m 52s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/63d6395061d34f938aeca8453e38e2ca : SUCCESS in 12m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8f623f28732b4741bfb6ab455bbf78cb : SUCCESS in 8m 04s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/365d904361e54f65b7386e70486302d5 : SUCCESS in 10m 07s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/379a46c3a98f49de8e9cb453119c1b61 : SUCCESS in 12m 44s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/cb9a1d9616f64be0bfadd9418e7288e3 : SUCCESS in 5m 46s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/616e915995074bb3b25184a4f265735f : SUCCESS in 10m 41s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/afdaeb80ac3b490abb3e087a6bceef15 : SUCCESS in 19m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e455dc328aa247edb9e51c9ff13daece : SUCCESS in 8m 53s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/a6fd8cfe45444ca1b65c0392b583010a : SUCCESS in 15m 29s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/1e41897645944cc8971e1023709c0de8 : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/96d49195ebd9438abed03c0222146450 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/03cde68e55a74648a4cccfbb661f35a3 : SUCCESS in 26m 09s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/997ec2115b68428887c43a62ac805da0 : SUCCESS in 4m 35s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/4b4ef99114fe4452bb08a4abc41e3049 : SUCCESS in 6m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/435e29e506664b87940bb6a1f23bc73a : SUCCESS in 6m 31s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/1cbcc9397de74eea842dd29d08e0ae5f : SUCCESS in 14m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e43614c82d944f40a112d11469fcf843 : SUCCESS in 1h 00m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a59b8dd0037f44869014aa11cc7f04f4 : SUCCESS in 1h 31m 43s","accounts_in_message":[],"_revision_number":1},{"id":"dcce6b6f9cef2f2c13b120ea39b49e0347487924","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 19:11:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e1564cc9c77a66c69247c23b9905fa7bee8ca4f","revisions":{"0e1564cc9c77a66c69247c23b9905fa7bee8ca4f":{"kind":"REWORK","_number":1,"created":"2024-05-23 08:14:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/55/920255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/55/920255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/920255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/920255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/920255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/55/920255/1"}}},"commit":{"parents":[{"commit":"8aaccc37bf096cf6ea19f0a4a1a67ed91611cf1e","subject":"Merge \"update constraint for python-glanceclient to new release 4.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8aaccc37bf096cf6ea19f0a4a1a67ed91611cf1e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-23 08:14:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-23 08:14:36.000000000","tz":0},"subject":"update constraint for python-barbicanclient to new release 6.0.0","message":"update constraint for python-barbicanclient to new release 6.0.0\n\nmeta: version: 6.0.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ief557df1f858aee0d81ea0f523f43ade7619b8be\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: If612d3eb3fe17617d6200933d50a55ddfb572315\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e1564cc9c77a66c69247c23b9905fa7bee8ca4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e1564cc9c77a66c69247c23b9905fa7bee8ca4f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
