)]}'
{"id":"openstack%2Frequirements~920259","triplet_id":"openstack%2Frequirements~master~Ib8b654918353c5efee0212b717729aaa04c56360","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib8b654918353c5efee0212b717729aaa04c56360","subject":"update constraint for oslo.service to new release 3.5.0","status":"MERGED","created":"2024-05-23 08:24:57.000000000","updated":"2024-05-23 19:13:05.000000000","submitted":"2024-05-23 19:11:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920259-new-release","meta_rev_id":"185509936975f18a65af69f29f0ca273815381ab","_number":920259,"virtual_id_number":920259,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-23 19:11:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-05-23 12:06:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-05-23 15:45:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-05-23 15:45:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-23 09:53:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-23 09:56:29.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-05-23 12:06:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-05-23 15:45:59.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"973df70fbed19b1a15f9bf8f184a19a77beb42ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-05-23 08:24:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c9781e49c42837d2b6f6381c746141e84e2a37f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 09:53:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f73ef93dc5054c74a880debceb5052d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/653e72445e8e4629a15fdf41b41f78d7 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4b997a470864639b7ec640aa648af49 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f21910d7f6854dd8a8655664b17abeed : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2313522afc4340898a1afeab0b5d16fe : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/194df19fcf844c5c832925acb6f1cb35 : FAILURE in 5m 03s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/49a77a63d58f4b40ab7752c0d471bc29 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/51c11096895a4ffcb7e6829cb4bb725b : SUCCESS in 2m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6f86ab29615d48ce95399a309b469f5d : SUCCESS in 7m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6d929f9fffbd428bb6d0f75118ded34d : SUCCESS in 18m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/99c1a999cd9549cf9b6ccff780d31261 : SUCCESS in 7m 19s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/28ab0f9a4415424a828eb9f12e5536ee : SUCCESS in 20m 33s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/a1eee4dc358a4fbc87b85d35a1150468 : SUCCESS in 4m 10s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/ededfd3a77c8426babf37d75e824b952 : SUCCESS in 4m 46s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/4f119b9b3b8d4425b5a1e10329c2a3dd : FAILURE in 5m 45s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/df4e8f262b0f4542ae3c53a621559646 : SUCCESS in 13m 13s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/95518c6ea29b4910b2555de525e1bc60 : FAILURE in 3m 42s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/7c54c13a99484de4b3d583ce70e1e18e : SUCCESS in 5m 15s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/324b38c8b2b147d4a97f14bf0986e133 : SUCCESS in 8m 18s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/52275afdfe304eb4a4f9aa79c6b58810 : SUCCESS in 11m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5a37bb302bd74d3c92f0d700d0314399 : SUCCESS in 8m 50s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/6511ff97c39547ed933fdf8f83e9e910 : SUCCESS in 13m 27s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/b5e487bc08934c0788734255332853eb : SUCCESS in 10m 17s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/b9c258619d764a7199eda3b86191e6ec : FAILURE in 4m 27s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/1ff3e569e6a74d15a90e4fdf5037dea2 : SUCCESS in 8m 50s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/c8bf2d00648a4446aa2beb31f0a315ac : FAILURE in 4m 06s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/fdf8660050924bf5a2823b20e9cc7324 : SUCCESS in 4m 03s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/884e5a309ac746be8030e8964df550e9 : SUCCESS in 9m 39s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/9fa92cd266eb41df92288d4e34d6f06a : SUCCESS in 26m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d6f57c93b24a4d02a4f1ac78906c40de : SUCCESS in 6m 23s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/26749fd0327c44e99443df28af0026be : SUCCESS in 9m 13s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/dc2facdc8ae54549a5bf664d5bfa22b4 : SUCCESS in 3m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9a0dc055836f44d3bce3042317f1d9da : SUCCESS in 2m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9ff1575be3404a5e91205ca9677135b3 : SUCCESS in 24m 29s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/c1529be15f0c4f79a063fcaa9718622a : SUCCESS in 4m 30s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/ff431518537448ed82b451a8beb78261 : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/be8241e107a34e26a81f6b5b05fb440e : SUCCESS in 6m 43s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/2daa3cdcad47485e91046fd9c4604892 : SUCCESS in 11m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7664882b764e4d339dcc019f54450600 : SUCCESS in 58m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe515b2c5107441eacf4ef4f065ddbdf : SUCCESS in 1h 22m 44s","accounts_in_message":[],"_revision_number":1},{"id":"c9688075e5a2fe568146223483e090bd89b6ef1e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-23 09:56:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2ce38c9a5f5340b01ca3182ae95b1f409852c755","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 11:39:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d5d16b9840e45a4928c3ff029f1a696\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/413b773e3d434ed8829fc3c2070a96ec : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/793462fdec8144edb6c24a0fe60d74bc : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f35d57848cd04ed3a03da0317c989034 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05149a7eb3ab48b390123b37bce57d2f : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e4f1bc0af2540149b06e8917963981e : FAILURE in 6m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/36326999068848c0880f49a3b713ad37 : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/239d1b2ce24141bcac4e40d6e2175f77 : SUCCESS in 2m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/73c8e5a22e574dabb2b5aeaebe2d9232 : SUCCESS in 9m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0b3c20d6ff0540eda2ef83a57ffb2a9e : SUCCESS in 17m 25s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/851591ee0c81490ab08ac5065ae27643 : SUCCESS in 10m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c2db694d86374f379aef3932a196145b : SUCCESS in 19m 54s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/0d94a864bb6d4479afc8aa0e64d54500 : SUCCESS in 5m 30s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/3e0c3c16412548cc9b3d972c072708d8 : SUCCESS in 5m 54s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/9b3c08378fe246949bebaedc801c8c84 : SUCCESS in 3m 56s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/a44d04c75abb48f6a0b434b272b49e28 : SUCCESS in 11m 49s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e6141c4f71fe44deb768e8acec19078f : SUCCESS in 5m 07s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/8a2400ecf4504931a906e28e21b341cf : SUCCESS in 5m 08s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/ce09e38e29fc48aab35d6f6b45f0296e : SUCCESS in 6m 03s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/6f246a3e05ae425384cae6b412628970 : SUCCESS in 8m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c64d430ae127436484f207118f6c5534 : SUCCESS in 6m 58s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/7d2bbd36f7f140bea80e31c0b9aec4c8 : SUCCESS in 10m 42s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/23eb3a0a7a3942bbaaf1445f15b93f20 : SUCCESS in 7m 33s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/20f5965a004644dea39478b168eb3b06 : SUCCESS in 4m 04s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/08d846ad20394e7cb87521787b3dbdf6 : SUCCESS in 9m 02s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/97cf97a3801e49a2a2586b1c135f7e6e : SUCCESS in 3m 38s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/cabd36a6d4894ec492821a137779c24b : SUCCESS in 5m 21s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/c1c41ace716a40ceabc1ddf7eaf2784c : SUCCESS in 12m 12s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/268489d06f184dfd882ade05feb40cd4 : SUCCESS in 22m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/588a8e3ebbbf47d99cc062d68cf08951 : SUCCESS in 5m 19s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/7c8e6cce333148b9b4869acb04b14cec : SUCCESS in 11m 06s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/bca7c67a76e548f288b17adda61d10ca : SUCCESS in 4m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7da39c83b6c945359b8740d5e1a6cdd1 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3a3e889fae62428cac9e93d3bc3b529e : SUCCESS in 17m 15s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/754dcd7758fe47bab6fed31a88425411 : SUCCESS in 4m 07s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/9d97fdf3f1114ffb8a0f994cba1580f4 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e7acdd0d068849668e7049d60ccc4ee9 : SUCCESS in 6m 14s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/3d85345cd3434b729ad12747a244fc43 : SUCCESS in 13m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/353e58561ce24717b3e50a180847e5f0 : SUCCESS in 1h 03m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76877a29f96e4204816bcd334d029c58 : SUCCESS in 1h 36m 44s","accounts_in_message":[],"_revision_number":1},{"id":"31c6aff9e64ad279d1903c1babb696ddef737420","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-05-23 12:06:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"37e594224337b6bd474f000b035d96d2920c62a2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-23 15:45:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fa21b300dc3499d1d72524b29f183e224a9d32aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 15:47:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf3b307aa7be2556ef97adbefb4eb80ad12a0d01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 19:11:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36d8a5beae454e1f9ecda3b245b3e424\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fb3e7ea280445af9c516549f0b64722 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e33d36da2764eaf98be736c7c881f25 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83ebe44f8420419ab3512a6a053f4db2 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b4e634ff7a994e7993420736403696c6 : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/15396a8694a14185a46c00e465424dd5 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c199e96cf47a4cb38075ffce277406ae : SUCCESS in 4m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3be054abc3ba4a838a77354ffa009fae : SUCCESS in 10m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/afa8d6c6ae2d4b429a00c5e297b8542e : SUCCESS in 19m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/07677c5cb7f04d8481b304dd2e5f53ae : SUCCESS in 8m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c35ae143fd2c43d9b73279eddfa4d9cb : SUCCESS in 23m 10s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/93a5956c3bff40d6a603f62b2a73965d : SUCCESS in 10m 07s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/04e8751729c3411d8f8401c9dd886fd2 : SUCCESS in 5m 09s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/b03f997b3617465d89de41e1dc6946fa : SUCCESS in 5m 10s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/da997b39170642099053ab39b7feafb7 : SUCCESS in 8m 40s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/703c5bee63d74154b3766a2d6c04eb02 : SUCCESS in 12m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/269790cbaf4b4449b1a949738507c3c8 : SUCCESS in 7m 35s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/af62ff8914924041a9038d97bc32c5bf : SUCCESS in 12m 50s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/789878ab92d249abba0b3f5ae9f339ab : SUCCESS in 10m 56s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/a47503a6ecce4489a2e5c08ea98ad100 : SUCCESS in 4m 48s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b42b5469c57f4fcfbcddd53d0c45bb12 : SUCCESS in 8m 13s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/2040818801b84effa304f096bb373242 : SUCCESS in 24m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9d38d421815d448c8c92f32b7b9ea412 : SUCCESS in 7m 20s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/085028ca0bc843b1acef6e6fab73bb3b : SUCCESS in 13m 17s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/74c876167772456cb74b8d77428e9e3b : SUCCESS in 6m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d384785303ff42b2987dc8dd624ffb9d : SUCCESS in 6m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/68f0a1e5b8e2475fb76b5746e5653565 : SUCCESS in 24m 45s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/5dd2a842e55d41389ef0d40c905a46e3 : SUCCESS in 6m 17s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/133e42021d2e410ebaf4d63751a73bbb : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ac311cc905504b0a9971177542a92f42 : SUCCESS in 6m 08s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/adcd4e88218e483f8d5775b2371611bb : SUCCESS in 14m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e37c2c42fce42738cc207c15601e1e3 : SUCCESS in 1h 13m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/030fda0cb4964b618a5afdb6588e7b19 : SUCCESS in 1h 30m 37s","accounts_in_message":[],"_revision_number":1},{"id":"2efac641fd79cb02fef1d10489a5c0d447b50f32","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 19:11:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"185509936975f18a65af69f29f0ca273815381ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 19:13:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e33e30b0cdb4a17a8011ed5c579ce80\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ecb9b6add3246a99f70c804bff0af4d : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"252da6176f08866cdaedfc68f033a538dacd7129","revisions":{"252da6176f08866cdaedfc68f033a538dacd7129":{"kind":"REWORK","_number":1,"created":"2024-05-23 08:24:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/920259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/920259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/920259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/920259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/920259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/920259/1"}}},"commit":{"parents":[{"commit":"8aaccc37bf096cf6ea19f0a4a1a67ed91611cf1e","subject":"Merge \"update constraint for python-glanceclient to new release 4.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8aaccc37bf096cf6ea19f0a4a1a67ed91611cf1e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-23 08:24:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-23 08:24:56.000000000","tz":0},"subject":"update constraint for oslo.service to new release 3.5.0","message":"update constraint for oslo.service to new release 3.5.0\n\nmeta: version: 3.5.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I965cd3b051a088cb131049837482541bac478dd2\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ib8b654918353c5efee0212b717729aaa04c56360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/252da6176f08866cdaedfc68f033a538dacd7129"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/252da6176f08866cdaedfc68f033a538dacd7129"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
