)]}'
{"id":"openstack%2Frequirements~920262","triplet_id":"openstack%2Frequirements~master~I50f19e4a1d95001be44880b9b36c5ba134278555","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2024-07-08 06:26:16.000000000","reason":"Change was submitted"},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-07-08 06:26:16.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-07-08 04:41:19.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I50f19e4a1d95001be44880b9b36c5ba134278555","subject":"update constraint for oslo.versionedobjects to new release 3.4.0","status":"MERGED","created":"2024-05-23 08:31:39.000000000","updated":"2024-07-08 06:27:56.000000000","submitted":"2024-07-08 06:26:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920262-new-release","meta_rev_id":"02e1b1b377726d7e5b7bee51055e7c24e30f5d88","_number":920262,"virtual_id_number":920262,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-08 06:26:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2024-07-08 04:41:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2024-07-08 04:41:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-23 10:09:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-23 10:28:05.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-05-23 12:06:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-05-23 15:46:39.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-07-02 01:28:41.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"cbfac47e4e0e68f57c02bf8cb094ec4988c0e56e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-05-23 08:31:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ec0950a0e1cdf529637e171d3eeff2cea7ce3c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 10:09:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/87a4ca20f8494af0bf0a5d7be453aaec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7ec31a6831e4d61b40cbcc5cc554326 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a1451439a99463eb41f47fe819a4f89 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c32f227f7744a2ea42475f24916b9ec : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f95fb7bc0b6c45179004001c61374ab0 : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd24a7ee7f464a769d0a233e030957cf : FAILURE in 7m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c95fb3521bee41bfbc38ac5ac96cfff1 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e4dbd4045ef447a7a5e8bf04d82b7655 : SUCCESS in 2m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/22b1424bc1a1439e929f52cadcb21cac : SUCCESS in 7m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/16f7b47cd93c4b1bbc98402f681700a3 : SUCCESS in 22m 09s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/06ee714f16f64115af14d1ede5221a20 : FAILURE in 2m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/66e8022a40d748579bf6f4f63da50d8e : SUCCESS in 22m 00s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/423a4f7d5ea84231bb1c758774672a74 : SUCCESS in 4m 02s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/a34be44600f74a91b369534f7d5a198d : SUCCESS in 6m 46s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/93276912516e4186824089b720fa124d : SUCCESS in 4m 09s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/42eac59c165d4bfeaefcbbcb5141e8f2 : SUCCESS in 11m 16s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/424ffdd344c5472f97122bee8b837395 : FAILURE in 2m 01s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/7f680f87e4194f589595e3623dce4e8c : SUCCESS in 4m 45s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/6fea48b6b45548bb85b06eb75f0c96b6 : SUCCESS in 6m 28s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/816296cacfc34b4ea5028f843e8077be : SUCCESS in 12m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b55e544e55b745e89fb301410ea40f69 : SUCCESS in 7m 44s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/b559fe6f2fc54707816c9666a38309a7 : SUCCESS in 10m 38s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/df3de67ed6ee4592ba40d4f591175e9f : SUCCESS in 9m 40s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/b7278964c4f440d7aed2d44ab82be038 : SUCCESS in 4m 29s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/a2f5429d512a418683bb8576b4c2ae9f : SUCCESS in 8m 50s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/cca24207b37d4ae7b002733423d751de : SUCCESS in 4m 33s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/41024f606b094648ae40e16e2a33b90b : SUCCESS in 5m 00s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/4ac3b2eaa63749df835e1a79345b3623 : SUCCESS in 9m 55s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/fb1073da2a2f4e9e99baf75f624a07c7 : SUCCESS in 27m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d400dc24c3b14161a76869e8596ecc8b : SUCCESS in 6m 21s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/f47303537a50431fa82f1839e1e9218b : SUCCESS in 11m 50s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/3ab338ed21d84579b5482b7148c6cc85 : SUCCESS in 5m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4105974fa8334a35ab183d9bda9a6d78 : FAILURE in 2m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c74141fac63c4865985988c395febe94 : SUCCESS in 21m 12s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/c3eb7e641ed24a25baf9d33c5a38fb7e : SUCCESS in 6m 02s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/1d45857548324c9786e38d62acb512c4 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5399aa817dd44e11a5df8b6b9cd64044 : SUCCESS in 5m 32s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/43b0a2e85b4e44e5b25669cb61c01d55 : SUCCESS in 15m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c7e6e3fb5cc48a2b0b0b9e7ebd704b7 : SUCCESS in 1h 10m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/737ad5de539042e8a667d7a7eaa58cff : SUCCESS in 1h 32m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d4e95070e96e5bb0c50c48f58f82c62acf25e6f1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-23 10:28:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e2a73e24000f83d46a6842eb54b0b51afb56c2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 11:58:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e5c5d7cb2ad441e9676eb1cb18666b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/792e7aa8b79140fe82b8d89f78e64e2b : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/763613d5afd7449fa12f0c256bc9334b : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bbe28768bd144b987b6232dca99d341 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c0e4170195142f1bb4519690d096b05 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a28046931f49464eb71bc22750d152b7 : FAILURE in 4m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d5efe02a8b1642e5893eda9fe648c6f6 : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/014102208c504a0b9b6c919e0b416a65 : SUCCESS in 4m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5501ed99feac43b6a3af7e35b027cc42 : SUCCESS in 8m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/16c310df4deb4edd9fe1ff1a7065266a : SUCCESS in 20m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9dd2015f2beb47ba884f3115b0b68f54 : SUCCESS in 9m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f4c28a69de2d4d3bb48acb036cc4f429 : SUCCESS in 21m 18s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/a613e3c684134ec39d72a1e8ffb21c85 : SUCCESS in 4m 54s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/c61796174eda4ffe97c7d284c0369777 : SUCCESS in 6m 36s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/83a5de9d1b3c44558e2fb3d6067312a2 : SUCCESS in 4m 10s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/33d5e0b700254b5ebdb71d86fa8d5d1a : SUCCESS in 9m 33s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/5675d22e290a409abc22c78ad0622395 : SUCCESS in 5m 07s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/9d9317306e7147e3beef49fdacca3695 : SUCCESS in 5m 30s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/84641cc12356431fa580f51b9ca87efd : SUCCESS in 7m 05s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/709a66348fe043c483ed8dd24e910d0a : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3a89c1291a91427dae66635ab2fcb296 : SUCCESS in 6m 37s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/79ba214124304962a5256fa2b4279b7c : SUCCESS in 12m 17s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/a0be9539f1b241209b01448f257a6b34 : SUCCESS in 11m 35s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/247cd14728444079a2e50016789eb120 : SUCCESS in 4m 06s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/066a0e79f6da4456a38e7ff5b289f339 : SUCCESS in 8m 07s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/825af55cb73145c4bf2cbcc165d7c9ab : SUCCESS in 4m 36s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/d7cf75ede5964c858dacfacc61e8a3f0 : SUCCESS in 4m 01s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/023ce86fb1564773aa396cd429bd2597 : SUCCESS in 10m 56s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/6e0ee48881e64b9ea7d2ab980f6c31c5 : SUCCESS in 20m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/daf0236bfd5840d28f4ce0afea862a2c : SUCCESS in 6m 35s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/1684f8b481e04d1ebf38e7cb7cb8c6c8 : SUCCESS in 8m 50s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/b8ad1f89cbca404c83aa6be9a6814824 : SUCCESS in 4m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ad845beb3cf744e9843b30b81e235769 : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/161b73a3b5c949cb8692ca8d7ddd4893 : SUCCESS in 18m 59s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/b3916bfc53854678aae3ead8e06b432d : SUCCESS in 3m 48s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/2c958da678c84f1999dc134e7d899ea0 : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/49d39635ad634d2ea53fe0ae2b2e9163 : SUCCESS in 5m 43s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/6668dab3932e4424bba82d4b075ff05d : SUCCESS in 8m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d587ec70b4a49f9b8b048f2185d4d3c : SUCCESS in 1h 09m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a67485a9a8174fa5a30b545eb89d7b19 : SUCCESS in 1h 09m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8e081711c9b3e15ec8e89cbeafb0d2cb79963ae6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-05-23 12:06:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3683555bd2c3edcc932ec9a054ad10f3ce8fbed1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-23 15:46:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ea6469a94a45f9748b3cadb0cb5e1bab7142a815","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 15:47:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"930ebe2fa8480d912f5ff367624204f80b70da94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 19:21:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f77d9b34addc46aa919b2189a0ee9e01\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66ec5123aacb440bbc4492aaefc5a704 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d19b409f3ae4e8d93ec6f6364b5b4f4 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4444ff7d43d0405ea741902ba2ae2af6 : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e1d1a9eb2874f7a9ffe719bbc24a756 : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f58e263927f4455a71e2d3c44e9704a : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7e0aaae16dfb4aae945453f9612a18ec : SUCCESS in 5m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7f476f4f6f9748b880d6143ac00c22f4 : SUCCESS in 9m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7cfd9c0077a84f9686fa64bffae1b0a9 : SUCCESS in 19m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/154a1cf4a01c4f8a934be0e8f8bd8772 : SUCCESS in 8m 56s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3290f9987c7d4d3198b41f3297c8cd90 : SUCCESS in 28m 49s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/64e7fb603ecc48949568937408967a45 : FAILURE in 13m 30s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/79a2aae55b254793b11c906926591ff3 : SUCCESS in 5m 33s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/a7690cf33cd54339b16262bd9294679b : SUCCESS in 5m 56s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/f8197c84e0d74d268bc7415955d50edf : SUCCESS in 9m 07s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/c0be7bb6eb994463830a6cb1bf97619e : SUCCESS in 12m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/60cac3e65cac4b3fb00f6b2161ab1c3e : SUCCESS in 8m 22s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/ce21a80c86e74e0a8b9ecf96ab499ea5 : SUCCESS in 13m 26s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/565a8f4a61ab437c9438fe71d3766b3e : SUCCESS in 12m 52s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/8571f6027b16489d8cfce7cd76fb5748 : SUCCESS in 6m 02s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/1880d1181ac34df0b6a0d7a088a41276 : SUCCESS in 10m 48s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/5baff5b64d1a4272841db38a60934920 : SUCCESS in 31m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/64cb84d648a64e98abd63107a733c9b4 : SUCCESS in 6m 38s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/f91a2ca85ff14da4a7ffcd1a28c7f7aa : SUCCESS in 11m 15s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/01e00baf5a7346289c4fe0fea11b8d05 : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/330170402f014a9696ad67b408461652 : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e4e4fc78b8344e2cb44d2f03367db41e : SUCCESS in 24m 11s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/1ed213a7354c4baba2e123800eefd5d3 : SUCCESS in 5m 58s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/302f81a581344b5cadf6c1edca2f8db1 : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/42c35a06255545ad8c4fb792126b6ecc : SUCCESS in 6m 16s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/f41199491201477892f5d5e21d0eced8 : SUCCESS in 13m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3a2dee49534466ba391d5ee75be4ca1 : SUCCESS in 48m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b05044ed935548e2aced4d354687a973 : SUCCESS in 1h 32m 09s","accounts_in_message":[],"_revision_number":1},{"id":"2f0bba9e352d8d79f0fc298e0b08cc720cc941aa","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-24 10:05:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e875fbbcfc3f5906d0ce294774e17f42dae0d3cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 12:18:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/957af2eab42543e28b96f5a42f75feb3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8483f965fcae46e3bb03de11714b946d : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0162bb9e32e045e4a74ef94fe64fe951 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b67712d539144158a43c97107a171521 : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1cc52701bc140ce8d34a3491b63dc80 : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b236738d60b54b848a5cec780a6d6ea1 : FAILURE in 6m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8d597c15e64649449197f4992c833b7c : SUCCESS in 2m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/420cc62013e54403a9063cc9481467c8 : SUCCESS in 3m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1493a6b93ce742cb8ec87801b1d06f91 : SUCCESS in 9m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/be5ef03225794909a807b0d365880de3 : SUCCESS in 21m 18s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/327195abf2934ef7b52a91cbe1c3f27f : SUCCESS in 9m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8e49d50d052440dda13df0a05e7ac53b : SUCCESS in 23m 08s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/6c1fa8e327c048549baf10715e06e173 : SUCCESS in 5m 26s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/54cb6c0c91374573a50f125cb5e15096 : SUCCESS in 5m 13s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/d06b1969af4644748ea15ccb78cb2b04 : SUCCESS in 3m 52s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/14dd7483ae884f518c27af4b7c0f356b : FAILURE in 10m 21s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/d7fdcc50988e44928d32d76fd86770e1 : SUCCESS in 4m 55s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/6e003151706a4c8eabf6146d745df643 : SUCCESS in 5m 11s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/3bfed6836fb9408e94c6abc2834edc7e : SUCCESS in 6m 08s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/98f1a76bdf894ac7853fdb8338ec552f : SUCCESS in 15m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f1443307f8054e72892eabbcd7bcc2c1 : SUCCESS in 7m 19s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/05d2295eeef348a08acea481c4245fb4 : SUCCESS in 12m 18s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/61627922458d4944885f9dac30546c81 : SUCCESS in 12m 52s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/572ade88902a4940a65c328884838cf4 : SUCCESS in 4m 00s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/432eeb52734e44f2aa4c3fb249c2e136 : SUCCESS in 9m 47s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/5161f5ded3d14c4e99a750f6bfbfffe7 : SUCCESS in 4m 46s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/84205fe369284eafb697184d6987ce26 : SUCCESS in 4m 32s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/9c1aa6e4eb52488eb3e79ef15307edb2 : SUCCESS in 11m 38s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/19632b2c4a144e59929415a587a80240 : SUCCESS in 26m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/eb66c4fea6b14d4db037172680e75244 : SUCCESS in 6m 25s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/05b065a0cfca4368abc073cbf0aeaa24 : SUCCESS in 9m 42s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/981e9d9b81c24681b4c30a2713b9dd83 : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e2a4f76e58694bb299329f1e2b710cff : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4c802ea24ad64b46a76ef3933b0739c3 : SUCCESS in 21m 58s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/2c5cc1b37aa047578ddf70e6401c1212 : SUCCESS in 4m 20s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/5d82dbd851e9409abb83ef82d7aa753c : SUCCESS in 6m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/82957f1d1d664a15bee862e83cc001e6 : SUCCESS in 6m 16s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/61cf26489e1743f493f2559e2ebe7bd4 : SUCCESS in 13m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/628193803b55443ab6c60c91965a07ad : SUCCESS in 1h 02m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f94f5296ca73467ba34203e0a2c73c63 : SUCCESS in 1h 41m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3e72dee45779e7d9ac4793f91635523687136492","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-25 09:33:48.000000000","message":"Patch Set 1: -Code-Review Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e4ea4756148c851360ed2dcd08833313f6ee03a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-25 11:09:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2dd64fe70deb465eb944521df61d179c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/054ee799dc7d43739733a3dfca385fdc : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6503e341e85b45439fabe2f59bdca76b : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57f15791f9c6466ab3f93bb1165ca82a : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a83fc1cdfbad4f6f9daa5e8200f848f0 : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/893243d7b0f5489380d0cff9878259ae : FAILURE in 7m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f7346042aced43db984a0e8a98bc0ff4 : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2c6293c60d444b20a5ae5dc2faef6af1 : SUCCESS in 3m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/299bb55aa56c4fb68b33d2174a9fe54d : SUCCESS in 10m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2de741312c00438a84a23fc3866c1bc5 : SUCCESS in 19m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e1c51e2a0509493ebdc5a377df96b3e2 : SUCCESS in 9m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/55c5b10af2ab4266a5f3539c4278f551 : SUCCESS in 22m 58s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/c735cbed52ae45519c7ecc59ef7b8b5d : SUCCESS in 5m 08s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/aecad1af242848dc920dfbf9e07d0fc5 : SUCCESS in 4m 32s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/f58774b4afd04ce5b3375f0fe9141673 : SUCCESS in 4m 22s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/273128bf95664cc686bdd0bae86ebc0b : SUCCESS in 10m 19s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/66df2c22859641cbacd11bb031b309a0 : SUCCESS in 4m 27s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/cedd65bc4ea34163897caaf62ed96f62 : SUCCESS in 5m 58s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/72599a4f34ef4d4689a79754edc3c2ff : SUCCESS in 6m 24s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/11340e85cfbe4acfb59425ee88669127 : SUCCESS in 13m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/70472f491b874b7fbcec669903da2a01 : SUCCESS in 7m 27s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/f6b3b178d63a4a8cb6c9e84593f3c467 : SUCCESS in 11m 20s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/c6df38f3419a43a1adb31fb75a3d360b : SUCCESS in 10m 59s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/ca8357b41dd448df94c813908ae5b983 : SUCCESS in 5m 13s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/07f9c9677c5e46349593d22fe685f61b : SUCCESS in 8m 56s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/4a110bfab45941e8a9cb3fb17d232717 : SUCCESS in 4m 28s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/5bfe9a374acd4fc89800294a35148a24 : SUCCESS in 6m 40s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/8b1fa4d96e254547a96ee154bd529013 : SUCCESS in 10m 53s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/8354d60416204f24a75a83ac7d6942b4 : SUCCESS in 26m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5044e56879754b748cc463d08e5545e0 : SUCCESS in 6m 59s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/e6414d8f80e9498ebc2b37d4c7e06a22 : SUCCESS in 12m 56s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/768d7ab16f51497b82ecc00a675506ce : SUCCESS in 3m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9122fa116a6b420081ba515b674b862f : SUCCESS in 4m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a3092c18fdd84614bd5ee977b78455e5 : SUCCESS in 21m 37s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/417e57b0c99e4fc88b2e365fba394395 : SUCCESS in 4m 16s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/5a8f46dfcdde452d817ccde24cdbbd3a : SUCCESS in 6m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ef76ef7433fc4c8f96f9a074f5af8afb : SUCCESS in 6m 32s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/4a9397417eeb4af9b82a78724ddb97cd : SUCCESS in 14m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6465fcaa7b5043a2a8593930a5e0bfac : FAILURE in 1h 05m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2918820f0be948e7b66bf34813f23d4e : SUCCESS in 1h 32m 46s","accounts_in_message":[],"_revision_number":1},{"id":"2f8bb72b9011e9b0d1f1f073a0513a3695036911","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-07-02 01:28:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a76e74afc21cfda72afe203e15b8f4cd1dcea3a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 03:23:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a41aa738ff834bdf8a00f0adc19ea642\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaaed76fc979422fb5f589dd2cb281c5 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adcd87e926f44b67ba94027901acc681 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9e718edd2c84391aa05b213fd84aa0e : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/323cb98cd86d4a8f830dc22f62c4a63f : SUCCESS in 5m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0aabe651b4b74e2ab1dc6a18696e6e3b : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/13f909829a3445f1807192eac1dbcc55 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8da9f2028f624090ac5de95cbb068ed0 : SUCCESS in 3m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/89b7277bb73c42d18542cc82b147ee62 : SUCCESS in 15m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/91ba3b34abdc4902b5f1e13fc196558d : SUCCESS in 17m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b161933eb562435aaf8373c8b97751ee : SUCCESS in 10m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9a75ef5aba2d4069942831fd1e327fb9 : SUCCESS in 16m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/35377f152570404c82ff58d110ebfe72 : SUCCESS in 27m 45s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/34ae1f4e9ee94b0687ea115dc8fc76f9 : SUCCESS in 5m 14s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/d6d1f6a73f76470fb452bf61cc8eb955 : SUCCESS in 4m 50s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/33d39c18bb9e422890bd30076152525f : SUCCESS in 4m 44s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/cbc1dbd3f78c4d7d98f7bc803e271cca : SUCCESS in 10m 26s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/2bc9922c5d11411baadb41d509c69d38 : SUCCESS in 4m 13s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/608db79e6e984127b3eef73e70f1d68c : SUCCESS in 4m 48s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/c86c0ab3bb934febaf40a4a41867b89e : SUCCESS in 6m 11s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/854567d99ee149e0ba6dcfeb4f21b898 : SUCCESS in 13m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5d140b80c3c541b9876b318d4ece4866 : SUCCESS in 5m 56s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/522879a9154a403ca82366523daf5289 : SUCCESS in 7m 55s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/76c79092b93e4c65a7acecb66e63596c : SUCCESS in 8m 47s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/1548ac6dd68e4e48bb4010a3def20466 : SUCCESS in 3m 20s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/9c6cbaef8ffc4de8b3e5225dd2c6719d : SUCCESS in 9m 45s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/4bededc420534edea839cc92e28c3ed6 : SUCCESS in 5m 24s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/789b4d64be054d059662203b1e676f28 : SUCCESS in 5m 15s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/595e56a618074650868657d781962bc0 : SUCCESS in 12m 19s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/4095c3f82a164d46b66cc552bac23c11 : SUCCESS in 16m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a30e21fd75664eb784d1535bc3e395ef : SUCCESS in 7m 52s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/c6045905138347b0b21f6e2d9976ac43 : SUCCESS in 12m 51s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/2b0b8270c0034e92952fc3d5278447af : SUCCESS in 5m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e9742d80db234abdaf889d256eac500d : SUCCESS in 5m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/444af4c19ec04f9a86e9eca439106047 : SUCCESS in 22m 15s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/f69b466dc0734e528856413e4bb6a037 : SUCCESS in 4m 00s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/9efbe23d980646e5adbdcf3b8644d3c7 : SUCCESS in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7212aea4bbcf4c1c8ad802c144af3371 : SUCCESS in 7m 07s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/e2bacb2194174f59870c3b2ab54255ff : SUCCESS in 14m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82a23281b6ba48cd9e5b84eee3450b6f : SUCCESS in 1h 05m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8acda80f50b844c8a76883f3173f1c10 : SUCCESS in 1h 49m 23s","accounts_in_message":[],"_revision_number":1},{"id":"21cd8c410551c39132d396a5c7f5252a843061bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-06 18:16:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1, Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"55c3b28b8575720ea49567f415e6bf035ae9e2b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-06 20:03:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67471df3bec846c3aba0996438e6da24\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f86d12921ef54970b36077bea1d02dc9 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1363e8f8e0bd4b8b92a87f7c7c98cb4d : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e8e6dee37e44c7b84fd53dfa0cc20ed : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff81374f23d04d7dae672f2f402f686a : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b310be6b2fb47ac9f7f63aa11e3061c : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f906628531ac43eda22cb75dc1c2cdfe : SUCCESS in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dfc71b07c9af48418a70db46793944f4 : SUCCESS in 2m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/89dbd13f8ef54fb880f782969085374a : SUCCESS in 7m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/94319ee9fbe6441181e6df0c556069cc : SUCCESS in 14m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/aa5023746fbe4c4c8f8b957cbab69e85 : SUCCESS in 7m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/82ad2552cdee4f09a389bb635cbe0be9 : SUCCESS in 15m 52s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2428d3490dd541c18bc53c6bd6e99cf0 : SUCCESS in 30m 33s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/b569ea7d743f4c399d5c90efa771be9e : SUCCESS in 4m 53s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/d4a081b76b984a4dbbb6cf707cb3d0f9 : SUCCESS in 4m 59s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/cc7ebf669424464383d8a122c1f9d515 : SUCCESS in 3m 31s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/5cd4d4e81a56487784d311a9c0fed1a9 : SUCCESS in 10m 23s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/581bb5dfa0c94a7b8f0771bd0f0cc77b : SUCCESS in 4m 41s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/55345b4bdf674e5d8ee4ecd1b240d95a : SUCCESS in 5m 57s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/9a777a4e653848a58386778c30f7a71e : SUCCESS in 6m 38s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/fa2d5199eeaa471eb319f817277bf10c : SUCCESS in 13m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2959cfef07fc42e19d135b019fc809e7 : SUCCESS in 7m 21s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/906e25140c0e488b844760a174404c38 : SUCCESS in 8m 03s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/984a23303d9f46858f0ca2ff9c81d11c : SUCCESS in 10m 36s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/d5bf46f11d654869aed5cf9b6a7188dc : SUCCESS in 3m 42s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b12b25344ee648e28b306b2721d1222f : SUCCESS in 9m 16s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/1582a21438084e51b4bca70df57fd890 : SUCCESS in 4m 10s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/00ea09f8ecd447459a90021d0d9e1721 : FAILURE in 4m 35s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/83b5642e47d54294a8fdd6dc583b4a77 : SUCCESS in 10m 08s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/efc15a900f074a55958f117584c3d19b : SUCCESS in 17m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f12861e8928f473280f88699aef02cb3 : SUCCESS in 6m 35s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/481a29e2f7c14db482b00fa9ad49f8f4 : SUCCESS in 11m 29s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/bfdee784226f491b967e6a7f3ab33ea1 : SUCCESS in 5m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ca8319e18ab24f008cf5efef4ebe9fa9 : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/84fb8fc4dc1a4c76b90b8bf9191052b4 : SUCCESS in 18m 13s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/0a8d3f756caa42cfb0c9aa5f0286f41b : SUCCESS in 4m 10s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c8ab625ca3f04acb8540f08ff836794f : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4d24963ce1604476a03715cbde09db07 : SUCCESS in 5m 45s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/da9e862cea2b4e75ad33c3e40e4110a9 : SUCCESS in 11m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9dc3522732e0436babf01b6bcbc1711d : SUCCESS in 54m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f0a8f39b173473187c64dd193ed7f6a : SUCCESS in 1h 35m 59s","accounts_in_message":[],"_revision_number":2},{"id":"286e1547f6e0197efe8a69b992aa81ea80b77a33","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-08 04:41:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d4e94d9378f9b6c625dc6b8f4b6f12e2c7826720","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 04:41:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9dbb7f73c1a8164e34ac6afdfe0ebc17b3e35e9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 06:26:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bf51c9042d242c9b3ef67bcfde81a3a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5420a5acfe4844258d0514adc1933a5c : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a8f69b695f044b78ae23af4f2e62d71 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bcc48262af04c6dbca394114ee6ad48 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/77a989b95da54ae9a531285b4ed39680 : SUCCESS in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/34d03f7421134a44a9e7ed25702a3429 : SUCCESS in 2m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/de89bcd96b2d4f6bae0aed03a1a48375 : SUCCESS in 2m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/192b895062f24e798aecac38655069ac : SUCCESS in 7m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4c55d42d760d412bb447aebd5d584d40 : SUCCESS in 13m 33s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7538c5e036cf4ce18bdb7485e4ae754f : SUCCESS in 7m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bc94d43c43ad4017b0421cb316ef5e52 : SUCCESS in 15m 15s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/5018f2fa881f4981bbe95e5f998577db : SUCCESS in 11m 16s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/207e4b40a7ca458286d506053a00c4ab : SUCCESS in 4m 17s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/6675e2db62bf43a49ddd7654bee5f8cb : SUCCESS in 4m 46s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/4a2b2bef428542d198fdb8d85b5d06d5 : SUCCESS in 9m 33s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/3dec01e5d0784cd8bb922bef819e586e : SUCCESS in 11m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/17a52a74ceb144ceb0bc6c3baaa752cd : SUCCESS in 7m 21s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/a04a873ce06c43e18525eeb32d7224b2 : SUCCESS in 7m 37s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/d1328b4b5cee40c8a5dd7c63542907cb : SUCCESS in 11m 08s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/58bce0c0728f4eeab4ef1016779fae13 : SUCCESS in 3m 46s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/1f821d4917834bb3867066938554bbaa : SUCCESS in 8m 46s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/15dca892c2bb4d87bd3e056aa0de4387 : SUCCESS in 17m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/48d9a28dc4ac4d43a74e4a57da802cc1 : SUCCESS in 7m 48s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/40d54dcd11ad4610a6b3cca054f05b0c : SUCCESS in 13m 47s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/0aa9067eefec4d1eaac799c1d3c08cc2 : SUCCESS in 4m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/542a5f1c408a4505aa0fa948969e5c42 : SUCCESS in 4m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/703394f07f6d44f482eb3f173c670b67 : SUCCESS in 20m 22s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/f6217857157e4760bd4ab1d5c9eb53c9 : SUCCESS in 4m 32s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/4cda0492c3b6469d96a14afef487b337 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ae0d66d83a124eb39069def908e97078 : SUCCESS in 6m 03s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/63a10eece5bc404498d5047e69067815 : SUCCESS in 11m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45ec3e5063bb431388c47560c0e2608e : SUCCESS in 53m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3fe4216df0a4bca85a4bc19006043cf : SUCCESS in 1h 44m 15s","accounts_in_message":[],"_revision_number":2},{"id":"8cabb44ad8209dfed6a17da934a474d4a8572f6a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 06:26:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"02e1b1b377726d7e5b7bee51055e7c24e30f5d88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 06:27:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26668c5d30b14cc18a6cb9186c420685\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c93b3710c48f448db1b4027946b17bf2 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"05b9c17e559d071ef208cc1d2a5479e1494440a7","revisions":{"9ba38b9d5975985d181e1dce4a45795ff91f6e37":{"kind":"REWORK","_number":1,"created":"2024-05-23 08:31:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/62/920262/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/920262/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/920262/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/920262/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/920262/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/920262/1"}}},"commit":{"parents":[{"commit":"8aaccc37bf096cf6ea19f0a4a1a67ed91611cf1e","subject":"Merge \"update constraint for python-glanceclient to new release 4.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8aaccc37bf096cf6ea19f0a4a1a67ed91611cf1e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-23 08:31:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-23 08:31:38.000000000","tz":0},"subject":"update constraint for oslo.versionedobjects to new release 3.4.0","message":"update constraint for oslo.versionedobjects to new release 3.4.0\n\nmeta: version: 3.4.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ibfa19d239efd696030ae630ea57555aa1eae9b40\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I50f19e4a1d95001be44880b9b36c5ba134278555\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ba38b9d5975985d181e1dce4a45795ff91f6e37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ba38b9d5975985d181e1dce4a45795ff91f6e37"}]},"branch":"refs/heads/master"},"05b9c17e559d071ef208cc1d2a5479e1494440a7":{"kind":"REWORK","_number":2,"created":"2024-07-06 18:16:03.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/62/920262/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/920262/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/920262/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/920262/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/920262/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/920262/2"}}},"commit":{"parents":[{"commit":"1201da60c81309d69dcb4551d8415d67d5e246c5","subject":"Merge \"update constraint for python-observabilityclient to new release 0.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1201da60c81309d69dcb4551d8415d67d5e246c5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-23 08:31:38.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-06 18:15:49.000000000","tz":120},"subject":"update constraint for oslo.versionedobjects to new release 3.4.0","message":"update constraint for oslo.versionedobjects to new release 3.4.0\n\nmeta: version: 3.4.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ibfa19d239efd696030ae630ea57555aa1eae9b40\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I50f19e4a1d95001be44880b9b36c5ba134278555\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05b9c17e559d071ef208cc1d2a5479e1494440a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05b9c17e559d071ef208cc1d2a5479e1494440a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
