)]}'
{"id":"openstack%2Frequirements~920283","triplet_id":"openstack%2Frequirements~master~I8b1654e58c42acb0e4ade0c5ef3ba518087560b7","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-05-25 16:24:06.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I8b1654e58c42acb0e4ade0c5ef3ba518087560b7","subject":"Updated from generate-constraints","status":"MERGED","created":"2024-05-23 14:02:10.000000000","updated":"2024-05-25 19:27:25.000000000","submitted":"2024-05-25 19:26:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920283","meta_rev_id":"7622f1edda6f521113b8a5b6f9923e7088cbe5d4","_number":920283,"virtual_id_number":920283,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-25 19:26:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-05-24 13:22:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-05-25 09:31:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-05-25 09:31:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-23 14:02:10.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-23 15:09:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-24 13:22:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"db713340754e7733d179f8ddb3022f0ac245477d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-23 14:02:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"074195ebfe685a9476e3f59fe279af3cf0637b95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 15:09:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8d280a0ae924660895d7dc32b505343\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dbfd3d614c941f5838ffdc8390d27ed : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3c775a9177b4c36a921d151bbe755a2 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51744ff64ed5487cb636c769580a8b92 : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b6238055fad4969bbfa314018e5625c : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ae0f98ba5c74e84998f3a0631d9af7e : FAILURE in 6m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e11d23074c64addb36f07573190a851 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77a72a46171b4dc686a7d338ff0d17c5 : SUCCESS in 3m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/24013c9a60f84700b35026bbff208660 : SUCCESS in 11m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3b784d1650364bb29d131f7f8eb4cdd0 : SUCCESS in 16m 22s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2387c8c5dbf744b58c4d2340f05f3a15 : SUCCESS in 15m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ee5fc790d8ba45128c2df0a18b5df7f6 : SUCCESS in 17m 44s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/132307f245424aa0baa43a4e6fb53285 : SUCCESS in 5m 55s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/e93b81c59fe74552bf0c290f28afb3a7 : SUCCESS in 4m 44s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/ef8472ac583f406cba5de6c0686e1ad0 : SUCCESS in 4m 32s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/01d6694df3334d69830e53cb33049373 : SUCCESS in 11m 57s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/09adb530f1674723824f074ff8d69894 : SUCCESS in 5m 10s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/ecdb08729bb1413aa7dc6d1cb1d0fd0d : SUCCESS in 4m 53s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/8474df6bd7c94fb18b3d348fe1c587cb : SUCCESS in 7m 29s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/94253648006e4441b9962c8121bf71d8 : SUCCESS in 12m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e96ce4fb8e51483f9a9994560dfc07f3 : SUCCESS in 8m 35s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/e888cf93a0ac499ab6cdfa9595771c7f : SUCCESS in 8m 32s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/54da24e16e59432ea9092c44fe470a2f : SUCCESS in 9m 29s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/4bf89bf0432f43ada8e84bb1b7fe08ad : SUCCESS in 3m 54s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/1238282dc357491086f9b44891bb7025 : SUCCESS in 7m 38s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/f96a20e722704203a4ce036d35be0715 : SUCCESS in 4m 11s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/44188bf2350446fabebe2a29f82baa49 : SUCCESS in 4m 42s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/cf1cc8b4d1d149288117350b86d5641d : SUCCESS in 13m 36s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/5e3b0ddff7dd41269514b10e009e43a0 : SUCCESS in 28m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8d98345f359e49fcb9c1f1865cd67d2f : SUCCESS in 6m 33s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/1b6f565a6acc4a6c8aebd708024ea04c : SUCCESS in 11m 07s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/3acbec76a4d94de7910e7ebdce64e7f3 : SUCCESS in 4m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/08b19da02ecd4781ab2b28fa7822d460 : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82999b894bb54fa3ac461d6ace4f1559 : SUCCESS in 23m 32s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/51f98d23425a480fabe0ea80f3628232 : SUCCESS in 4m 33s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/a5dd4f9378d04e1598a82ed36dbd7602 : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7598c489793943c4b378aa6a55037b77 : SUCCESS in 6m 01s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/3500e06bac144322a91d46d71f0806bf : SUCCESS in 12m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83dae985f08e4a6ba6aac42f6819ac2e : SUCCESS in 1h 04m 13s","accounts_in_message":[],"_revision_number":1},{"id":"4bff751c625151df88c60c3b76cd118e4654e2ee","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-23 15:33:17.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d7e69155fb0f4e111acf8ee330d3a27f8042e458","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-24 10:13:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e8a11226096fefe863a7f18bc01089d0c691930a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 11:19:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7530986fc81641acb94d1a01da38604d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/349337005bff4b15b30ecef28d1fd195 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47d03b4f12d748a8b25f4ef20b1505c2 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee336c94ebc04298a2f62a976f6bbe3b : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/722c051c2a604581a44f2ecdc4673c28 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ce8fe6085f749f0b10fc7e0cc24802a : FAILURE in 5m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b6cc59c3c53d49c89137b086dbefd498 : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ded96f9d896a4d31a6a0f738991f409a : SUCCESS in 3m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/001f148523334f8a9bcc0b31c402d3e1 : SUCCESS in 15m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/23acb9f5f0d84cd3abfccb805585f6f7 : SUCCESS in 18m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b3e50037b6474859998fbdbae231acf5 : SUCCESS in 8m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/92dda2a1d56c47edaceddf635b82dc71 : SUCCESS in 19m 03s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/6dac1a2fcc87408ab78946f287b4471d : SUCCESS in 4m 29s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/3f603eadcdcc43e981501efa621f9cb1 : SUCCESS in 5m 17s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/b12f3316f8aa4d99a7dfa2aa9cba39c0 : SUCCESS in 4m 20s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/8a35c441f3154a8dacd6e5531283a144 : SUCCESS in 10m 58s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/ce532ff48fc5433bb43031ec417ef604 : SUCCESS in 4m 59s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/68e0ddbd3b2f494ab2d6c0ba5b773164 : SUCCESS in 6m 47s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/438cf2e89f0845ec81a24f9a64f82b84 : SUCCESS in 7m 20s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/ff7852ceb3584a7585b771fbc4466af6 : SUCCESS in 11m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/81d6d98cb73f47af838cd30c5bf8ae80 : SUCCESS in 6m 54s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/6662ebc13f9d41b29162dab026dbf867 : SUCCESS in 9m 08s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/a39aae3c7c794d97a94d38ace5b0e845 : SUCCESS in 10m 31s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/43652625d5d649349bbd68ba87beb8d2 : SUCCESS in 3m 27s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/3f1cf4040cce43b3bdd61c421c64ddaa : SUCCESS in 10m 37s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/1c5df3a357f64daa9832fc9c08a59770 : SUCCESS in 6m 13s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/169026584d514b6cb0a62b7644584fc4 : SUCCESS in 4m 10s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/d4430beb628e403a80c5447e8dda9d23 : SUCCESS in 11m 03s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/9a3ceae9b6eb43cca7a4146843970263 : SUCCESS in 28m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/36cb90b55f384dd298204522c4d9d1ad : SUCCESS in 7m 16s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/df60e8dcde2b41ef8ce5d79acdb33499 : SUCCESS in 11m 08s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/7edb3e02dc5147e3860731bf04e01886 : SUCCESS in 4m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fe630e83ce1842afb18a76a45601573b : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4b9bb7791496468eaf98849198778138 : SUCCESS in 28m 04s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/72ec2a418fee470796b1b8cad602cc42 : SUCCESS in 4m 40s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/d15871fa2385436f8d63a4f97df14e44 : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fdd35d97cc3c463da95036428f3979b6 : SUCCESS in 5m 52s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/594ba218100f4cce9477daf6ce53b960 : SUCCESS in 14m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f72da31862b4a6bb93915e70602a782 : SUCCESS in 1h 00m 35s","accounts_in_message":[],"_revision_number":2},{"id":"cbd9978c2a2688b001343a69e6b910bf517396fb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-24 11:35:35.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba8ce233a5db63f90ec34d43e10af5a0018923cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-24 11:42:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c81d9939ee1a80b208c1aa04ea4a0d432b57fce8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-24 11:43:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7de628a1e675605656b8c154a9792f777c7b412a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 12:58:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44525cc5a8624df4a510050715a806aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c24be03d8dc7425aab98138077e6228f : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e1ec93c17e8439d805314348b3dbd09 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2a4b158a1eb4aefaeae9bc7778230f8 : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1334afc826d4d21b17003f73c0d54e8 : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2657e033498d48f080e6f467120368f7 : FAILURE in 4m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8d96196424a8410bb2600ca8b6e846ca : SUCCESS in 2m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cf5db5a256f0443a8b6db51bfaad8185 : SUCCESS in 3m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ee15273e45364908ad1541b4ccb96b49 : SUCCESS in 10m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/79f8328683d340d7b22a6a86267fbf20 : SUCCESS in 14m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e9ca9ea660ba4df2ae7db469e3df44bf : SUCCESS in 10m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ae44fb001a3b4bf49c881a0c7eacd4fe : SUCCESS in 18m 48s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/1b9cde59d91643a992a65cdb5f750d1f : SUCCESS in 4m 19s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/55bc308a05384aa3aedeac98a9c9bf96 : SUCCESS in 4m 13s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/b0b6b0ce4a424fc4b4c2323298878e7f : SUCCESS in 3m 38s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/ba9ba209f9fc41599eccdf68a69f79d5 : SUCCESS in 8m 47s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/dca4a5796eb5480f80a1ef0f67c169e6 : SUCCESS in 5m 15s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/cc9aac374fa24cbf8433eff4c2caedbf : SUCCESS in 5m 39s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/29771eda40004b79b0ec1f6651c88ed6 : SUCCESS in 8m 23s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/9cc7fcde195540308002fbe76386ce86 : SUCCESS in 11m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/80da736062394bf5ac0d54eaf4678d2a : SUCCESS in 7m 49s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/c55786d381ef4133b89dcc015ca2c445 : SUCCESS in 7m 31s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/bef651ff0d674d20add1fdb8bff5d5d7 : SUCCESS in 11m 32s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/6941898341a544eca12af6950a58a0bd : SUCCESS in 4m 04s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/5cf9e7438b4d4b2b85559999d3673b7c : SUCCESS in 7m 55s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/e26f23c5a74c4c6ba4806219c29e114d : SUCCESS in 5m 51s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/ef8bc596bee346bfbc6547d48695f345 : SUCCESS in 4m 33s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/910b3e1ce0ad40c386c78b474e952ca5 : SUCCESS in 10m 24s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/316a53d1e211404bb5435aad0ef91fd5 : SUCCESS in 26m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7b6a96822a28467aa155e2ca19037788 : SUCCESS in 6m 18s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/6b75ed31612c48d48ab279a34fde3bed : SUCCESS in 9m 31s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/da50df228b4b4ce9ab577978c2d9db46 : SUCCESS in 4m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d1caa43b56fd4aecad1e2699df649771 : SUCCESS in 2m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a9abd810b93b460aa68af4508c4c9f5a : SUCCESS in 24m 40s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/4d3e0f69bb2e4127a609d213ba0de6ec : SUCCESS in 3m 39s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/bb8fb73cb0a44a22957930a9832a2a9b : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5568bf50d76a4fa681172e4217eac9de : SUCCESS in 7m 03s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/1ce0bb84015046b68a64eda06f2bda52 : SUCCESS in 13m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b96b1e1b217045ed82b7fad9f81e8f03 : SUCCESS in 46m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f7a856c872e474b86b9d947f7749b73 : SUCCESS in 1h 13m 52s","accounts_in_message":[],"_revision_number":3},{"id":"6a69e412b54dcc4beba44e015df30cdb81f26126","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-05-24 13:22:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"63a2c87bb8350849a1fffc78124c7946d6b3a1e7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-25 09:31:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"87af55d148b2c3d142946d13a6594263806e041b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-25 09:32:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6ec38a179b5efdba5ca7efc9a8424c38b32766c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-25 11:07:08.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a33960eb8a494774a6d30f476d9ef734\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0aad84acc704f4586a5d6c171ac9a49 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/badb1fcefd7640149b18cfa31f6d4e19 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78e978170e7b43229ff485bcb5110e52 : SUCCESS in 6m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ebefaf99bb264083ae91330e04f3cd97 : SUCCESS in 3m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b6784ebaaa9d47ac92d8477b79f6a30c : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5c96458f33b7496bb015709f341f25f1 : SUCCESS in 3m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/89ab7e073da1489cb00f01cab0066b3b : SUCCESS in 10m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/89acb07d14a24ac5b1f5113cd52293fe : SUCCESS in 16m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/02dda0cd3c16417e8d158b119b037607 : SUCCESS in 11m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0640cabb101f47638d8165f83653af28 : SUCCESS in 18m 24s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/b3d95989ed3f4e1b92f43684a257fcd5 : FAILURE in 11m 17s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/bf442914f3d64677888bad27b019c403 : SUCCESS in 4m 49s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/299690f45af24f1691bc48d8ec5f36d6 : SUCCESS in 5m 08s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/b74e674ae2be4886a30f2cba00440a06 : SUCCESS in 8m 34s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/8791fa119fe04d02be6cd928c82047db : SUCCESS in 12m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/061f4e77592e4ea8afb7826ae48e25ea : SUCCESS in 7m 48s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/4e0375ebb11e4e6f9acb84701cc0402d : SUCCESS in 8m 02s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/c3f574353c2845af9565d027d85b0176 : SUCCESS in 10m 17s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/ad52766372f644ecb32adced456369fd : SUCCESS in 3m 57s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/fc26d2ea24b34eeab229f7ace1731ffd : SUCCESS in 8m 45s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/c1d790bfdd5b4a508e73636e56ce390c : SUCCESS in 25m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b864bdb55a7047d58d646fa9b916e010 : SUCCESS in 8m 05s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/9a313abc0ac1454e9c6e262733d17561 : SUCCESS in 11m 26s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/6f5fded03fb147daa6731c72b901d11e : SUCCESS in 4m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0b9af85559684cfabea06d70a558a556 : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c6538e0234f04b5ca0cd9f4df11e6c85 : SUCCESS in 21m 24s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/7b1b1c04f8e044379f594b4b461b0d04 : SUCCESS in 5m 41s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c00b7175b07744caa795e1cdcdb8b0a6 : SUCCESS in 6m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/523eaf1e534e44bd8f13a061b76beb18 : SUCCESS in 5m 59s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/ffdb3279da464d8e803093224dee0144 : SUCCESS in 13m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48150f1d34084eabacd8df2b54746cbe : SUCCESS in 50m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05d184ad586f42be86c0973cc3a2379d : SUCCESS in 1h 32m 14s","accounts_in_message":[],"_revision_number":3},{"id":"d9a18cdb71f99fe69979a41b61d6810e10f19f88","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-25 16:24:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"96a75666aea8c146b3b681afe3d3fe239d08c591","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-25 17:34:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a0290c7468a48c4b6d1b16bed65fd0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28d6e95bd52546648d8a77393cd17423 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7d05d0a5dda4d568978b70c6a72c766 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e353bb15ae44889a2f1cd5eb6a53aec : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a365e10ee98a4d57b6dae4e89a55ce83 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f10b0393024e44ab86f0d6a8e191159c : FAILURE in 5m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a73a23506b184b649cb186271191778f : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1868a5c36cfd43ae80b02bb04887f4c7 : SUCCESS in 2m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f49d89f406734005b349fab50b44405f : SUCCESS in 11m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c8b117d7fe6e4af1927ee6a4b0e7d0d8 : SUCCESS in 16m 20s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/774110492cf2470e8c6ae9806b4b9deb : SUCCESS in 10m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/935dd00acc8a444aaa7d39716335f2a4 : SUCCESS in 17m 02s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/81da6b38cd12461a86e2c84b061a0c78 : SUCCESS in 4m 40s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/3dc44f44e9ac467f8c1a0fc2ea756dc1 : SUCCESS in 5m 07s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/1a0a63caeca64018ae17f51e1cf1b6aa : SUCCESS in 3m 51s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/9c75fe418852431ba203a232a6697f5f : SUCCESS in 9m 16s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/30ae2fcc6649475fb3b4802119d8f50b : SUCCESS in 5m 49s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/d201f48d77f54a29a197402a51f23254 : SUCCESS in 5m 12s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/53e527baac134959bb7a6e35ab0efe25 : SUCCESS in 8m 49s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/0f5f18953da8407a8fe11dd156f95ec5 : SUCCESS in 9m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f107c3bdfe9e46c5a40d60da0288c61b : SUCCESS in 9m 02s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/d28a6ba562cb40c9a98320f5b5e33eb6 : SUCCESS in 8m 13s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/bfe21be6499a4ce4b483ca822d9a1793 : SUCCESS in 10m 03s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/42345bc8794b40f68c92266ec49eb1f2 : SUCCESS in 3m 48s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/616e0491b2374490978a6ed1f52c8ace : SUCCESS in 9m 03s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/15e84afd6a3345fa9a13dfeaf323cbe9 : SUCCESS in 4m 27s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/a4bf76852c564d749e9535b1d675d10e : SUCCESS in 4m 08s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/8532d05f24f74ecfb613b8270d010de1 : SUCCESS in 11m 25s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/a5aae249ab6849239c97a4127bd72ceb : SUCCESS in 24m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1b3ad254312d4210a1920c8fefaf406d : SUCCESS in 6m 30s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/ea73e0c117a84f7a9c3f417490bcc833 : SUCCESS in 11m 28s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/751c7ff0cbe1466a9e68e1d1337766e5 : SUCCESS in 4m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b1a51110b9714f14839beabd1545aa8e : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e2cacec296e488ab28c3b9e3aad74a2 : SUCCESS in 17m 52s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/374ac1f18dad4e119b43f797974bf882 : SUCCESS in 3m 28s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/9f50c351cc3d49528381672768eba065 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e1063c7011e49e9b6e39acc47a07f0b : SUCCESS in 5m 43s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/3b4a2bc8a8bc4897ba8271258cae890d : SUCCESS in 14m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89cb809d26a24339bfc73f279d7bd07e : SUCCESS in 46m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3e1803d5f3e4a678fb1df75cf2f8cd9 : SUCCESS in 1h 09m 54s","accounts_in_message":[],"_revision_number":3},{"id":"dfe9be263afe7e2055f4f8c2556a0ccb7da90755","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-25 17:35:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b05cd7f5a47a8a1b9309d2535e1a21b5e79d26c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-25 19:26:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/665ac5d016b145aa958157f3eb97a2c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bec0fafab1d415f9bd66a7988bf55c0 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b82ba2dca3f4590bc9451d1dc4269a3 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84aed701d086440bbbfada663f2ad64f : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45e62fcb911644cca06ea657b7e47b0a : SUCCESS in 3m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cec138d5373041ea8bbf6135baaa6d79 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8c3a7d42d0084822bb13535f50a8f743 : SUCCESS in 2m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2417b9b2b5bf42f7b753a591093d94bb : SUCCESS in 7m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f54bce783c794d31a4ca932323335c8b : SUCCESS in 10m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/baff88454c014b2897e70abc5160feb0 : SUCCESS in 11m 23s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/62422fc5a3464e4fa138782061aa64c4 : SUCCESS in 19m 36s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/6c4048b8a8034d058fdb42459b8a5e5d : SUCCESS in 10m 18s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/b458fb54887b4524becb46caa55f4321 : SUCCESS in 4m 47s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/d693760355de4c7987e80785f3043c07 : SUCCESS in 5m 04s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/be00899ec36c4d96825418ac39424f4c : SUCCESS in 5m 40s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/cbcb5682effa4b24a5f9c20b7f317603 : SUCCESS in 12m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a6b5b9d8a0f44aeb4f0b93a11de7a97 : SUCCESS in 6m 02s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/11389768adc3480c820bd54438d8117d : SUCCESS in 7m 42s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/d0f28e78e4174f8eb128ec56e12ead38 : SUCCESS in 7m 34s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/51859b437ed24e0ea40242bd5be1df3e : SUCCESS in 3m 11s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/5d372f1c4cd845abb574398ab1971b94 : SUCCESS in 9m 04s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/71f579cb711e47ffbc95eafb8ba51a66 : SUCCESS in 24m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2ac6073c9f204d11a67d4f5a30e8d685 : SUCCESS in 6m 02s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/8d86548b54864fdd9c7040f76a42482c : SUCCESS in 9m 29s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/99505793ebac4484b5b241fc88787de2 : SUCCESS in 3m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/32c9ae3a180f4ec5bc037f00278fd717 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/14530c9ce4db40e3985c3f6061b272d2 : SUCCESS in 23m 10s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/ad2eded9eb434f8a8f15c7bf92bdcfdf : SUCCESS in 4m 31s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/4ff7e2218e6a476b9627460f26f503ab : SUCCESS in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ee6933dcd83d4346bba5c696ba031623 : SUCCESS in 4m 56s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/4e0a6096974d465188d62148e1315e2e : SUCCESS in 8m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e9e5de1918343a9a49551d2039aab89 : SUCCESS in 48m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/321af34751414bcfab3589262b172a95 : SUCCESS in 1h 45m 43s","accounts_in_message":[],"_revision_number":3},{"id":"6233ea460dbdda304f341f6e084aa9d68845750d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-25 19:26:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7622f1edda6f521113b8a5b6f9923e7088cbe5d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-25 19:27:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/037888476af8424ab96d303484685098\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43a3e15113d7474db2a7ddcf0edbea31 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"84ef92ddf50a5ff4fbdc2f48f69446d8ad6cbc41","revisions":{"98b94a030c66bf7288c113f314d035f0fd21d203":{"kind":"REWORK","_number":1,"created":"2024-05-23 14:02:10.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/83/920283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/920283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/920283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/920283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/920283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/920283/1"}}},"commit":{"parents":[{"commit":"26e59920a7bbdc47118845704322dcd32510a2d9","subject":"Remove ansible-runner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26e59920a7bbdc47118845704322dcd32510a2d9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-18 08:12:18.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-05-23 14:01:08.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nBot-generated patch with added fixes\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- jsonschema\n- pytest\n\nChange-Id: I8b1654e58c42acb0e4ade0c5ef3ba518087560b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98b94a030c66bf7288c113f314d035f0fd21d203"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98b94a030c66bf7288c113f314d035f0fd21d203"}]},"branch":"refs/heads/master"},"a1f939354315c7e3fa787df763233e151fee199a":{"kind":"REWORK","_number":2,"created":"2024-05-24 10:13:06.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/83/920283/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/920283/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/920283/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/920283/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/920283/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/920283/2"}}},"commit":{"parents":[{"commit":"030937ba889aa862dc3b0a2827caad620fba3ef0","subject":"Merge \"update constraint for oslo.db to new release 15.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/030937ba889aa862dc3b0a2827caad620fba3ef0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-18 08:12:18.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-05-24 10:12:49.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nBot-generated patch with added fixes\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- jsonschema\n- pytest\n\nChange-Id: I8b1654e58c42acb0e4ade0c5ef3ba518087560b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1f939354315c7e3fa787df763233e151fee199a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1f939354315c7e3fa787df763233e151fee199a"}]},"branch":"refs/heads/master"},"84ef92ddf50a5ff4fbdc2f48f69446d8ad6cbc41":{"kind":"REWORK","_number":3,"created":"2024-05-24 11:42:35.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/83/920283/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/920283/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/920283/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/920283/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/920283/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/920283/3"}}},"commit":{"parents":[{"commit":"030937ba889aa862dc3b0a2827caad620fba3ef0","subject":"Merge \"update constraint for oslo.db to new release 15.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/030937ba889aa862dc3b0a2827caad620fba3ef0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-18 08:12:18.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-05-24 11:42:21.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nBot-generated patch with added fixes\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- sphinxcontrib-htmlhelp\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- sphinxcontrib-serializinghtml\n- pkgutil_resolve_name\n- autobahn\n- importlib-metadata\n- pysaml2\n- numpy\n- Sphinx\n- tomli\n- backports.zoneinfo\n- sphinxcontrib-qthelp\n- sphinxcontrib-devhelp\n- sphinxcontrib-applehelp\n- scikit-learn\n\nPin versions for packages that are not ready for updating:\n\n- netaddr\n- pillow\n- pysnmp-lextudio\n- requests-oauthlib\n- importlib-metadata (\u003c7)\n- jsonschema\n- pytest\n\nChange-Id: I8b1654e58c42acb0e4ade0c5ef3ba518087560b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84ef92ddf50a5ff4fbdc2f48f69446d8ad6cbc41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84ef92ddf50a5ff4fbdc2f48f69446d8ad6cbc41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
