)]}'
{"id":"openstack%2Frequirements~920285","triplet_id":"openstack%2Frequirements~stable%2F2024.1~I617b2d73054e2b2149ea1b0095ec9d08c0e4deea","project":"openstack/requirements","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-06-19 03:24:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I617b2d73054e2b2149ea1b0095ec9d08c0e4deea","subject":"update constraint for tooz to new release 6.2.0","status":"MERGED","created":"2024-05-23 14:12:36.000000000","updated":"2024-06-19 03:24:06.000000000","submitted":"2024-06-19 03:24:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920285","meta_rev_id":"c8a1aab36e6d8501ec3db0d5c31e777c19999333","_number":920285,"virtual_id_number":920285,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-19 03:24:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-06-19 01:33:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2024-06-19 01:33:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-23 14:12:36.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-23 15:59:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-19 01:33:16.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"b407af77868e5039601baca47387efe8d917300b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-23 14:12:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2226f6ff010864f8058bda43300063b785aa016f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 15:59:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/616c6fdd486a4668ba7ab93343c15862\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4afad4871d943f0a40aab992d4ccd3e : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04951ac8485c435b8e2bc3b2830ca42b : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/307879bd6ab44d3a8618abf21cbb12ee : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e3da2bdc6104df2b0054eecc48acd7a : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b6652848150b418a9e1212803c8697f9 : SUCCESS in 4m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6a137abd54fc446db3dfa029ed59835f : SUCCESS in 9m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1758fb360e864a55bbc37ded649166cc : SUCCESS in 15m 46s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f23416f2e07d40bcb7c8e0f3dd0a3c92 : SUCCESS in 12m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/49f37e312e584c32b802de0e53d08d52 : SUCCESS in 18m 40s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/ad6dcc612ea3459a839ff13cf998e80f : SUCCESS in 4m 34s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/275603afe31d4509b01fe3c539c319fc : SUCCESS in 6m 00s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/a37143fa3fc643b29154b359feddaac4 : SUCCESS in 3m 35s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/654da1acc4d44250bc667a5563cbb77d : SUCCESS in 17m 09s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/0930043ac1ea480d8a1ec264e2b4aca3 : SUCCESS in 4m 55s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/d88715acd9ed4aa983ddd1983c0ca5dd : SUCCESS in 5m 31s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/49bcff5daf084533b1639ba00c928841 : SUCCESS in 9m 34s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/d2d457098ab94db186b463dd14a8c159 : SUCCESS in 10m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1dba395685e74bb3b675c8f0bdbb837a : SUCCESS in 8m 47s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/367334bed759415281747a62b19476a4 : SUCCESS in 8m 03s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/b84d7bacf6ac44f89dd318a0c53bbd8c : SUCCESS in 10m 23s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/9f5692996b8843f7b07377c3e8b94c8f : SUCCESS in 3m 47s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/f32fc91603534fcc9938d47d9134bf75 : SUCCESS in 10m 21s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/59981b64c30c4d82a5e1a4aade1da1f6 : SUCCESS in 3m 52s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/09cabd597ba849d9869c937b9f58873f : SUCCESS in 4m 15s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/a3b30e3d312f4d918c7e63495023f0f6 : SUCCESS in 11m 49s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/49ac039c3e094b108719b7f655404f42 : SUCCESS in 21m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ca38cdc2dd98412fb67b1fbf1f42bccd : SUCCESS in 8m 29s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/da9f9a3420c84d9fadd5ce28a0fe9e75 : SUCCESS in 11m 44s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/d8ed9f01a9c24ac1827995922d3f4d9c : SUCCESS in 5m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/73b9c2ca260a408dbeb8681e2c0f598a : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a405b4e81ee45a894491face16d51ca : SUCCESS in 22m 23s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/4ad231c8ed984a3dba93cab8eaacafca : SUCCESS in 4m 29s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/977bc75488dc4eec9bfdc471a559fdfa : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8083926adcfd4e018cc2a5c596734d07 : SUCCESS in 6m 26s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/7b5bfa8a48e04ad08d0929bccd628815 : SUCCESS in 14m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50bbfa1880474163a54a9b9a92f6ae0e : SUCCESS in 56m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f70dedfe555416a826cb7365b7a2250 : SUCCESS in 1h 43m 36s","accounts_in_message":[],"_revision_number":1},{"id":"7fb7599d89a716756a67e262916010cf15bedd25","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-06-19 01:33:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"94b5eb8cd0b1a4dd40275a3bf16151c9fe1fec8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 01:33:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d4fa0921eca1019ab07a978b341186f36baf5185","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 03:24:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/362b3b98a5a94e25a4c47e14c1d390d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fae71ab48d24c75a58f28a6cf8c9af0 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1cc92207ffd4c939c0b38673189b873 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be7b6502f2a24ceabe5f035011459ab4 : SUCCESS in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1a43df0e4b74328a767f8b61032192b : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ecd783d827e34adcae65043df9399f10 : SUCCESS in 4m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/87a79b99394e48feabb3fcc1b400110c : SUCCESS in 10m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1dff9ed961af41518a2bb4af19e2b15e : SUCCESS in 17m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/31df8e728cd94f0899112912b5b83943 : SUCCESS in 10m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0c83468289954a1cb68214c5ed883e0e : SUCCESS in 17m 33s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/bdc39f3a6c6d49d5ac4c0edcffccd9cf : SUCCESS in 11m 08s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/3145cd8354774e83a153f9558cc9b2e0 : SUCCESS in 6m 25s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/dc691e4c76a9463495245975700b1937 : SUCCESS in 5m 56s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/fa0bed9090694044a757de08985ed8f9 : SUCCESS in 7m 08s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/115319804de04238bcc6ba54ca4f153f : SUCCESS in 11m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/789cad022c8f4fbc998d158b2ac8759e : SUCCESS in 7m 59s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/7c49d23a20af4b879307c8df8000046a : SUCCESS in 7m 31s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/bf85e616a87345f7be5d9a751b7f3b66 : SUCCESS in 10m 34s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/f18f54a7dff64ab899a79790be570659 : SUCCESS in 4m 53s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/2cd258a78137477ca03df43104166865 : SUCCESS in 8m 24s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/c1e4440a78934af7817a59621b3223f8 : SUCCESS in 28m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/35d8cbd4b351444ab4bdc1fc436f96b2 : SUCCESS in 7m 04s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/26365d0c46b54b8e9bf654c33856d39e : SUCCESS in 11m 38s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/2cacf3f00e9b4db6800b3f78d8fae423 : SUCCESS in 4m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3b6c5b8eb0ba413cb7722e747e6501d1 : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b8c8717fc57d4197951c88b8303480be : SUCCESS in 21m 15s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/57368d520ac345a1bcbf93cbd0ddfa96 : SUCCESS in 4m 40s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/12cccb682ba9405e949498f3606c65d4 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ccf6a94047e641f8b20535162e5b1685 : SUCCESS in 6m 01s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/14390d86e60c4577b202935b28d61405 : SUCCESS in 15m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f13a9d04c44946b696b7930be135bfb2 : SUCCESS in 1h 08m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80523e3ee19d4666abc499484a657271 : SUCCESS in 1h 30m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c8a1aab36e6d8501ec3db0d5c31e777c19999333","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 03:24:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c2b1961719d65165c25bc3bf023e07c0b65111b","revisions":{"9c2b1961719d65165c25bc3bf023e07c0b65111b":{"kind":"REWORK","_number":1,"created":"2024-05-23 14:12:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/920285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/920285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/920285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/920285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/920285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/920285/1"}}},"commit":{"parents":[{"commit":"5c68165572da1c438fb98533fcb4f609f9e8d6b2","subject":"update constraint for oslo.log to new release 5.5.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c68165572da1c438fb98533fcb4f609f9e8d6b2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-05-21 09:43:01.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-23 14:12:18.000000000","tz":540},"subject":"update constraint for tooz to new release 6.2.0","message":"update constraint for tooz to new release 6.2.0\n\nmeta: version: 6.2.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: Ie48b92ca25626de3ee00b0dcff755c2d8fe2afff\nmeta: release:Code-Review+2: Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I617b2d73054e2b2149ea1b0095ec9d08c0e4deea\n(cherry picked from commit 30e6988eba003895d9deb35750723cf9b61891d1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c2b1961719d65165c25bc3bf023e07c0b65111b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c2b1961719d65165c25bc3bf023e07c0b65111b"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
