)]}'
{"id":"openstack%2Frequirements~921968","triplet_id":"openstack%2Frequirements~master~I0ea11984e63b885668ed5842e1c6221739e18343","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2024-06-14 12:01:42.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-06-14 04:52:32.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I0ea11984e63b885668ed5842e1c6221739e18343","subject":"Update cross-horizon-npm job","status":"MERGED","created":"2024-06-13 14:50:13.000000000","updated":"2024-06-14 12:04:31.000000000","submitted":"2024-06-14 12:01:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921968","meta_rev_id":"55c8fa660d8045d673c7d8430aca5723bec546f1","_number":921968,"virtual_id_number":921968,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-14 12:01:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-06-14 10:15:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-06-14 10:15:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-13 14:51:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-06-13 16:44:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-14 10:15:33.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2024-06-14 10:15:33.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"}],"messages":[{"id":"9f04e1059ccd08a0d7e4c8db3c3581f13e3f17a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-13 14:50:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7eb06ea6852665761e83d804be9083a80baceee5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-06-13 14:51:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"319c3845fab9884b308a3870832519d389bf05fb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-13 15:01:16.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"34b9b7bd1c956b835f65d4662fd146e884595822","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 16:44:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/623c0cb82f214a66af7fe970a1900307\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75036c1aa4c0422982d087433e014683 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/704d27e04230404f824747d626aa408d : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb3f4c134c4640f8a3852826e4de1b2d : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32c17bbc49d540e5869835eacbd7c22e : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45f2c2ac34db441eac61769c680498a1 : FAILURE in 7m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d3ed05d5e4a4b52ac6755d8c0d658f1 : SUCCESS in 2m 49s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/359420a03a9247619f0ba648292a3e14 : SUCCESS in 4m 25s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/7832985c7c3a4bc6a83b9678025a9226 : SUCCESS in 5m 01s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/aa02a172f8734443ba032030080dec1e : SUCCESS in 5m 12s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/9792d66a80404709bbd7f59290dd50f6 : SUCCESS in 10m 59s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/c76258abd4654590b97ecc3156d5c855 : SUCCESS in 5m 30s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/2d7574a75ee34769bb5f493c58e85108 : SUCCESS in 5m 29s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/418a1d50b7dd4684b9e9ebc3a9324054 : SUCCESS in 6m 59s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/a660ea7f68a14834a52799b9ce49bbd4 : SUCCESS in 13m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cef84785e3b944e2985dcd4e9d42368a : FAILURE in 7m 22s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/694ef6781adc4eeeb53415fdaeb5df4a : SUCCESS in 10m 42s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/1fc730ad311c45519fed7f9aec489203 : SUCCESS in 12m 36s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/b9df9a582f4748f382315faaaaeca628 : SUCCESS in 4m 46s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/e5e4fb68603641c288e76de9a9affd2c : SUCCESS in 8m 36s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/b1327a9147444ad9b64b7c37c148803f : SUCCESS in 4m 49s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/48d6b2c501b844bd911bf8a794b68e60 : SUCCESS in 4m 50s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/0fd843a81ba24056a2db470357c49a87 : SUCCESS in 11m 29s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/d1aaa346c4b54ef89cae732475324523 : SUCCESS in 28m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/925e76454e1e4b55bc53b982d84ef552 : SUCCESS in 6m 55s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/2798954f57c44f499d3a9e29000aa1b8 : SUCCESS in 11m 48s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/d842c510fd624c88b9154cafc6300807 : SUCCESS in 3m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6f801f04a3ee43eb88909f181207b1ea : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d2f853264d954d3f9fe8d3be5fcd06b6 : SUCCESS in 29m 07s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/c376c838aa7f4b4bb2fe9e15916d73b5 : SUCCESS in 4m 30s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/40017c5ea9b14175bed009ada23638f4 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/df476db0981d491d8a7c9cf89a807e5a : SUCCESS in 5m 52s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/2d15f8ef5bc44bfdac82fe876d112d01 : SUCCESS in 15m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e37205a47e940ba9010addb6f83df6a : SUCCESS in 1h 07m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1554df68e12341a8ba8a151e95464cf3 : SUCCESS in 1h 45m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1b5c4ec0fa537201babf8edad3f2471851fc4f56","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-14 04:52:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f109afd147ce670bfc0b2fdf108a7222df262813","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 06:46:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76fa8ac71aee498d948a4944a8fca86a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c973f42570d4dd58157f73ee45cd803 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2971403d26d48dd98ace279636eff54 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfd5e1eed9824bc48775bf66db2be97d : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/49c5d73f3d04461b80e91a131b60d27c : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/109ecb97509e413b9d3a53ed475d0b5b : FAILURE in 6m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/93e029b41f0149568a60bb7226191857 : SUCCESS in 3m 08s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/56b03b414a694dffac2f7812eb8a88b8 : SUCCESS in 6m 00s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/b3b3f160bd21423d941819b25f071dcd : SUCCESS in 5m 16s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/ad58d4d042b746429d90129aabd4fd60 : SUCCESS in 4m 21s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/d35eb41819fd4e32b015f56732a6b43c : SUCCESS in 8m 26s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/95ca0c6b31de4cbfbf7c0982bf3344da : SUCCESS in 4m 26s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/e2db1635f27b4d17b0644b8ac291591c : SUCCESS in 4m 30s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/9ddfaf2edc30419a906121e0a4e6ec07 : SUCCESS in 6m 43s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/ea57a967f94e40ec8d591e18b181fbfa : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eea263c50e854d86bc412daa7534b82d : SUCCESS in 9m 27s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/27ac8e62874e4f379a13fba12a6e21c9 : SUCCESS in 7m 43s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/f4f32d97c01440b4b3f027f02460c27f : SUCCESS in 11m 13s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/972413d324af4aa38c1581e3f9671561 : SUCCESS in 3m 51s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b5bdaf1e58b0451caffa1319b67dfd3e : SUCCESS in 8m 44s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/7afbef39a2c8462d8dc83a669cbb5d3a : SUCCESS in 5m 17s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/13e371f345a6437dbed78504ad5ea604 : SUCCESS in 5m 05s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/afb07874cf5c40d79b59c90708e340dc : SUCCESS in 12m 51s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/bebf99ac61cd40899bf586e3c1457e9d : SUCCESS in 19m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d82f6127c10848608b1fc9c9b5a16331 : SUCCESS in 6m 12s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/b59a07b5730549d98a3662367b68da5b : SUCCESS in 12m 33s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/f1380ffb957546a09b6b73912d3ce59f : SUCCESS in 4m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/75aa10ddfd8845f1a8b336c7f9679321 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9826ff77797a48208314b10ac847c2b9 : SUCCESS in 18m 57s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/bdd79364e9344e5494442bb2fbc8318c : SUCCESS in 4m 30s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/6d0b6877dc734d148fc3db60bd226d82 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f9fa96896474ef6b8c56726d22c85a6 : SUCCESS in 5m 41s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/371a7dc03236410286ce986c14bae03f : SUCCESS in 9m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6b338bb73b544a59ecc2bb8e543799f : SUCCESS in 1h 03m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/364c69bcda3346619f48e538f7a2110f : SUCCESS in 1h 47m 59s","accounts_in_message":[],"_revision_number":2},{"id":"09fb5f156f5bfe5d0e24bd0d772cbffcef555015","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-14 10:15:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"29e26f77d758f949e44ced32cf2ecfc56ca5b0f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 10:16:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"facf4a626c1f92ab87c6ed90cd393c3173e9f190","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 12:01:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e587a89d0a244e7a923efcc28345f07\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc4edaaa568e42df91bf74b14ff35842 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/485429ffe7e74e8a8db91049dbd6c530 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15930ad96a584f039e4dedcdfa78a1c5 : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4821ea67e3814b4ab902c25047e44af7 : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/770fcdfa778048c0b151d9cc57d01ea5 : SUCCESS in 3m 21s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/be5346b097f544c293ce2f9c5f83ef66 : SUCCESS in 11m 54s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/db1aeab124cb4baab86b654c275da471 : SUCCESS in 4m 58s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/ca628f4f9fe1412b9946cda4a3255d06 : SUCCESS in 5m 01s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/910033a56adb415cbfc736598e9aaa5e : SUCCESS in 7m 10s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/5b0dcdb7668f459f8110b6aeda4d16df : SUCCESS in 15m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bbb3c545adc34667bbb58af2584803e0 : SUCCESS in 7m 56s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/edf34584bfbd4cd5bf1a24b81ce1800c : SUCCESS in 8m 07s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/09b05bd64cf746a8bdd7d6b914f89f8e : SUCCESS in 11m 33s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/3e18bead9379453b96e04e700ccaf1e8 : SUCCESS in 4m 05s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/f174975c904640d5880561a73580bc11 : SUCCESS in 7m 48s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/5323395b7f8046f4a005031f62484145 : SUCCESS in 20m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e6fcf339490142f99843c74ac931dda6 : SUCCESS in 5m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b2630cc0e54c4966bfd8da1384cd5adb : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3a5390f6d91449689aa87ff96afb301b : SUCCESS in 21m 21s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/6cdde526f1f14adb939d8829f3658f6d : SUCCESS in 4m 41s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/ec818ca6f9624672b782b1e6c10d6ed7 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/beedd3ee4bac419bb46d446d4f58a560 : SUCCESS in 5m 39s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/196ac12ec47f4c1299c1595cd7dd9afc : SUCCESS in 16m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3bebbf763f24436bbc3e5cfbf5858a2a : SUCCESS in 1h 03m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e8099bd7c1d4438b240d344a7d59416 : SUCCESS in 1h 39m 30s","accounts_in_message":[],"_revision_number":2},{"id":"90673f46b05ff620abfcd5617d198586142aaa9f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 12:01:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"55c8fa660d8045d673c7d8430aca5723bec546f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 12:04:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a3d74c6bc2d433d8596a90dfae071b6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28d152fca9044d0fba9da100596ff5a2 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1768ef70b9e4e49fb9f0b1648e76b76ecbc0ea3b","revisions":{"466f2eda4a96236499fcf1e2dff6046b3cb2200c":{"kind":"REWORK","_number":1,"created":"2024-06-13 14:50:13.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/68/921968/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/68/921968/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/921968/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/921968/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/921968/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/68/921968/1"}}},"commit":{"parents":[{"commit":"c1131bbdcfbf778816dd5bf838cdae9d84c17674","subject":"update constraint for neutron-lib to new release 3.13.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c1131bbdcfbf778816dd5bf838cdae9d84c17674"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-06-13 14:49:12.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-06-13 14:49:12.000000000","tz":120},"subject":"Update cross-horizon-npm job to jammy","message":"Update cross-horizon-npm job to jammy\n\nHorizon has dropped support for python3.8, so we need to run this job on\na node that has a more recent python version.\n\nChange-Id: I0ea11984e63b885668ed5842e1c6221739e18343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/466f2eda4a96236499fcf1e2dff6046b3cb2200c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/466f2eda4a96236499fcf1e2dff6046b3cb2200c"}]},"branch":"refs/heads/master"},"1768ef70b9e4e49fb9f0b1648e76b76ecbc0ea3b":{"kind":"REWORK","_number":2,"created":"2024-06-14 04:52:32.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/68/921968/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/68/921968/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/921968/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/921968/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/68/921968/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/68/921968/2"}}},"commit":{"parents":[{"commit":"c1131bbdcfbf778816dd5bf838cdae9d84c17674","subject":"update constraint for neutron-lib to new release 3.13.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c1131bbdcfbf778816dd5bf838cdae9d84c17674"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-06-13 14:49:12.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-06-14 04:49:46.000000000","tz":120},"subject":"Update cross-horizon-npm job","message":"Update cross-horizon-npm job\n\nHorizon has dropped support for python3.8, so we need to run this job on\na node that has a more recent python version. This also needs to use a\nmore recent node version, then. Make the job inherit the same parent job\ndefinition that horizon is using in order to avoid duplication. Drop our\ngeneric requirements-cross-npm template since it is no longer being\nused.\n\nChange-Id: I0ea11984e63b885668ed5842e1c6221739e18343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1768ef70b9e4e49fb9f0b1648e76b76ecbc0ea3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1768ef70b9e4e49fb9f0b1648e76b76ecbc0ea3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
