)]}'
{"id":"openstack%2Frequirements~923375","triplet_id":"openstack%2Frequirements~master~I84888695e2dc156bd58bb4de86b1388de3f6db48","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I84888695e2dc156bd58bb4de86b1388de3f6db48","subject":"update constraint for python-neutronclient to new release 11.3.1","status":"MERGED","created":"2024-07-03 09:42:48.000000000","updated":"2024-07-05 01:51:36.000000000","submitted":"2024-07-05 01:50:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923375-new-release","meta_rev_id":"cd9c80d7bff0b8abfc713ae78a16cad0341b70f0","_number":923375,"virtual_id_number":923375,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-05 01:50:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-07-03 20:51:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-07-03 20:51:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-03 11:45:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-03 20:51:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8e3c5b9fbccd9b0f68f0e8b1651339f707f501ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-07-03 09:42:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"201d2b1690eda25c2335b53d3fb771e71afb6b3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 11:45:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18c2648c34af4accaa736ca540e4b769\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55cf73db29064f1d99fc077fead6d6ad : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/109bb3a4579043c28b00b268b440b716 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/401f641c06d645eb84145399f0ea3405 : SUCCESS in 3m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c10a9742844491cbbbe7a42d7e9e9fa : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/66fcb9273e3c4259918847a77afe7851 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0406ea44a9904413b31fe761f4cddb38 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f807ec365ce9488a8859ddc56437d3f0 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8e0db866565f450c974f355d9bf809f8 : SUCCESS in 6m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/056b4edca9054b3fa69bbb0178a3499c : SUCCESS in 17m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/48f03bb199d348bea6eab41957ebb978 : SUCCESS in 8m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0aedea34afa249a78895ffc116ddc977 : SUCCESS in 18m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5eaedf0cd4344fac91060a548d07ad76 : SUCCESS in 26m 27s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/afb6d6ab7b7d442a9e7be569f4bba4e9 : SUCCESS in 4m 45s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/c089127ec08f4f8fa9fa5ab50edf6873 : SUCCESS in 5m 05s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/2e13c1a8c6c04870af10aa2c692ed108 : SUCCESS in 4m 57s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/60838cbb90cc44c88ba402ecc240f2e4 : SUCCESS in 10m 22s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/c7b5577ff18d470fb511c63c8bc7de11 : SUCCESS in 4m 04s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/31d3c5b3e0564c89945ef46bdd2830fc : SUCCESS in 5m 12s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/6e6e414a6b964fddaaf1d068f5fa846c : SUCCESS in 9m 13s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/27772d64b761406b9d543ac1e1e04e70 : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/99d40873324c412983ea5d6bfcb89eac : SUCCESS in 9m 46s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/1d8d6d2080f54a6facb28151204eeffe : SUCCESS in 7m 59s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/a295ee8e16994add904af6b76d880db8 : SUCCESS in 12m 17s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/da19cb2469e8481f98e604c75d6c23b5 : SUCCESS in 5m 06s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/62ce44240b3c49749bc7ba556bc31cc0 : SUCCESS in 7m 42s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/fc3819cd459c4e7fa70deb8132ef1e3b : SUCCESS in 5m 28s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/7c6232a33fe148cd930d65cafc7cbc55 : SUCCESS in 5m 29s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/778fba1ff20044e4bca8017043faa9dd : SUCCESS in 10m 37s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/57f5c437571d48c382a45c4191257b7f : SUCCESS in 18m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/72e873b9e98549e58e26b2d834306858 : SUCCESS in 6m 49s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/bec5bdeec9944b1fab6f34afee159c50 : SUCCESS in 10m 41s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/892ffd3cf88846199f75b659f2a3e0d6 : SUCCESS in 5m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d62796d560a34f419a0a5ccf0f6eb4f3 : SUCCESS in 5m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/26f27302c59f4517b70d1341f64cc232 : SUCCESS in 22m 23s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/1db0f7e207b04445b97b9d358ea7f953 : SUCCESS in 5m 50s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/e0628ed40eab466186f99047389e94a0 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/67a2bfaae6eb45d99d3aad814765d5fe : SUCCESS in 6m 07s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/ea80aa9e75fb4f4fa9d8dc43111f257b : SUCCESS in 13m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08db009be95c420e9e8439c9ef989210 : SUCCESS in 58m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e40c76cd1bd54219a2ea38f4006a7ddb : SUCCESS in 1h 38m 20s","accounts_in_message":[],"_revision_number":1},{"id":"616ffc3fe8bac4229f2eaab6f32c397c1bd4c1fa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-07-03 20:51:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3eb38f121cb64ae95da3ee4dd01704e422394cc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 21:10:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c21c67dde8e51b05058140ae9345dd2661045f29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 22:37:52.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1a54933a252cf06bc0120351190cf69be8521baf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 00:05:59.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b604c689d745f91e8ebef5e53b498b7972f767e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 01:50:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2da110c1634c4e619f630a6fecf79c3c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8519d00cbda44408029bdc66479cf6f : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b074a23d34aa49ef9236304251b58395 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21214e738595468e90debae83228d41c : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e6bb786269e14fae9fd9e63a723bdb15 : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ee29e93ed0947a18366fd068899253e : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7e118ba622c54aae8bb0e61fb8555ee7 : SUCCESS in 3m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8ba2444bd6ca40efb9e792a5abbe5854 : SUCCESS in 7m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/28b9aec4d03c4fe3b1e46149cad44421 : SUCCESS in 17m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4fa7256376c34dda99732c54b1c86ecd : SUCCESS in 8m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/499b693872cd40629781f19dc7092afe : SUCCESS in 17m 52s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/468fe7b53ca44d94a1f8170f4047c782 : SUCCESS in 13m 20s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/6bd46b1f2dba4c2b9e107fbe8096c612 : SUCCESS in 7m 18s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/d5b31afffc6348d28d594c7c2d687673 : SUCCESS in 6m 52s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/b77fa452e9a8491087823511fac02b47 : SUCCESS in 10m 19s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/7f49896f908e4950bffb24aee6f1223a : SUCCESS in 12m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/972387a5283b4192a85e5d62292770ea : SUCCESS in 6m 25s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/3d937d7de3ab4c3c8e98051b5c126631 : SUCCESS in 9m 02s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/059e91545c25418ab52311a48a9ffd56 : SUCCESS in 10m 27s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/3a9d813b392e4d88949c01b60cde4c3a : SUCCESS in 4m 32s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/c86ca9e988364f1a99f8c84f198344f6 : SUCCESS in 10m 37s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/009724cc81444f7e9d97ddcca87b3cac : SUCCESS in 28m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1f928d1354054b62a51cba90487095b0 : SUCCESS in 7m 41s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/7a5beee1401d4189893577c1415400c5 : SUCCESS in 13m 03s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/7f5c5b790ca94f7d90b6a49fb019b646 : SUCCESS in 4m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b492929f6126413cbe6f6ef1e303b322 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8ddc79c1399a4bb38694a9835541bb26 : SUCCESS in 22m 06s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/5828953683cd413391a7b28f0c28036d : SUCCESS in 6m 06s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/a499b108b0a346ff942729cba3145105 : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ab5562aed6ab4d06af533b0a59b60a7a : SUCCESS in 6m 13s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/3ca2d14630764db3a02310e05df33768 : SUCCESS in 14m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad9834050c144cdd88f983178bc0d9fa : SUCCESS in 1h 03m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a1a781c64e3747c3acc0d81cd857246f : SUCCESS in 1h 39m 19s","accounts_in_message":[],"_revision_number":1},{"id":"97a6506189b8d0d8900395bc13d5d39079afc22d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 01:50:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cd9c80d7bff0b8abfc713ae78a16cad0341b70f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 01:51:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13f2377305dc41b3924458f7b9da8cd1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/923df58c37de4f66a30536e21e4d3537 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96dcfedd11a03126d81c80538c1dbce5218e4c9d","revisions":{"96dcfedd11a03126d81c80538c1dbce5218e4c9d":{"kind":"REWORK","_number":1,"created":"2024-07-03 09:42:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/75/923375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/923375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/923375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/923375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/923375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/923375/1"}}},"commit":{"parents":[{"commit":"44f9faeaa5eed040e777e8736e88d1f0454a16ca","subject":"update constraint for os-ken to new release 2.9.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44f9faeaa5eed040e777e8736e88d1f0454a16ca"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-03 09:42:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-03 09:42:47.000000000","tz":0},"subject":"update constraint for python-neutronclient to new release 11.3.1","message":"update constraint for python-neutronclient to new release 11.3.1\n\nmeta: version: 11.3.1\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I41031fd6a204edf95af3fee01c05bf44efc7e384\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I84888695e2dc156bd58bb4de86b1388de3f6db48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96dcfedd11a03126d81c80538c1dbce5218e4c9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96dcfedd11a03126d81c80538c1dbce5218e4c9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
