)]}'
{"id":"openstack%2Frequirements~923480","triplet_id":"openstack%2Frequirements~master~Ia2f526070fdf501af6e928a1d66617f6e1f940d8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia2f526070fdf501af6e928a1d66617f6e1f940d8","subject":"update constraint for oslo.cache to new release 3.8.0","status":"MERGED","created":"2024-07-04 09:55:32.000000000","updated":"2024-07-06 06:22:02.000000000","submitted":"2024-07-06 06:21:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923480-new-release","meta_rev_id":"d6868f85380e85e20dbf610696190e9a57834a7f","_number":923480,"virtual_id_number":923480,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-06 06:21:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-07-05 19:00:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-07-06 02:14:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-07-06 02:14:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-04 12:49:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-05 19:00:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-07-06 02:14:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e08aa48f981d38465c07ca278acbf4c59e81095b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-07-04 09:55:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56f573256bdfea5a3c0a5138a4aa7a45f783c9f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 12:49:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a872702a607747199894044de7a0e895\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/197d7f0682ce419cbbd942f5dfe905b4 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/282849c295114b7dbaa8b1c3cb0dd4c2 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c70658b57b9e48c8a6ee71eb486b6e89 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c7fa2a9af9c4f5493b944ee735fbc4a : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8a63a38711be452b9318c4c2dde7f9a4 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f0a79c3f79147479913075871692e7f : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/41c21ba0ee924cb785d398e69f34ea89 : SUCCESS in 2m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/864f308d573a42c092d066b411dc1682 : SUCCESS in 8m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ab94cc98920b4bc2b31e2cb7cb24000a : SUCCESS in 15m 50s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/273523c17db54ec3891fa9a0f9242977 : SUCCESS in 9m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3c9ab09875ad4151b1965b386ceadebd : SUCCESS in 19m 02s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cf91c6f4523d4347a315860b7f94b8e4 : SUCCESS in 29m 31s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/5a2fc214532b4bbaabeeb9385dc26c15 : SUCCESS in 4m 06s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/bca97406baf748eaa71a81c7a738a5eb : SUCCESS in 4m 50s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/afce8a967e5b4498b458157a7dcb2e79 : SUCCESS in 3m 55s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/c86a29463f6b426e8578223ef8ed3675 : SUCCESS in 17m 52s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/19349f69716140769710d0f156ba8dcb : SUCCESS in 5m 05s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/84fd9e79c3014a68a47853bb68d99e0d : SUCCESS in 5m 12s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/52c9f34311d44aa4b2070b2b6754eef2 : SUCCESS in 7m 32s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/4225f4a017524cd5b5b37d8f07b3b7c3 : SUCCESS in 12m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/193950b6253749c8ba8f5262e915b0e3 : SUCCESS in 6m 56s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/7ba79eba0054462fbc0e6f7869ae1e49 : SUCCESS in 7m 19s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/ba79f758c23a459c9220c8a0c4c64ada : SUCCESS in 10m 08s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/1fc039875ad94c5e9ef700f8f177db9f : SUCCESS in 3m 52s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/1d19e7354c9e43d782d2f4335ca95f0b : SUCCESS in 8m 24s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/09b570e6fa724fb8b746adb06e92dc77 : SUCCESS in 4m 17s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/b5eaa515d2694bd4a59c6d14f3a45fb5 : SUCCESS in 4m 23s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/9c8b0d7f3adc4e94b6101e1b0dc1e6f9 : SUCCESS in 12m 00s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/5bd4383eafbd47608e85606f2b4f9c09 : SUCCESS in 20m 49s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a2d1ed984cf746ada54eb3619322ed7e : SUCCESS in 7m 06s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/d466d923dcf4416da00514beb4ffeb0e : SUCCESS in 17m 58s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/8cc3d4714ed44797b50ff4d0bb960972 : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/31011186d71e4cdd9a053bec16dba802 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/61eed59dc922454e9e81be89742ead4d : SUCCESS in 24m 21s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/6dc9ef5aaf564d7badc6b250a6a50fde : SUCCESS in 4m 05s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/435b6097d0b94ca69ce79de3c10424ba : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/103d80e83b7b461f97757374164153e7 : SUCCESS in 6m 00s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/3f06e50a87694299aa4279530419d067 : SUCCESS in 15m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3fae23fbf5b4b2191898398abdab508 : SUCCESS in 1h 06m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4fc4ea9a254428faff4c465ea6f84e3 : SUCCESS in 1h 55m 09s","accounts_in_message":[],"_revision_number":1},{"id":"0548b31de2c34a627853e107e4b91103eb48c47a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-07-05 19:00:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eeeb17caada718002994d5c0f428ec7958314895","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-07-06 02:14:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4af4a030a3b92115bf121407541b8f5e82f15240","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-06 02:15:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad5384c9734795dbd8cb2f22f312217dee297f3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-06 06:21:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e6e92148352408d863145b1daf3f567\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11d1810da9b14ad6974dbdfd57352e65 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da35eb42d1514f3b930b9156c3f14d44 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19698f4679344203b05e1acd33c1f106 : SUCCESS in 3m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2075427d866f4b79b0df58d85069aa7a : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/53b0365f9497428b9012d9f48c54d4a6 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0e49257e36bc41a7920779a3b372f100 : SUCCESS in 2m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a71d9f6042934d28a89b85d9cbfda481 : SUCCESS in 8m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/32c67311a13048cb82aee97a14602896 : SUCCESS in 15m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/07990c9219e8458ea0b03b64c611ec18 : SUCCESS in 8m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c626270dacc6409bbc12ad223e0b1e85 : SUCCESS in 16m 47s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/edf900c2a9fa47a4a0dbbd0ef9da5dee : SUCCESS in 10m 39s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/867c276868a84ef1bcbea8e41679655c : SUCCESS in 4m 46s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/12ae8cf24acb42288d26ec24c015d9f3 : SUCCESS in 5m 13s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/2efbd799a57b40c1a2761e56df80a711 : SUCCESS in 7m 04s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/63d9384c2c98440fa0d8d410bdb574d6 : SUCCESS in 13m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5e637f0548b343b68519d99bc8db8ecf : SUCCESS in 7m 03s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/0a91bd6ca78241cf89ed3ce2cca4f3f3 : SUCCESS in 8m 03s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/a08617c759664eb8b64bb8c5e6ae9eeb : SUCCESS in 11m 00s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/9a683bc3557c45a9928b8a741ef71722 : SUCCESS in 3m 39s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/c08a4e48a22e4d3fb06621e47535ce42 : SUCCESS in 9m 15s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/e95491dc0c8a4057a8297d7c256a4c30 : SUCCESS in 19m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0862f4a25a664e2fa58210b8e7eda3ea : SUCCESS in 6m 49s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/6d81c942f19a451e8ed9fe29a9d6507b : SUCCESS in 11m 46s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/6f4202faa9c841ac9e45409714ffe959 : SUCCESS in 4m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5199582aa9ec4bb2a3d789f8a0afccf5 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc9d48908a5c4a4d832e084c32d90256 : SUCCESS in 22m 45s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/04b6665d79284d45a271bdfac6a8cd9b : SUCCESS in 3m 56s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/14328ce5ddb143dfa30ea89d9ac167ba : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/57c0742c3aeb4d0b84dfcf2279e48d34 : SUCCESS in 5m 40s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/40c3e65932b84bd7aa24126aaae85e6a : SUCCESS in 13m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a61ba96db8e040fd850fb8fa35f8b960 : SUCCESS in 46m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0c1fcbb1d0b49da82ca059a85e2639f : SUCCESS in 1h 18m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c5ba494957cd92f72a776bea6a86570724a506ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-06 06:21:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d6868f85380e85e20dbf610696190e9a57834a7f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-06 06:22:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88d3c5c4fe344878a0461086ee8dd1b9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc1402ed70a3487395b528ced6c50e56 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a291c08ad1ab0d2f5ff2fc24dac0ef9ee37cef9","revisions":{"3a291c08ad1ab0d2f5ff2fc24dac0ef9ee37cef9":{"kind":"REWORK","_number":1,"created":"2024-07-04 09:55:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/923480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/923480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/923480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/923480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/923480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/923480/1"}}},"commit":{"parents":[{"commit":"0c378a1de344023719c887b3a9eff3fac7f6a66e","subject":"Merge \"update constraint for python-cloudkittyclient to new release 5.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c378a1de344023719c887b3a9eff3fac7f6a66e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-04 09:55:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-04 09:55:31.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 3.8.0","message":"update constraint for oslo.cache to new release 3.8.0\n\nmeta: version: 3.8.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I805be1f601a742ce01e594106ff072a3c5592c69\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ia2f526070fdf501af6e928a1d66617f6e1f940d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a291c08ad1ab0d2f5ff2fc24dac0ef9ee37cef9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a291c08ad1ab0d2f5ff2fc24dac0ef9ee37cef9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
