)]}'
{"id":"openstack%2Frequirements~923481","triplet_id":"openstack%2Frequirements~master~If39cbf54762b95cd0b9f78e92ac516a3d1013755","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If39cbf54762b95cd0b9f78e92ac516a3d1013755","subject":"update constraint for oslo.utils to new release 7.2.0","status":"MERGED","created":"2024-07-04 09:55:38.000000000","updated":"2024-07-05 22:33:45.000000000","submitted":"2024-07-05 22:32:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923481-new-release","meta_rev_id":"186a9bc23c8e7f1231f9f7ddaa285a9797a78c5c","_number":923481,"virtual_id_number":923481,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-05 22:32:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-07-04 12:00:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-07-05 18:59:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-07-05 18:59:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-04 11:52:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-04 12:00:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-07-05 18:59:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4dd2c30fb1bc8d92f6f3a701901f5bc67000729f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-07-04 09:55:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"862450863046f2e85bb2d63bf9d8a02df0d1d188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 11:52:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ea964847bd84c0a959828cfe98688b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/405197f7f49341a691c7fd00bafdfdc3 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2706a0f7cbbd4beeac384e0926ba29ad : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/296d658cd1b14944be2ec59328f965d6 : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f7fd3765bbb84e6e85369b2fea4f321c : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e953851037454949b8a44ab81a2d1faf : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce351ce4d4f04b92b3d5ffbace4071ff : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2deb1bba3857454b877be24c7d8d3546 : SUCCESS in 2m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cc07668669864083b9d4eb510eaf70f7 : SUCCESS in 8m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c51919ca7eef4b849921e8b4b0bafae8 : SUCCESS in 12m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4792e4803de44beca4db5aa92684b9b1 : SUCCESS in 9m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6fa08f8ddabe458099ada53cb449b94a : SUCCESS in 19m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d161cb2914d54a83b77d29b54b1668e3 : SUCCESS in 28m 22s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/1c56b3f6e6244801a6c953fbfeada6e5 : SUCCESS in 4m 13s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/30d137af48e648759c48ef6edfea3fab : SUCCESS in 5m 49s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/5482346677f44424a1dcf4a865c82935 : SUCCESS in 7m 16s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/36442f610ca44331b7dce00602573c2f : SUCCESS in 10m 08s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e6a0f6ca9497451aaad6eae49dda0817 : SUCCESS in 5m 05s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/75a44a5079224d06a46415136bc10f46 : SUCCESS in 7m 29s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/c483ee0976c84835a8292ccfa9904892 : SUCCESS in 7m 16s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/2a9b60c0de284a1683aa7295a4d4d3d6 : SUCCESS in 12m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9f773ea9963e42ba858e586828774248 : SUCCESS in 6m 15s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/4ce8251c5995442ca4f5386a322fc7f5 : SUCCESS in 8m 50s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/e19655ab9f6a454ba7de368abada5e88 : SUCCESS in 10m 43s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/64b355d354124af89cfe1efdbd9e1ef1 : SUCCESS in 3m 39s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/8d1a55fb7aec419fb8edf2d3ee196ed1 : SUCCESS in 8m 17s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/3d4d3f677cfb4a819cdb037bf0c6f18f : SUCCESS in 4m 09s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/90ac3c81fd4f42c9bf17faa25cf83109 : SUCCESS in 4m 38s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/442690a9ebed4468a527771dc2b808fd : SUCCESS in 12m 10s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/0c6d4dd163274bf39e504e64f908bb18 : SUCCESS in 20m 05s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3849da47c24f4477aa3de14c2b769e98 : SUCCESS in 6m 53s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/49247ed84cad44e28acc26e753016512 : SUCCESS in 11m 24s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/2fa547b7b875421aa27bf2c26693bcf6 : SUCCESS in 4m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/905074a53ddb47e88e8d8ef104318f3b : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b0a26d127ec24d69a82518699d9d77e9 : SUCCESS in 22m 52s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/ee3b7c047a1b4961b4ba9add9e03af58 : SUCCESS in 4m 23s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/dc969f2b5bf84d8e989523ec2364de37 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6ab169777c841629127e0996c05b71b : SUCCESS in 5m 48s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/14603ecfdfcb4b3db3d5276b4d89bf46 : SUCCESS in 17m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/646aabbbad6e463fb093cf97fa2ec9d1 : SUCCESS in 1h 04m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3bf2c16e92bf4fc0b820ac7b67b84934 : SUCCESS in 1h 52m 36s","accounts_in_message":[],"_revision_number":1},{"id":"94b3d679bcba632c5e62122e0d414ffe72e840b1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-07-04 12:00:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8df1f056cc61da33eb6bb25788987420c3bac765","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-07-05 18:59:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d378548a3092b28df77f02f63748d0dfcbb14ffa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 19:01:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10c38e64b270c05c45c6f0a0ed24319b703b4766","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 22:32:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7d7bbd925ad46a99bf4de87ad9896ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa25149f3bbf467e9806ead3b9e1d438 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6215c8322128493c9a816b04947d337e : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5080d669c2cf457791309da17b14a42e : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94d4594c85d84f14ab16fcef82b4348c : SUCCESS in 4m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1fd14b1a7454cdaba5c65e6302c09b5 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5e5c73a3fedd4563a310e64e2a993fd1 : SUCCESS in 3m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/745ae1ee81d9475e9c177a8a1d3af594 : SUCCESS in 7m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2f4c3f7d9ee447b2b37a7a6071896f9c : SUCCESS in 11m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d903ec43b71c434c81a9f2dd55147da6 : SUCCESS in 11m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c0a890c49bd844ccb775b07d47a1fde1 : SUCCESS in 17m 02s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/bdf7073461a640f2b5bcd39fc4304cbc : SUCCESS in 10m 30s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/6b58d7440cb4491ca40e459efd6703a2 : SUCCESS in 6m 26s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/c0f827165c7a43f2a8e69417def0a6fb : SUCCESS in 5m 37s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/5e94a1374a574a36af6ed9384e6f4573 : SUCCESS in 8m 57s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/547d7bb1e13b4bdc8771a3a1631e21f7 : SUCCESS in 12m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/167460182c05412f91e08af7c8fdda75 : SUCCESS in 6m 51s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/79e2d1198a654dfd91e1c5b4a994c6b2 : SUCCESS in 7m 50s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/16590167106d47828d774c9b1208b062 : SUCCESS in 9m 24s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/c3fd77735dbb4938b9af8df11589f56d : SUCCESS in 4m 01s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/63271a8502a346a9bef97d5b9ea5efa1 : SUCCESS in 8m 24s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/1a799d0087bc480f9362fae7639593b6 : SUCCESS in 19m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7a4ec3d64c45457ba44ca5a22a9b8408 : SUCCESS in 5m 44s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/713a8b065ee14a5a9b464e84eac09db0 : SUCCESS in 11m 54s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/8a4e67f849bc4177b7942223e54c68e3 : SUCCESS in 3m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9c293c7e495e40378128f96c754cfd2c : SUCCESS in 3m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0d69c5b313f34e6cb97f70e0d95d3f83 : SUCCESS in 23m 21s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/73b0d84520224fed813e38d155b19151 : SUCCESS in 4m 12s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/26f8d8daf6f845218678f6740192f998 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2384496e42c74e09a8dd3476bc20b192 : SUCCESS in 5m 37s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/a9e400b559724a7d811b70f80a142d4d : SUCCESS in 15m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e244b263dd148ac83c590b4011a037b : SUCCESS in 1h 02m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f049e31e0ce4ecfb8871bbe72f08ea3 : SUCCESS in 1h 30m 39s","accounts_in_message":[],"_revision_number":1},{"id":"1cea98858b244a324157c94ac2463eae7605525d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 22:32:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"186a9bc23c8e7f1231f9f7ddaa285a9797a78c5c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 22:33:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69b76427653543a584bc7e0782e5aeca\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0900fc50a5d42d3a2dc67092c953996 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1729681f5896a547a5d7a9f384aeb0e5444447ec","revisions":{"1729681f5896a547a5d7a9f384aeb0e5444447ec":{"kind":"REWORK","_number":1,"created":"2024-07-04 09:55:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/81/923481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/923481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/923481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/923481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/923481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/923481/1"}}},"commit":{"parents":[{"commit":"0c378a1de344023719c887b3a9eff3fac7f6a66e","subject":"Merge \"update constraint for python-cloudkittyclient to new release 5.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c378a1de344023719c887b3a9eff3fac7f6a66e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-04 09:55:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-04 09:55:36.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 7.2.0","message":"update constraint for oslo.utils to new release 7.2.0\n\nmeta: version: 7.2.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I8b84063c5d785d740258c53596d9f06425085d8f\nmeta: release:Code-Review+2: Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: If39cbf54762b95cd0b9f78e92ac516a3d1013755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1729681f5896a547a5d7a9f384aeb0e5444447ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1729681f5896a547a5d7a9f384aeb0e5444447ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
