)]}'
{"id":"openstack%2Frequirements~923482","triplet_id":"openstack%2Frequirements~master~Ie0e12195ed31eb7baffd4f1f8752688caa8adcb1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie0e12195ed31eb7baffd4f1f8752688caa8adcb1","subject":"update constraint for oslo.db to new release 16.0.0","status":"MERGED","created":"2024-07-04 09:59:39.000000000","updated":"2024-07-06 07:02:20.000000000","submitted":"2024-07-06 07:00:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923482-new-release","meta_rev_id":"c48171fdf869b10f91a3084d8d8322e17426785d","_number":923482,"virtual_id_number":923482,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-06 07:00:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-07-04 12:00:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-07-05 18:59:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-07-05 18:59:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-04 11:52:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-04 12:00:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-07-05 18:59:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9ac79226edeab42ce0c192c731f92b6106318fd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-07-04 09:59:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56870377a3381c4043a5727d728472429b3835d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 11:52:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93eebef9172947129dac928f714f87f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a123e8c9e48c4b399b5526c2bf77e326 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e249149af0794195a4fd35d6b04992f1 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c32cd06c60324d11aa6aed382909e8c7 : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fafcf8b85097454d9739beeea367c9ca : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4523088c25b84e0a84ec8c2a5c63e3dd : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/24191cf74cbd44ca9956080eba575850 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cdf3710bacac43ea99b30523a543b561 : SUCCESS in 2m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/960299e50c87462dbf37d2f46976b5c7 : SUCCESS in 7m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a48619a812944d7fa906d14426d4fdc7 : SUCCESS in 13m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/aac4afe97cec4c58a6212ec7a9175de9 : SUCCESS in 8m 37s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b58366fceea446bf91d4d3684f575e28 : SUCCESS in 16m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0a4065a653dc44028e6e3cf179cdc62d : SUCCESS in 35m 08s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/0d7fb6a0681d43b5a338947d43e86be5 : SUCCESS in 5m 29s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/2414ccb71389419081e06f1c8967491d : SUCCESS in 4m 40s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/5c3a2b00145a40a19eac6fb6bf373285 : SUCCESS in 4m 10s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/b3c37cfda8704b9380425d4a6ac8a169 : SUCCESS in 11m 25s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/ed908a350ac84f3a954b784fd7946d0a : SUCCESS in 5m 00s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/535ec19c8eb344d789009521928bbe01 : SUCCESS in 4m 59s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/34b11ee2a16846c1b9ed3f98af8ab5f1 : SUCCESS in 7m 17s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/55f8f96da8354f48b1eb701e7ba59ef8 : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/542e178348e5416d8fd2af83c8776e91 : SUCCESS in 6m 31s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/240adb6642f445748e9b56d831ccba2f : SUCCESS in 8m 25s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/414d86dd393f40a9b37e2f7af60ac323 : SUCCESS in 10m 56s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/b328c7c30fc74aaebb915cf026b0a97e : SUCCESS in 3m 43s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/3b642bf1223344e29d26df2ede7cfe21 : SUCCESS in 8m 24s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/e3ff7d16398d486bb02dcafa95abf543 : SUCCESS in 3m 55s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/0285d2b48a9d4df4b5cfa2c4e58d7eaa : FAILURE in 4m 27s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/34432481754d40e39cf3ec25b71d42ab : SUCCESS in 12m 14s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/bc89de6890d545aabee067038f806c2e : SUCCESS in 19m 52s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1632bac84c374ffeb13ff3166e337aa3 : SUCCESS in 6m 50s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/be7e60490ceb4934b73ce3943870b8b4 : SUCCESS in 12m 33s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/41d070ce6c8644b183f0b33cb14a7971 : SUCCESS in 4m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd4bc7cc0644450182967e9914e3c3fe : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3fbdda4957eb4e4a8d85312b03c8c529 : SUCCESS in 23m 48s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/7075ad63e3ee4421914a4c4127959350 : SUCCESS in 4m 24s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/ff6a58314643420c9ad3db9811ec6637 : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fb5c317d83734ad5bf668334f2aa0840 : SUCCESS in 7m 14s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/7e9411bf12344d8cb6dc0dcc34c55b51 : SUCCESS in 14m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a2546708fc046adbf19a5e11b8d8a03 : SUCCESS in 1h 00m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d87e03d5b9e04f54952ff89afa67240d : SUCCESS in 1h 41m 20s","accounts_in_message":[],"_revision_number":1},{"id":"fe4385d6270c95238960563191f0cf69bb716176","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-07-04 12:00:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7f619b9edd3af276c37fbce5622cbe458f2c8ae7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-07-05 18:59:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a0f2d163250c1fb9fcec43eee6187f6ce1e593ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 19:01:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"df8e58441ea52099be8d9d435068d648ea6ff2f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 22:32:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1bba99d0af5e42d59457ce8a705aa767\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42af175732f748e188778c9f9d06120f : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01d6666a03df484fa0f49f0cede34302 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bda7d389d910460b82055b385b353a5b : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0dc544fd283b456c99ad09e566cf426b : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c4d007b93ec144459c8e6ada2156647b : SUCCESS in 2m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/76a90e31e6f24ca69590540b32f85b95 : SUCCESS in 3m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6d10ab250c7b4a2d9e8850882d646e8a : SUCCESS in 7m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/854a1b83927746858edabd890f97a265 : SUCCESS in 13m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b9f9ca01e1af4685b64048eb4360b60c : SUCCESS in 7m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/db2570cc3760478a8d1a97adb432913e : SUCCESS in 15m 49s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/847069008bf942fab361deebcaee07ab : SUCCESS in 11m 21s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/6d3d5cc1410c424895235abf55f43b8b : SUCCESS in 4m 24s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/8f97f40ee9564540905f30c42c2515e4 : SUCCESS in 5m 32s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/475459d3074b4809a4ecd40c5d8007f2 : SUCCESS in 7m 30s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/939a020272474ea4a960a1ed4c96bb12 : SUCCESS in 12m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4196126e20844e98ae1ba0632aa38e83 : SUCCESS in 7m 09s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/0b2371dcc9f34f25a8f4b064022064f0 : SUCCESS in 8m 17s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/223e7dc52754456aa4e988ba2bdeac86 : SUCCESS in 10m 27s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/6fea5aa6d76042e2a6834dcac634b3fe : SUCCESS in 3m 52s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/5559624ef70b4e689ca2d3f473702909 : SUCCESS in 8m 03s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/31b6c27e26f443f9bd93ea60dae20aa3 : SUCCESS in 18m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8b31d5d3e0a54965b5dc0f5ace535aa5 : SUCCESS in 7m 22s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/b03239238b004113877802b0628f5aab : SUCCESS in 11m 56s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/905b5e8c20254e3b9b69350d158725a2 : SUCCESS in 3m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f79f61dc834a470ebbdbcbb38e9922f6 : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5dfcab13e80f4d43a17de24d91c2127b : SUCCESS in 19m 57s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/e4c66f8cb7f943e98e873ac84c3418df : SUCCESS in 4m 49s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/7cf5b32171a84856950f1545c079b8f8 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c357730210fe4befaac72a31728d0981 : SUCCESS in 5m 15s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/1347964a41394e20988f80e8afe25d38 : SUCCESS in 13m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a82a0f266a354bef8946fbfd1f497375 : FAILURE in 50m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e6bf5b233174f6cabba07aa9ebdcfd8 : FAILURE in 1h 19m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b487d645f2462c5cc2696135449874209bb82461","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-07-06 02:14:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9c64d615fc8571a5460b090e735fa74094f3f133","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-06 04:11:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e5a636be31b43cdb429e91fb1256cf0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e46b45b307b4a3081294282568df28f : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c85cc0984694fd1835e90ba8a4f7d29 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eab893ef2d324e69ba009845be740fc2 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/24fc5f8312e143ab8766ea0bea198286 : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa3a20dea03b47d9b0a4fccf9fca469e : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8ff28adad0bb4a98a9b59afe66f170b2 : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0e79ec71fb1241d49f1f65dd106d7a98 : SUCCESS in 2m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1c5998dd49fc47edabecd0b9470076b3 : SUCCESS in 8m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7aacfe0ed66d45c3a37e531c92531a6c : SUCCESS in 17m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/011649645068487f9df4fa6f23b5a312 : SUCCESS in 11m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4729d31c778e4cb6b74cfd5276baf4b9 : SUCCESS in 17m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/617779791df64d989581e7dec4b7bf8c : SUCCESS in 27m 14s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/7fb6c8652e0244d1aebe69de6621cd68 : SUCCESS in 6m 13s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/c5daa5caf77f40498f8ef67cfb6216a0 : SUCCESS in 6m 10s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/4362dbd272f54390ac4cb5b4031517bd : SUCCESS in 3m 41s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/086c326f15444fa38a6af9644b404e3f : SUCCESS in 11m 04s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e29ad5cf35064606b2520fc289559b7a : SUCCESS in 5m 05s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/9906adc1836e4368a710ccfc2c48cb44 : SUCCESS in 4m 58s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/6ae621ac4082464b940fc5eeb482377a : SUCCESS in 9m 29s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/cfea242506204ba687e074376bc59608 : SUCCESS in 10m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2d32b63431004da6b48b60e2afc338c1 : SUCCESS in 6m 56s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/fdd321d613a1420ebb8ac47e4183ee1b : SUCCESS in 8m 52s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/84ce7c7f2fc4443caf41e734b5c33db2 : SUCCESS in 10m 38s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/d96e77be6548497da0eef9872324f08f : SUCCESS in 3m 54s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/5060ef36faeb43689c74d37f6a31b14e : SUCCESS in 8m 33s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/1b1d8bfe73634aa38f2cde0657dbf597 : SUCCESS in 5m 04s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/a03e088471a743bab97077b735b85d89 : FAILURE in 4m 27s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/d659a1260f4f4a2988f1e52da48d8818 : SUCCESS in 13m 27s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/8f99c0f928ee4bfdb2d4be54e1b32b13 : SUCCESS in 22m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4bb7b0f0bb114df5a091d9abefd3d9e0 : SUCCESS in 6m 38s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/ce178c0ccfbf436298342c20207168a8 : SUCCESS in 14m 39s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/4e30d0002b8c479eace010089e8d28dc : SUCCESS in 4m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/92f8acd6bc234eeb89715eb0694df8b0 : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7535f36b387b409e9a375e58c4222437 : SUCCESS in 25m 35s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/3d40eaa6b16d450499c12cd2d6d5167d : SUCCESS in 5m 12s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/3a1820e2be1e4c74bbf68c20955ed86b : SUCCESS in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4227a12d9c274560851cc2cc272722ec : SUCCESS in 8m 09s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/df2f9def8033402e84d2e2d1ab09b605 : SUCCESS in 15m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec9f21dc4b6049eeb2d7f9b383cb35e2 : SUCCESS in 1h 06m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9ac1bbcddc9457991ed32a57c8c361c : SUCCESS in 1h 36m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ffd50c6807aa51a8cbfb6e7cdc0cbb5b66f00066","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-06 04:12:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b6159b46a7ef0e5808e9716113809aadf6a4bc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-06 07:00:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d491c942ab24a9f81fd15ca56c558a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c14ba9231104487987b8f741e10a2167 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bf5c080a5714d88ae90a3aed75b0900 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d26fdfc804eb4cb7b7f077abbf0a74ee : SUCCESS in 6m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bb03746a12b3449abd72e481b2a2769e : SUCCESS in 3m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/41674c738b184017ab3507f9d7dff802 : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0329ebe2205b44a4a4d4c50adeeabadc : SUCCESS in 2m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a86ffa2337c6480f99d71f52e8cda455 : SUCCESS in 8m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c4cb11cc93db4397ae98ed491610a610 : SUCCESS in 13m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/85f83c63e772466b8d32a3fb09adba3b : SUCCESS in 7m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2304082475d943c0a4fa64700de25539 : SUCCESS in 16m 44s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/021186911ec04525a48b44a937a152c2 : SUCCESS in 10m 59s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/4250816ba14a405088dbb89434422c56 : SUCCESS in 5m 38s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/3afa5fe4ae74475fbf7bca48fa66ac6d : SUCCESS in 5m 03s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/ef8ea9de0de3415e9e73b331a77085e1 : SUCCESS in 7m 18s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/fea83aa2d67f4f5ba3cbc2ba839a74f5 : SUCCESS in 11m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/98dbbde6cd97420589aaad6622931f70 : SUCCESS in 9m 15s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/cdc669b8287a40c49c363a4cb4224bc2 : SUCCESS in 7m 37s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/fec06cc060334c5e9e1614ac5322df43 : SUCCESS in 10m 10s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/8cb7021f13444a8b87a53e9c2b3ee160 : SUCCESS in 4m 50s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/74714529168f4719b2ab6fef0d229721 : SUCCESS in 10m 06s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/ef44a97f51f04f84ad1b6c60346be7c6 : SUCCESS in 19m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0e6c875c424547d1b1da441c696bc6c5 : SUCCESS in 6m 20s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/8f1521252a984fb99aefbb12b94ec46d : SUCCESS in 11m 26s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/56830518b4274b5b9a06640d64f21ca8 : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0c7654125f9f46608d39b73115d36928 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5e26bd93ce444242a1d4ebd7c6783eb4 : SUCCESS in 21m 47s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/791ff7c048c84f24a7d6ddf3c9215d7d : SUCCESS in 4m 14s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/df00af78a6a24f49abb990ff77f2726a : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d84ce16c846c4214b1bc4bc6c3704bb9 : SUCCESS in 7m 09s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/e870bc87c0cd437b9f07e6fa2029a6ab : SUCCESS in 13m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39e6128f5736485892c5c9cbdd51a682 : SUCCESS in 1h 00m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66020280afbd40da97ba7da70c079808 : SUCCESS in 1h 35m 11s","accounts_in_message":[],"_revision_number":1},{"id":"96f01b9c0943b8a5af526d6939af614127e019a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-06 07:00:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c48171fdf869b10f91a3084d8d8322e17426785d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-06 07:02:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f2a447c1094448abc35a9304b5cb7cc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8412900503f740db8e63ffae93e3f1cf : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2265c4d6259e684dbbf545626eb31cb8a635e952","revisions":{"2265c4d6259e684dbbf545626eb31cb8a635e952":{"kind":"REWORK","_number":1,"created":"2024-07-04 09:59:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/923482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/923482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/923482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/923482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/923482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/923482/1"}}},"commit":{"parents":[{"commit":"0c378a1de344023719c887b3a9eff3fac7f6a66e","subject":"Merge \"update constraint for python-cloudkittyclient to new release 5.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c378a1de344023719c887b3a9eff3fac7f6a66e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-04 09:59:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-04 09:59:35.000000000","tz":0},"subject":"update constraint for oslo.db to new release 16.0.0","message":"update constraint for oslo.db to new release 16.0.0\n\nmeta: version: 16.0.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I944993c7cc4d863bd1e0b2ee288d7b2f21452e7f\nmeta: release:Code-Review+2: Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie0e12195ed31eb7baffd4f1f8752688caa8adcb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2265c4d6259e684dbbf545626eb31cb8a635e952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2265c4d6259e684dbbf545626eb31cb8a635e952"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
