)]}'
{"id":"openstack%2Frequirements~924485","triplet_id":"openstack%2Frequirements~stable%2F2023.2~Ia7a18dda573aa614cbd5fb54be21259033a49e44","project":"openstack/requirements","branch":"stable/2023.2","topic":"new-release","hashtags":[],"change_id":"Ia7a18dda573aa614cbd5fb54be21259033a49e44","subject":"update constraint for oslo.messaging to new release 14.4.2","status":"MERGED","created":"2024-07-19 10:35:14.000000000","updated":"2024-07-31 05:46:49.000000000","submitted":"2024-07-31 05:46:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924485-new-release","meta_rev_id":"0bea58d777b4b330a81b78df221c3f7dba4814e3","_number":924485,"virtual_id_number":924485,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-31 05:46:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-07-31 04:03:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-07-30 13:09:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-07-31 04:03:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-19 11:59:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-23 05:16:32.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-07-30 13:09:24.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-07-31 04:03:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"e6818dd76d4066d9e507ab01590d77c7caeb410d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-07-19 10:35:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd3e4e87f060cfaf6f977ddb0fb76eefb7155384","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 11:59:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4cf7b9b0d3a4a199617214ca7dab6a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d2c3e1f514c42e68394c227803cfdb1 : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8247ac2393c048f28afc00746883a44f : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42b44b7296884442a9e333f3dea8e8de : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/021cab57c2404f1a8e9c35f251febfbb : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b9232660ea3840608ad1f0a9041e0765 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3ef604bb636345f1bf677840120a0a7d : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4bb55e7a45ca4c8188c516c7507e9769 : SUCCESS in 3m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f45712bb52b44934a372b7a94eaad0b2 : SUCCESS in 13m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8ff01fca39394b6082cf2f7505e3e265 : SUCCESS in 12m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/15b478128c1a401f8605f63b1d1977b3 : SUCCESS in 11m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4f74d81760a84bd19f202bf4354ddbd9 : SUCCESS in 15m 24s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/fc86e2ca10fe40ca80fcccb2100159b7 : SUCCESS in 9m 02s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/0d130d536c2e4f7ca7935e3132a1c734 : SUCCESS in 4m 34s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/98fc45ba8b6f4348add5205ced99dac2 : FAILURE in 2m 40s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6bc0e99442a44ef69181703a56622a84 : SUCCESS in 11m 08s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/39bfd9ef6da1481eb34d727b80a86ae0 : SUCCESS in 4m 29s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a122cef2ae054ba78df275f77f320b5f : SUCCESS in 4m 28s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a47dc16f80b7458a9d3ce1f674e9f268 : SUCCESS in 7m 15s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/2ecb728008c3438b83b59acbf13ef8db : SUCCESS in 9m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7e66b1ecb112431f822a3912dc3c390e : SUCCESS in 9m 03s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/1b13a0b7762e4f04887e7613bb836930 : SUCCESS in 7m 55s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/ed3c2e6f96d5460895ad074e3872ad02 : SUCCESS in 13m 04s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/626af6ec2d0342ea95e303e43f1e2e23 : SUCCESS in 3m 30s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a16d122e18394388a715327bdbc4b6ad : FAILURE in 3m 11s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/e6d73836485e4a16bfd1ac939c8d8491 : SUCCESS in 6m 16s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/b5e912f8495e42febc11dd64935319f8 : SUCCESS in 10m 32s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5f42d05ed8894c9d81043651058a7936 : SUCCESS in 27m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a02a921fb53b4b268a53892c64729bfc : SUCCESS in 11m 51s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/aec0c99f7aea44c0ae09051942446a17 : SUCCESS in 12m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/7b8fe18019444cdbbabaf55ebcf55cc1 : SUCCESS in 5m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/55e468fd8ef34d12a61275431dcdffc4 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3ca17c1783ab4d03b2cdee53f7629960 : SUCCESS in 23m 16s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9f29b8b84344464d94964b43cc521d0a : SUCCESS in 6m 16s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/85e5da51645f4bb0b9c87476b9e91c6d : FAILURE in 2m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e8d262c3dba446689d662da301e4e81d : SUCCESS in 7m 20s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/a471a3344d2545ae9412fa07ccb26fd1 : SUCCESS in 11m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04491d3cfb5d417c8937cfbea1ce0be7 : SUCCESS in 1h 17m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5915845c9d5340d2a1668efd89a8da74 : SUCCESS in 1h 18m 00s","accounts_in_message":[],"_revision_number":1},{"id":"4a64eedaf9848e24cbfcc9216b480b669f62ec80","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-23 05:16:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"779275eea79ea25a1442c5173a2a67450c073753","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 06:48:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/802a22a6af9648ed97d06d2e46850217\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d9d6684c93c4e6aaff865712dde1820 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dccaee606634ebdb5ab1d2fddc3fd69 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00b15848fd48438fa08e3bc8a01732df : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8156297b65014a6fa051f0922cbbf2e6 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dbf6405ed0574e6bafbc94804ed92c4a : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b79c815a7b8646c7ba922ddf03f3fbd3 : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4f2dcdb19f974845ae49f6bbad8833b1 : SUCCESS in 2m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b1c9c059c65e48fc8f3c7518d4ba9ca6 : SUCCESS in 7m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0b06bd6084624d5f9ff0371f8ee4d403 : SUCCESS in 11m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/abe62d6d0b554940b017cf9c0d21bc52 : SUCCESS in 8m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4fd7c1c683e848ddbb265bc16b934109 : SUCCESS in 11m 28s (non-voting)\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/2ac7d1979edc409b8ff785d06016375b : SUCCESS in 7m 03s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/0f7f3be8b98940c495c804d1225521e0 : SUCCESS in 4m 24s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/dbc1b82072b442d4ab288fc9b213b89d : SUCCESS in 4m 41s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b13e14723df240e59787b60cf6cff909 : SUCCESS in 11m 02s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/901163ff63fb42cd9e9c383f86c5e63d : SUCCESS in 4m 21s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4e417446966041cc9a619bf97bfa6136 : SUCCESS in 4m 50s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/52ccc7b1587747d983bfc1dc262c0e26 : SUCCESS in 7m 46s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ea4609e0c5c048a390bcebb6ee28e5d7 : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/06dc94ba4f304082a8d9fa7f3e29177f : SUCCESS in 7m 30s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/8f36cfe4a13c4dc282e1531e47ea16b5 : SUCCESS in 7m 41s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2ac5b10a7d07406b8481794157902f13 : SUCCESS in 10m 23s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/494104036d1d45228d76ce5be278fddf : SUCCESS in 3m 32s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/f6f5df6d37914ab08d1cba726ff3539a : SUCCESS in 9m 44s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/b8a2307298ab41c29f0e8fb6f1cf5beb : SUCCESS in 4m 19s (non-voting)\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/daba3183252d4e34950da3b7a1814ef0 : SUCCESS in 10m 22s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e7575b51ab3e4b80943fc9a8fa476153 : SUCCESS in 26m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f80c12ef067d48249891774182297e41 : SUCCESS in 11m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/7046ee09729340abacf219a318d3ea65 : SUCCESS in 13m 57s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a165054f1756440687fda947f11d2591 : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d17c68e411754b5cbc6b071e962a5308 : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/45b248838cd7406a9f80f2d4d03ca10b : SUCCESS in 23m 06s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e73b4eb6be844868809a8a356f299516 : SUCCESS in 5m 01s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c2d7b225b75840fa8b8bb9aa8f8b6c7b : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9dce2c1947fb43bba5e9f861b0c0f39f : SUCCESS in 5m 35s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/36333d94829148348a0ee5bce4105295 : SUCCESS in 13m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76c56b79b3b44b6d82d41800683028d9 : SUCCESS in 1h 00m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/689af418eaa44e639defcaee3a340556 : SUCCESS in 1h 26m 40s","accounts_in_message":[],"_revision_number":1},{"id":"cf0c929c412b4676e7b4c5f41bcd3695571ca520","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-30 13:09:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4beb990a20c6ec8312fde1e12ba38351c8f466a3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-31 04:03:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c721a0ee94cb4084c6569415cfd168506541c4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 04:04:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7fc6a10e8d923fcc857aa1fc6f89fdd8ec9ad5b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 05:46:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7655e53e40c64a539b56de8b9d05a99b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/971c98fc25544679a186fafede5acb8f : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/358c301ff9154cf7a086af9d9d7d683c : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bd690f8e4cc4f9292be313365096daf : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3fa1b37f4f44e31bd09cc339a5e280b : SUCCESS in 3m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f7b67961aa9e4b8b80d1b5e520a57508 : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/56184dfe746e4167bf489d13e5be87a1 : SUCCESS in 3m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/af7177ab6b414b6b9e3124e73851d611 : SUCCESS in 14m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c8bedd6973bb4eddafa5865e8b30781a : SUCCESS in 13m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e37a8bfad4334227a7aa3afb6a23f2c8 : SUCCESS in 11m 27s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8fa2de9129b446949f93112490060815 : SUCCESS in 10m 40s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ffcf302b65fb45d1bd431154c82604d9 : SUCCESS in 5m 03s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7f97df080ca548599d0ef81f7837f5b9 : SUCCESS in 6m 05s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/257c60f6c94f46ec847c11b71f307937 : SUCCESS in 8m 12s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/7467ba60ff654351b4d98d21a11aa962 : SUCCESS in 8m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/14816c6f1304440d8316524efbbe8bf9 : SUCCESS in 6m 51s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5563401daa1a4e60b999571ed1fb2648 : SUCCESS in 7m 31s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/6125e89914b54e9bb91b89c5a3ad0c44 : SUCCESS in 10m 40s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4dd1acf2a6f64da2842672ddece9b58b : SUCCESS in 3m 35s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/123a5de64dcf4033953e8ac039d77c8c : SUCCESS in 10m 12s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/87395accf0c84497833c5d237d50feb6 : SUCCESS in 27m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1301903e821b4410a085041a59cd9d84 : SUCCESS in 11m 23s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/f587aebe19f54dfa87ce9e149c282ad6 : SUCCESS in 12m 09s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/98310d6d6d4248cbb1d97048aa322b68 : SUCCESS in 4m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1b2c5b91b62c44498b91f0f88426fa03 : SUCCESS in 2m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2cf2ce7dbe6e4bf580b8aba6df974b8d : SUCCESS in 19m 23s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/49135ad5c76a4044bf61ca224dd6f109 : SUCCESS in 5m 07s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/1b227730bc1e427dba3d32c67a2bd85c : SUCCESS in 4m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6b722b3b344a47bbaa2185b68872b5dc : SUCCESS in 4m 47s\n- cross-swift-py310 https://zuul.opendev.org/t/openstack/build/17e0ece4fb1346b8b01254a72b68b039 : SUCCESS in 12m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47c9dad9192b4f9f93020a898bb04968 : SUCCESS in 41m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9140cb5816fc4e388c8fe99006866ebd : SUCCESS in 1h 32m 31s","accounts_in_message":[],"_revision_number":1},{"id":"0bea58d777b4b330a81b78df221c3f7dba4814e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 05:46:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c623aab8031d54e32418972e7e839738dd824806","revisions":{"c623aab8031d54e32418972e7e839738dd824806":{"kind":"REWORK","_number":1,"created":"2024-07-19 10:35:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/924485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/924485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/924485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/924485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/924485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/924485/1"}}},"commit":{"parents":[{"commit":"dffc6aead7b77fd221b1820359de378fefcef546","subject":"Fix pkg_resources exception handling for py312","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dffc6aead7b77fd221b1820359de378fefcef546"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-19 10:35:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-19 10:35:13.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 14.4.2","message":"update constraint for oslo.messaging to new release 14.4.2\n\nmeta: version: 14.4.2\nmeta: diff-start: -\nmeta: series: bobcat\nmeta: branch: stable/2023.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: Ia3e46af25ad59494ff6f5048e933b82e47a62a17\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ia7a18dda573aa614cbd5fb54be21259033a49e44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c623aab8031d54e32418972e7e839738dd824806"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c623aab8031d54e32418972e7e839738dd824806"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
