)]}'
{"id":"openstack%2Frequirements~924486","triplet_id":"openstack%2Frequirements~master~I736ff07a67d918c4af4e9655dfba23d7dd196286","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I736ff07a67d918c4af4e9655dfba23d7dd196286","subject":"update constraint for taskflow to new release 5.9.0","status":"MERGED","created":"2024-07-19 10:37:03.000000000","updated":"2024-07-23 06:44:50.000000000","submitted":"2024-07-23 06:43:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924486-new-release","meta_rev_id":"4b2a2d30940db17ff5c16163f4bb19de6e657d7a","_number":924486,"virtual_id_number":924486,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-23 06:43:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-07-19 16:12:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-07-23 05:06:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-07-23 05:06:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-19 12:18:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-19 14:36:25.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-07-19 16:12:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-07-23 05:06:09.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"5c7c56dbb3b0a3cd3f5a6fc197ef1edf794041ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-07-19 10:37:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6aa4c9f4fdea8cf8d53d40c24f3ccde2091987ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 12:18:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71d80fa679294fc789596567e6ab7c74\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b24cdb1871294890896181f6e6d72817 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a3307b6a7eb4183bc29c07c06413471 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7808ec222f5e42e99595e6da09041d97 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd980f4e220f41608b7578596a587b81 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/47de03b22ff24ec7aa8e82457dfa62d2 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/93e0ca3df5eb467b8f4a3e331d92bc3f : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/95d683eac1d64b4699dcf10c81523355 : SUCCESS in 3m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c7419bcc8eed4fc381323204fbe83c2e : SUCCESS in 10m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/99ac001262e34dafa0b69429d1287be7 : FAILURE in 3m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9f403a5d06e04cc8ac34f3dcafdeaf12 : SUCCESS in 9m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d7793a1984524441b4b7420bcf6cf5b1 : SUCCESS in 13m 38s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b4adf22805914f9ea585e08bc5d146c6 : SUCCESS in 34m 16s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/b67ccaf55abe4cb2ab7351b1561e3612 : SUCCESS in 4m 29s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/5441db49ba4248d8bc4d156f6cea8765 : SUCCESS in 5m 37s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/75ab8ec22f6a4288afed3ae5249947c2 : SUCCESS in 4m 02s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/b24aee65147f437abf23aa157c8abaca : SUCCESS in 9m 41s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/2ad04bda69ce41ec8ac962aa07ee126d : SUCCESS in 4m 59s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/30a224a4b91f496591e24ecaeefdc65d : SUCCESS in 4m 38s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/33488cbf91184b2b89b08567276a054d : SUCCESS in 9m 06s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/23021a3ec03a49588459171b97671fbb : SUCCESS in 13m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/549308c453e3497db16ca225a8c2bf06 : SUCCESS in 7m 47s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/9033bb020f5a4b1c82f0e6e6023a26d8 : SUCCESS in 9m 06s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/6a297175f44545a8b58eae86f3ec0a48 : SUCCESS in 10m 20s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/e5c309b7f21f454caaa71cd1a360ff37 : SUCCESS in 3m 39s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/9ad9de81f02f4a4b8d28924712b3f8c0 : SUCCESS in 9m 40s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/fc66ab9ce64b456e8ce3333b57b8f636 : SUCCESS in 4m 16s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/77621e849fa8409cb13af26275cf9a48 : SUCCESS in 4m 08s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/f4a833e126a14e4fba379c31368d4d88 : SUCCESS in 8m 33s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/66b41f1f96604a6b979234ec3a8f0ebe : SUCCESS in 18m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6d084b4f0fc845359b69f661c4dbf606 : SUCCESS in 6m 57s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/aa06f4822010455e863c9ebdb0454beb : SUCCESS in 11m 36s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/fb205cf55606479a8c0c36b70ff0c72e : SUCCESS in 5m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4e3384f9b587447e961c1760065003cd : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d04329c762554ca5b5ab008a332f5f7b : SUCCESS in 22m 50s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/c84c8887e6654dee9b6c6b9ca81121df : SUCCESS in 4m 02s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/f8982d02c2e04385bca6c0644e9da1dc : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/17f1a50299bb47138de6b46287136a79 : SUCCESS in 6m 16s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/023cbc247332410a94f5922f2bdbd644 : SUCCESS in 15m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fb33c19aa484fbf9337cc52dc598434 : SUCCESS in 56m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/62dcd0a663524633a11025993b242a61 : SUCCESS in 1h 40m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2fa5f8bc57a5f61d82f02fd200a7d25a9a208126","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-19 14:36:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c6c951e48bc9dc04466c41d3042edd2aa106bb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 15:52:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ca4c1346fe94cf39eed4a14802dd07d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec6199bfadfa4460a6e99e463b983832 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/596ae4a80fa444c2ab772cc6ed9568ba : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ade8900dc80d426ebd80cf6fc5dd0a64 : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/12831ecc438943e79a5e70c1fedc64ee : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64171e2675214fcbb089b724b4dbbef1 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8540fa602380464aafd6f360ef94a70a : SUCCESS in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/93e8b9aa7dd44f2e824b1dabac98dfa9 : SUCCESS in 2m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e7c90d870d3f43dcb0bc015bc6d01895 : SUCCESS in 9m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f37eef5700ee4004986b9cdc5b866685 : SUCCESS in 13m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/75f6eb9a383c46628e44ff29e984e77f : SUCCESS in 8m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7cb33e172ec04e1cbad8a3eab55d6144 : SUCCESS in 17m 35s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0afc61d5b985492bad3a482cdc19bbb2 : SUCCESS in 31m 00s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/afab70fb4d5c4800af678f977478119f : SUCCESS in 7m 00s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/366bbfcdf6674742a9338a1035ae9bf7 : SUCCESS in 4m 49s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/07c5980fb5344bf8b29464080c24532c : SUCCESS in 3m 58s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/2dcfa042c92e469fa1aac2c8cb6d4ccb : SUCCESS in 11m 46s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/259e04d7d90747d49cf16a2d28eb5c46 : SUCCESS in 4m 47s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/dd85ef3d9aef44e2954d4ad35ef1ab6c : SUCCESS in 5m 32s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/117f53964c01482f89db3b947cfcd094 : SUCCESS in 7m 26s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/47aa374f53184404871d45ec7ddbc978 : SUCCESS in 14m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/825360a9bd7a4454bfced39673627511 : SUCCESS in 6m 29s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/90bc3fb4e2a1463c93f162bceff09513 : SUCCESS in 7m 42s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/7e2491cdeec44e148bb00f52825a1308 : SUCCESS in 10m 04s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/126c9adce6204a088ec9855f530fd256 : SUCCESS in 6m 02s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/341320e474d34ca1b9cce0c91ad41b52 : SUCCESS in 11m 48s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/e605c5c5dfc14c708261ff3ec247d28d : SUCCESS in 6m 24s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/dc5401e755b54502aaa5109254b21398 : SUCCESS in 4m 19s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/3cecd73980934e9c9337fe102f20317d : SUCCESS in 12m 01s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/7891f2dbc38747348e6984579715bcc7 : SUCCESS in 19m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1e22c460167b47c6af6abb9e647d4a29 : SUCCESS in 6m 26s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/e2abf1762f894c07bc7b0ea2d620dcb8 : SUCCESS in 12m 40s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/7a996d5e61594235966a71f53f21debd : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/889d334e57014382b9ed241a9af8995f : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8135710a64194b90b0e49c8624521ad5 : SUCCESS in 22m 46s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/b1cf0c3adae44b71afb37e2b0bc4ef7e : SUCCESS in 4m 45s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c1c8ccd52823427d8ca60a4fd3a10658 : SUCCESS in 6m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f4717ea2744748e2b6880deddffee7e2 : SUCCESS in 6m 29s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/d6b091dfd5f64dffa2a7a4c0b35afaa1 : SUCCESS in 16m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fe74a2259144774b829179b2eafc50e : SUCCESS in 1h 01m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6dde6297104143d7a159ac7e908ae9d5 : SUCCESS in 1h 14m 46s","accounts_in_message":[],"_revision_number":1},{"id":"39728f882bc9ca8e980c3c5611aaa5696a76db72","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-07-19 16:12:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fb49f1b5320640de35cdd748fc59124f88c2cdad","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-23 05:06:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"68ed318aa1993f6e2c8f2ed75c3a0fe3b5258253","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 05:06:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c3c2626b80680002bba9aeea1fa82770f2cb4b38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 06:43:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b090954e7c6a4aeb944faee2a9401cda\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/771f6178414443bdbbfeb1bbb2c60881 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c1f3e5f42bb431eb93203a6c2461280 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fc33d1ce5114d2ea930bf3b6c199ea7 : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/723cadc856774c38ad576ac7dd7a8218 : SUCCESS in 3m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8804bc1bcfd44830817046740c0335b9 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/28c47b91e7814c7caff7e1f67f2bd65c : SUCCESS in 3m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8231c751d088430f8e0c684899f255dd : SUCCESS in 8m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/30f21747879948859a606ce39b798c07 : SUCCESS in 12m 40s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c8e2b942303a4575afa869545c0f18e9 : SUCCESS in 8m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6a27ed6afc4f40529dda188734b71514 : SUCCESS in 14m 26s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/b734d90314324330bb4f39efc7148a62 : SUCCESS in 9m 36s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/d420bf34d6714f60a98f686b98128c13 : SUCCESS in 4m 29s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/63d85b8ef6654786a210586e0bd475a3 : SUCCESS in 4m 39s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/443c65393840456187d8a0ec5bdc324d : SUCCESS in 6m 23s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/453459f7df604abdb8a7ff1d2c080c37 : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/acb8074eb7c94338a20ec5415bdcfffc : SUCCESS in 6m 02s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/3d29f0ef9b8d4df595a0dd3ecdeb2e96 : SUCCESS in 5m 34s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/318f2ea6f2344ad69c8ce3bf2d58d64e : SUCCESS in 9m 32s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/75b9e01258734ca8a2b586cf345760e0 : SUCCESS in 4m 21s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/6d3f4093e298421ba56e24ea46c9b5df : SUCCESS in 7m 56s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/046ff0888f464ab7960d90c34409166e : SUCCESS in 21m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6498204eef68472aa1afbc83eac91dee : SUCCESS in 5m 58s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/9c527e80751f4a9ca4e3f55045e4c3ba : SUCCESS in 12m 25s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/2fc19a07b0a846238f325502dfcfce70 : SUCCESS in 4m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c04fd79afd2d4211bd6401e8d649374c : SUCCESS in 3m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/af491aa6c334455789ab1f40456148a3 : SUCCESS in 23m 14s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/dc6233e6ff8f4af48020e4cc562c67f1 : SUCCESS in 4m 08s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/606dc1568c4d4b3796c68a6caee63680 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/57d7226178654227aa108b0859a90f01 : SUCCESS in 5m 15s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/9f5d01f130974169b4a81317de62a2dd : SUCCESS in 16m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0dda279aeaa4d4cb7bac60afb2c124c : SUCCESS in 1h 05m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d8fb833b4964605a4c149762c63261e : SUCCESS in 1h 31m 41s","accounts_in_message":[],"_revision_number":1},{"id":"2004498e61a7e8d69fc1197c0ad2cd3d13cdc337","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 06:43:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4b2a2d30940db17ff5c16163f4bb19de6e657d7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 06:44:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/875be1f092e64b6e9052125b9c875ab0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b3839191e464aa6bd6a9bac933ce183 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06fba58f858c13541a6525993859f78eb8e96b1b","revisions":{"06fba58f858c13541a6525993859f78eb8e96b1b":{"kind":"REWORK","_number":1,"created":"2024-07-19 10:37:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/86/924486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/924486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/924486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/924486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/924486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/924486/1"}}},"commit":{"parents":[{"commit":"cfc58b8fc976d5604c7a5f0c41ab0149193e067e","subject":"Merge \"update constraint for openstacksdk to new release 3.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cfc58b8fc976d5604c7a5f0c41ab0149193e067e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-19 10:37:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-19 10:37:02.000000000","tz":0},"subject":"update constraint for taskflow to new release 5.9.0","message":"update constraint for taskflow to new release 5.9.0\n\nmeta: version: 5.9.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I361826b891c256b713ed5d6214f8dad1bc283f1d\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I736ff07a67d918c4af4e9655dfba23d7dd196286\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/06fba58f858c13541a6525993859f78eb8e96b1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/06fba58f858c13541a6525993859f78eb8e96b1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
