)]}'
{"id":"openstack%2Frequirements~925059","triplet_id":"openstack%2Frequirements~master~Ied4be4aee568e6d90e2f4264568f85ce20778eaf","project":"openstack/requirements","branch":"master","topic":"bump-jsonschema","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-12-13 08:17:00.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-01-12 10:33:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ied4be4aee568e6d90e2f4264568f85ce20778eaf","subject":"Bump jsonschema to version 4.23.0","status":"MERGED","created":"2024-07-27 09:54:20.000000000","updated":"2025-01-12 10:34:10.000000000","submitted":"2025-01-12 10:33:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925059-bump-jsonschema","meta_rev_id":"8b10188224691915b1b399115eaf9dcf4a34f01a","_number":925059,"virtual_id_number":925059,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-12 10:33:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-01-11 10:43:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-01-12 09:15:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-27 11:11:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-10 16:40:59.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"}],"messages":[{"id":"6726c41e6e8f8ce00ea9de4fb37dfd93bbfdd962","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-27 09:54:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00105d8770baf3733ce5697ef4a7e770e3c470f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-27 11:11:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca599815521d472e881d2bfec596173b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6e7b343b3dd48188d009d08b71622c8 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a59bb4504c947caaf50e5dddc83d776 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21c949258968405bbf47346ca62b3146 : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44945fb95da640039a22319390030c50 : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b9ff60a12c3e482f8f5b72734149dae0 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2666bb781aaf4aefbc1e385d0530afec : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71409ad247b44dd1b335555779c3a902 : SUCCESS in 3m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3384311645a14929969a2b94a658966a : SUCCESS in 8m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c6d1a93c51a84b72bd1a18884ddd9213 : SUCCESS in 10m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1e8cda5d832c408090bb99bff459604f : SUCCESS in 8m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ca81ee19caa046b0bc60007194c23035 : SUCCESS in 17m 30s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b1b586cbc82b4e8e871e1246b6534e24 : SUCCESS in 38m 56s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/ca43f736b1094910962410cd2b8c1e20 : SUCCESS in 3m 38s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/8a9cc809345f43739ebd7fe0a1c79bdb : SUCCESS in 4m 29s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/188a254ec9cb4e3ab8e9c89a889611b5 : SUCCESS in 3m 25s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/3fbd32fa5b56484c80119f0feb1081ff : SUCCESS in 10m 07s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/7517d1c103d04183b8b2d37932e43710 : SUCCESS in 4m 56s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/a8f70dbd2a4149e6aed04a6b59502cbb : SUCCESS in 4m 42s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/44ede7e19d6e44289564efedbb913fbd : SUCCESS in 7m 59s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/50107adc0c4f41a58c32390fe4bb75b7 : SUCCESS in 9m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/62d011aedc754a55a1ec43021950d208 : SUCCESS in 5m 21s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/de67a1c4f1ef4481b4e242c65600c5bc : SUCCESS in 6m 26s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/429a3832177143ea9b92d5504e110409 : SUCCESS in 9m 18s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/169dc1c76a29436386052f0a3dbec135 : SUCCESS in 4m 39s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/798c5e55dfa74c5a8dc55e429ef17cfe : SUCCESS in 9m 07s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/8799e6158fa34370ae163fd0ffe3e3d0 : SUCCESS in 5m 01s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/fca7b612e392499b99cbc6f08e307448 : FAILURE in 5m 16s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/5a8d819dc26c433aa30b2a3dcbedd857 : SUCCESS in 11m 09s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/ef49e313002e4d2e80fbadea7f799a4f : SUCCESS in 17m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/803aca8806ef409b99eb6aa63a48948f : SUCCESS in 7m 21s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/bd0bf1bac0c74b6f9a79e950a6f5ecca : FAILURE in 11m 57s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/fc3a69eaddc24ffcab335c11b6b6054b : SUCCESS in 4m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bda5751595474d5cbe3308eb3c0b07ca : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/546212fccc9a465bae66ca1961b8b0b0 : FAILURE in 17m 46s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/c7480761cf0247db8373513053f50183 : SUCCESS in 3m 58s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/336d4e1dbf374ab5b253dbe70f700870 : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1c13a1eef00e44729053993c93a84a48 : SUCCESS in 6m 00s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/4fd928c878ca46b3afa36fa2a2ec7a99 : SUCCESS in 15m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aedd090995be41cbb448dbbbca0ccb35 : SUCCESS in 59m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a7d95b923e1402981fc054e5d3c1b48 : SUCCESS in 1h 15m 02s","accounts_in_message":[],"_revision_number":1},{"id":"f2fd7b0245c42dd7418150589c2f673b1695e76a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-30 07:29:51.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97fe8d00fcd1a9664a9d3cf324b9248b64d0b73a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-09-27 07:47:40.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"348369f9066c05a2a0fa39418b0770adf3446241","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-16 21:01:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6b190379f5b30c681b286031567ff0d8cd7375ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 22:10:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/670eb348711342e9b29d751e5b24737c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d59211e12124329bbab2dc7c17367a1 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0d778953c3c4f29ad76dd234d2c391c : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d6fcbc250db46e2a1669356f2110f6b : SUCCESS in 3m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e8a3d077969447d5b692a4c4bbca1fc0 : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2957565782ba48b09189f495ff3c99bb : SUCCESS in 2m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71c7a7d4c2a7457685d64622fa1451dc : SUCCESS in 3m 20s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f953419f07ed4d2b84eb28acd8efcdcd : SUCCESS in 11m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/65624453105444a0b0488f714c735194 : SUCCESS in 5m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b5692c9626a34295b196d02715b8389c : SUCCESS in 12m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/568308d5993a4892933f075fe269408f : SUCCESS in 17m 11s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/397d85df7eb94965aa741625f5f1ccb4 : SUCCESS in 5m 24s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ee66ecc5c3664b74807ef3a3c8eada0b : SUCCESS in 3m 48s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/8d734d694b5c4b9f8a1dcfddd79242d3 : SUCCESS in 3m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/07972958959947f6a0b9d8155eebb43a : SUCCESS in 9m 11s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1ca9dbb6aab541beaeff40eafe7c64f9 : SUCCESS in 5m 01s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/09e039e54b3e47d3a52717c3b4ec391f : SUCCESS in 3m 50s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/906f705d649949809ca59b36a886e214 : SUCCESS in 7m 55s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/bed6efa65fff4a339bb8bbb8176b49de : SUCCESS in 13m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d586af0fb18e41a5ab6ca942006cfb9d : SUCCESS in 6m 47s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/2020624f338549baac24f46243b01e48 : SUCCESS in 7m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ea53d295da3c45bd9a1fdd2f0d4f27a4 : SUCCESS in 11m 46s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f63cefa1155c4ee8a5ecff821ca6eea5 : SUCCESS in 2m 10s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c4710a22bdf840ce8b8be1a9455c6695 : SUCCESS in 9m 47s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/aaad9a58529c458cb9cf5c0c747dc2d2 : SUCCESS in 4m 26s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/03d060c9e09841a4be73abd827ced43e : FAILURE in 4m 23s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/45a6c8df7be345a38437ee0f81f6088e : SUCCESS in 8m 42s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c9faff631d944b14a215cdfc8135b1a4 : SUCCESS in 16m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c4c587e8fa564f7eb428b0acf1468ebe : SUCCESS in 6m 10s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e253ed9bec644e43a51622d42ce9cfd3 : FAILURE in 9m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d2f3540570454ed8b909e8204c10fc9f : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e8a628e179124e048d8aa036d26442b3 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/541e8cac9f2946eaa9ad70e81ccfa9cc : FAILURE in 19m 19s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/bae7f6159dc3472ab2cdd2d22665e1a7 : FAILURE in 5m 41s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/716279e94883440cb0f96c18e7402588 : SUCCESS in 3m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f4d444dd1cc74400bdfbbf2cc68ca693 : SUCCESS in 5m 02s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d870ad4b406e4355a9ff5f64ca003874 : SUCCESS in 14m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e23c086bd384d318b0a83ac22154be0 : SUCCESS in 1h 03m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff4a2b017d094d46a53f74eedc36bb73 : SUCCESS in 1h 06m 29s","accounts_in_message":[],"_revision_number":2},{"id":"8c59f8a53ab239dc1e7d6a9c5aeda43cd564a0fa","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-18 20:10:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7653ffbd5feb44ff32993eb6d54d61d71e912f10","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-10 16:40:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"19e9c4105e8acd7a85d44e60a35254b824b87707","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-11 17:25:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"29361c03cccc9a152f3bf9bd9c945dc4a3cbde14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-11 17:25:48.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"72d70c3c4e7bf66883cf8b7278ef1cc3a8f66b21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 18:26:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d15cdea342545a49ccccffda40ce0e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/946c356dfa424bffb53177b127860c8d : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db4fefa87df1478998e96413477335d2 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90b197cc42504071a6c276fbb6b5c379 : SUCCESS in 5m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d324c6bc00224ac5b1ca24c17128fe3e : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/750e136a83cf4ec5b65c86b9864fd8c0 : SUCCESS in 2m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/86133e7f1e854a268f824fa27a145e06 : SUCCESS in 4m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0e31580f74194dc899c10d9695978272 : SUCCESS in 16m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/de7d09b378b04d6cb74ec8a06bc135f4 : SUCCESS in 7m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/abcf07e5954c4bf7b38c65037905831b : SUCCESS in 17m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/38df0708458a403baec97171d92bc62e : SUCCESS in 26m 57s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b4ff57f2d9564865b0fbeb80527a6817 : SUCCESS in 5m 32s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fff979ff0b58447caf0ebf1f1aeb4052 : SUCCESS in 5m 33s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/3e052461bcee4a8aab5d9477a6660828 : SUCCESS in 6m 23s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5614444caf084bb7ad176cb863472e41 : SUCCESS in 12m 42s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1dbdd5abb9564987ab58c6f7b19efec7 : SUCCESS in 5m 27s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b540f3172032456e8e2d563a24ffd17f : SUCCESS in 5m 54s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/16ee8394ec7e4e9c8e46d716f7e44ef4 : SUCCESS in 7m 23s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0e24f936ac414cb995f022eba063aedc : SUCCESS in 15m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/29825eedaa76464e96811d755edc4918 : SUCCESS in 8m 24s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f732ad142d6b4d1b9f3666f8111b535a : SUCCESS in 7m 10s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2650a2143bba4971b0fe03edd394baba : SUCCESS in 11m 21s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3e69521dedd347e5a9372e914fc35417 : SUCCESS in 2m 33s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ade239efeadb42f088d7a57eeb72e803 : SUCCESS in 10m 52s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8c91ccc07d4646eebf703e418a43c32a : SUCCESS in 4m 45s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/25ee018184454abd854a2b55ff6f8424 : FAILURE in 6m 21s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4e1fed52f82540bdbc3bc5f49a082968 : SUCCESS in 10m 45s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/27abf349ea244730b6fdcf034459245c : SUCCESS in 16m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/03b5b3bb00464130a0741d3214e52415 : SUCCESS in 9m 35s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/20459df755284daca62b9fa4bbe8fd00 : SUCCESS in 12m 19s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f6b802c74e294246a93b8d28e6d74f5d : SUCCESS in 6m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/272c8071313942fba80941ca8b30ae5f : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d718a5b598bb40f4a9624be02daa3b7d : SUCCESS in 24m 18s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4354466f8b8e4287adb638c9058a899f : FAILURE in 6m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/acab74b1ac5c41719c920077bf0a26a0 : SUCCESS in 3m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2f7bbb7ffbc8454e91349c1dbe129c47 : SUCCESS in 4m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/392ae687ad13491b8fdd8a7dd92a6deb : SUCCESS in 7m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9668c0e091ed4ce3bb47f531cee3f49e : SUCCESS in 53m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c585e057e714b999cc03470b62c0cf4 : SUCCESS in 58m 59s","accounts_in_message":[],"_revision_number":3},{"id":"1d94c9a2057e4103658e7176ee0dc3ce7dc408b0","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-12 08:18:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c9d5b1371ed06b2fad7907ff1caa8c60458b9903","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-12 09:00:08.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a3cabf701a70707aa5ad6ab3fa6c1eee7c9e74a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-12 09:08:45.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"b4ef10c860dab067b639664c0329603edb36b889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 10:25:05.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bca18bbffdbc4eaca9295cd159fd1987\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e35008a5a74e4eec843a2e61f7a1c9b6 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/690d1bf7bac9431ebe1750a43670f13b : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/683d2ad62381420bb0ab5a8f94e61583 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d035ca5503df48ee899ea739193c6c92 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/360840a8b2644541945c9e94d19c2855 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d05af6851a5a48c7acc320c7664fa73a : SUCCESS in 3m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f71a706ff91546878a2eb5a04af112ee : SUCCESS in 14m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/202db3266a824106b99699f12ed36f8a : SUCCESS in 4m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/759ecafbbf9b496e90c2494ca6aa22d4 : SUCCESS in 17m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e912ebb01fc6445fb980e645d3cc0bdc : SUCCESS in 28m 11s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/1715f39c193742afb25c365b692aff73 : SUCCESS in 6m 27s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/7911925863cf4ddb8c00333efa2a1854 : SUCCESS in 3m 49s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/6912f5c999734136af9cd4db8f90d418 : SUCCESS in 4m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/66427c3ffcf94657b94add3bd9259d3c : SUCCESS in 12m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a970203678a24a2fad996992fc3cdad0 : SUCCESS in 5m 55s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/fe1b1cfcd6414d529a03e75cbcb6fe7e : SUCCESS in 5m 51s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/bb5a9caba1574c89acd3191253b5fdc9 : SUCCESS in 9m 10s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/740c182980564d299f44135fcddef82d : SUCCESS in 14m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a4a65a0f20a4b92a96f78054eda63e4 : SUCCESS in 4m 49s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/bcc4ad0b940a47d89511d9b5bb799834 : SUCCESS in 8m 12s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/14e18a4aeb264778a62b6254076d706c : SUCCESS in 12m 09s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a134577974d54ada81d83ec47364cbcb : SUCCESS in 2m 19s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/022121a5bfe645b3a36fa1760bd52a3e : SUCCESS in 12m 46s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/e0551739d5ff495495d8c38b54f09b12 : SUCCESS in 4m 51s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b49b27b194004e97985f3ec6b7261969 : SUCCESS in 3m 29s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/46de599192e44501ab345731b9cfcf8e : SUCCESS in 11m 19s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/861d7933e56e40d49f6fb37fa1208991 : SUCCESS in 19m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e653b572110b46d6a2deff3e15701b06 : SUCCESS in 9m 06s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/1cd98b46c8334073a6c7997fab7fa6fa : SUCCESS in 14m 06s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/845e93c7e6ae4127a1cb1fb9e2bfdd7f : SUCCESS in 6m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d2e3b195710843948b20bbbe0b58ca5b : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a78b2c3201794141a00c61461722efe3 : SUCCESS in 24m 08s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4792fcc898ff4ad79d6c4ca0435ed268 : SUCCESS in 5m 43s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0c70a3ab8a4042099ff55f9b00eabb2e : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d6f1cfa81aa141cab90f835d9d87da39 : SUCCESS in 3m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/32345a6fa2de4fb999ec4ad55e660649 : SUCCESS in 18m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f659fde6aae491e992be4e270b0233f : SUCCESS in 1h 09m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d4590ff4d2d430ab3d82f63168fb194 : FAILURE in 45m 46s","accounts_in_message":[],"_revision_number":5},{"id":"b88f9f7b977f4668676b1f621fb9f303d8f675a9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-12-12 15:50:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e1760d5cbd8749d662340938ab32dd6b0d03e76a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 17:33:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bb7af9ff6f14b22ae5bb10a6d69cc3d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9ef641be4d24b2e993e9af37dc30705 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/887f637aac624ee192e662ca7e5670ac : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b11403dc88b47c99489228db65ba9e5 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/67f0708b54d247aa90fab462b1b0bc1a : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e3e6d9dfd2445239dfcb8e847111b96 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d53984dab6440ebb9ffd4474031da28 : SUCCESS in 3m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d3646ee64421458d85914953911160de : SUCCESS in 15m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8c926e4727bf4adb893dd53fd6db46e0 : SUCCESS in 10m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/056df58ba86b443bba5154bd602c18b5 : SUCCESS in 13m 10s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5c826fc21398425a830988b0efb7d2da : SUCCESS in 27m 50s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/959a320bc90a45098fc9481d5821aee9 : SUCCESS in 5m 54s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b2d20189086e487e82c3719723beddac : SUCCESS in 5m 43s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/396ee7d1a5bb4ca0af4daffad7dbb082 : SUCCESS in 5m 06s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a75b9db4437c4e35b620515b19994d04 : SUCCESS in 11m 36s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f419c3cbc5024c42ba7d6dd32f190826 : SUCCESS in 5m 43s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/68d1a8c5ac04453890334e7bec4de2d0 : SUCCESS in 8m 12s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4b6663017da54577bfb7c044d8d17bea : SUCCESS in 10m 58s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/119730cc822049d6b7976be3422caf22 : SUCCESS in 16m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/79d89958fc7146fd8a7ef8f1f04e57bd : SUCCESS in 8m 58s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/db1b63a1e0af411180a6491bd2b59aae : SUCCESS in 10m 12s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c3f35f31eec14b29b6638fa81b9ba13a : SUCCESS in 14m 10s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/804b24def5cf47448959db45b2412946 : SUCCESS in 2m 28s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/e7a14dca89584b259512119a89659080 : SUCCESS in 10m 15s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/e9de040d50434ab8bba11061510cd23a : SUCCESS in 4m 55s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/933d611286654eeca15c634f82b28f27 : SUCCESS in 3m 39s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/69054555688b454aa6255483fe5dd486 : SUCCESS in 11m 45s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/341c682f5f354f11993ac3f9c1893bae : SUCCESS in 21m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4991c8cacce645cda2bfe48207584618 : SUCCESS in 7m 34s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3fb6f166b5a345918602f19f6d3631b9 : SUCCESS in 14m 13s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/f3eee86dc50e49d9890d717e3fa7bf0d : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/42c1e9cf7a394e40ac09d3bb8d6cff2b : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/15f0960b48c143d2b8305f13386fc6c2 : SUCCESS in 26m 05s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c4397abc62ca4aa2b6d5d90ff1bf7ad2 : SUCCESS in 6m 49s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b8020d90d1a04f91954c3049e6fcb1a9 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0c7745e26d84473b931ef69fe4769b1a : SUCCESS in 6m 34s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/a73c67b1524d46eabba5b835786c02a8 : SUCCESS in 18m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/180b43901c3e4f3482798f297d7355c0 : SUCCESS in 38m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be58a49ac5ff453f8070d875262ad5a6 : SUCCESS in 1h 40m 45s","accounts_in_message":[],"_revision_number":5},{"id":"469e227258e642ab67c5a52d45651d4f49a49a5e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-13 08:17:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"40377df3da422a69da4037aebc9632da48ceabf6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-11 10:43:15.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ad3ddab61bfe6c7d8c52c05a49ef28d9021adaf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-11 12:32:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/221f7ac800d1494b9eb818e5c7711002\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82d51b2b00b241cfb180290a1df5cc40 : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/304d0d50c336455b8585dd76039d717f : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/551e2889e93048a782b8c9fc04f054dd : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4146957a2beb4247845a31dde0c9d6e9 : SUCCESS in 3m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/23f419d6df8548a19e52c106b3e64934 : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/681114acf44842759b289931407f2596 : SUCCESS in 2m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/735c988769fd43d0b22a4d1c1cd7f1c4 : SUCCESS in 7m 04s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/68f6af8b13634b62b2a97a2c6dbc6f79 : SUCCESS in 6m 52s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/945216aa91e24dab892759ec9544e09d : SUCCESS in 12m 13s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a6d76d2640bd4a00a28ad487bddfc216 : SUCCESS in 9m 31s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/d825fec320f749c2a77df061190c0f9c : SUCCESS in 4m 25s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/dc0ced860046473abbf2c50e3238188b : SUCCESS in 4m 17s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c079b4e081604efd915bd2b20202df26 : SUCCESS in 3m 40s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b69ac817319c4d3da83a64d3bdf57566 : SUCCESS in 7m 59s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/198563f9166a43b2b4976a7f36006632 : SUCCESS in 3m 57s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e626620e99f44c869cd54c23c625b8dd : SUCCESS in 4m 44s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/948aaa762cdc4b17afb671a11641ec25 : SUCCESS in 5m 56s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9f8517f925d94cdb85129844a46c8d58 : SUCCESS in 14m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a86b7a26d9b945b1aac5e2643d4669fb : SUCCESS in 4m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b13ad6c8b0784ecf832f870d73343359 : SUCCESS in 6m 47s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/71c95b2bb96a4c5382cac4945b876c01 : SUCCESS in 8m 27s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2f526c07f2a94b7c80d691c277ff1388 : SUCCESS in 2m 03s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c24f82d53c944b10a56b9734d32248f3 : SUCCESS in 7m 40s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/26874afaccc940f49b7a40220dda4d05 : SUCCESS in 4m 10s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/33e1139ee9584e20a36a56d75db4f7b8 : SUCCESS in 5m 12s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/5e959b6d19ba47f287f89e55292fc220 : SUCCESS in 10m 22s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/efe24a404f3f468db7db93e455f65114 : SUCCESS in 16m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6302a545ac314f52a9fd30e41750779e : SUCCESS in 8m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c943b797d6b8480dafff31a567aee311 : SUCCESS in 10m 26s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e3a66354fe25450cbe1705fa27057c4c : SUCCESS in 3m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/351cedee47b84f358c344f47114d9dea : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/104590554788427bbfe063df4021373c : SUCCESS in 19m 17s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/802a91ec6473446394f41786d8d3f5fc : SUCCESS in 6m 23s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/374d482062a04c0ab5bbbdf154c84fa0 : SUCCESS in 3m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9cf4bbdfb91d40d5bba700d224064e95 : SUCCESS in 4m 30s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f435a0a75b7c495fa420993e740ff3b8 : SUCCESS in 12m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11b799dcd9994c50b87d05f25d811460 : SUCCESS in 38m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8425447133240988b0dfc5afb97d209 : SUCCESS in 1h 42m 21s","accounts_in_message":[],"_revision_number":5},{"id":"940215492c90d12fa61bdf7946c7b996dd4725b8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-01-12 09:15:35.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d9feb8c19100eee3fe01ef39ade795a14560b1e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-12 09:16:15.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b0ec49e558b5cef521732b21cebf04dfe2be1164","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-12 10:33:12.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13b4298451fe49318fc4dd8fdbbcaa80\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7ac3912b44846308b5a07f9ee1a7d06 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02076e3c67914c739560569c0b407f90 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11d6d7fa11d446f3bee9dc6e3051e3da : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dba35b2f68134d5c8a14da3780cd1054 : SUCCESS in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/06ee0f5fc43e4f1aa1fdda00d9fcdb19 : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b7e241d06d8b4740a1ac9f3a6bd6b67f : SUCCESS in 3m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dce3c979645d443388107b27c72f3a70 : SUCCESS in 9m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d0a11cb5a8c24a519e508489c585fcf4 : SUCCESS in 7m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e0cbbed1d796489c8ada32a03e3a9689 : SUCCESS in 10m 13s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e8b687b63dab402ca099d507b297ba6c : SUCCESS in 15m 57s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/cd441584eedd409a85aa78f710e47eb8 : SUCCESS in 10m 07s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/36f3e71b96294fa6adc84252eee96a8b : SUCCESS in 4m 31s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/7c904bc6452e43d7bff68b5eccd15ea3 : SUCCESS in 4m 33s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/35789b0394e04645afd244fa6511730e : SUCCESS in 6m 38s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/63e9a7352ff94a9993c1772dfd416bcb : SUCCESS in 13m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2523b3e0d4624891bf558f35f7ce4f7d : SUCCESS in 7m 07s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/a0e99094b647464e9bf0fb180997bef4 : SUCCESS in 7m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9c6d0dc9c00c4ea59c04ae5298a5bb5a : SUCCESS in 11m 03s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/551ad469d80f4e7f813273b228f793f5 : SUCCESS in 2m 07s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/379bbda334b340dd877434ae23a78674 : SUCCESS in 8m 30s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ecd4a4e147c44354922d2b567990d4ba : SUCCESS in 18m 42s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1cc960b7dcf144da974b430e977992c7 : SUCCESS in 7m 45s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b462d0dd41dd4ab890c31bedefd8126f : SUCCESS in 13m 32s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5ced1d70fa024a4fb3bcc749ab8dc4f4 : SUCCESS in 5m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e3f4d1536aa94c0e8197defafd6d94e5 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4359f861c8f54052b7242b3a779e37d7 : SUCCESS in 25m 41s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d5dfe48730904cce93b96b99de511cb3 : SUCCESS in 7m 16s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5f12e5f9d25a4a1a8914521c1a5db467 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/050f881fe08c45ddb7dee2a6074b168a : SUCCESS in 4m 41s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/34d076988fc540218557a6d14e521b61 : SUCCESS in 18m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d344885fbc8143aa807f3fb40a3b3b1a : SUCCESS in 45m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/20675b1a1b504b4fbdba838544885c26 : SUCCESS in 1h 16m 29s","accounts_in_message":[],"_revision_number":5},{"id":"2a5cb6a0289c8f8e9361deafcecc8339a6725fff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-12 10:33:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"8b10188224691915b1b399115eaf9dcf4a34f01a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-12 10:34:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44972ad6e38c4e1190d8804c2266eac8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc5c11fef1304ad1859edd421010e28c : SUCCESS in 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"97a8cf245605515587517431fe458ca19834bd17","revisions":{"c60449dae96692e8ed8845800810b8b6a686785a":{"kind":"REWORK","_number":1,"created":"2024-07-27 09:54:20.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/59/925059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/925059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/925059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/925059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/925059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/925059/1"}}},"commit":{"parents":[{"commit":"d820df1067d03e5d911edfd6770a49ba36fb1f6b","subject":"Merge \"update constraint for os-traits to new release 3.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d820df1067d03e5d911edfd6770a49ba36fb1f6b"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-27 09:53:57.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-27 09:53:57.000000000","tz":120},"subject":"Bump jsonschema to version 4.23.0","message":"Bump jsonschema to version 4.23.0\n\nChange-Id: Ied4be4aee568e6d90e2f4264568f85ce20778eaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c60449dae96692e8ed8845800810b8b6a686785a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c60449dae96692e8ed8845800810b8b6a686785a"}]},"branch":"refs/heads/master"},"e3aabf46d4de0f7ac41c79350aeba3a785e913fe":{"kind":"REWORK","_number":2,"created":"2024-11-16 21:01:41.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/59/925059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/925059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/925059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/925059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/925059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/925059/2"}}},"commit":{"parents":[{"commit":"1372e0d4d66c0640f63be5176a1bd7a25eda39eb","subject":"Merge \"update constraint for python-manilaclient to new release 5.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1372e0d4d66c0640f63be5176a1bd7a25eda39eb"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-27 09:53:57.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-16 20:58:43.000000000","tz":60},"subject":"Bump jsonschema to version 4.23.0","message":"Bump jsonschema to version 4.23.0\n\nChange-Id: Ied4be4aee568e6d90e2f4264568f85ce20778eaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e3aabf46d4de0f7ac41c79350aeba3a785e913fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e3aabf46d4de0f7ac41c79350aeba3a785e913fe"}]},"branch":"refs/heads/master"},"a5576c7518e1e4d2cc7efe6762a260acef065e22":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-12-11 17:25:48.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/59/925059/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/925059/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/925059/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/925059/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/925059/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/925059/3"}}},"commit":{"parents":[{"commit":"1372e0d4d66c0640f63be5176a1bd7a25eda39eb","subject":"Merge \"update constraint for python-manilaclient to new release 5.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1372e0d4d66c0640f63be5176a1bd7a25eda39eb"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-27 09:53:57.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibizer@gmail.com","date":"2024-12-11 17:25:48.000000000","tz":0},"subject":"Bump jsonschema to version 4.23.0","message":"Bump jsonschema to version 4.23.0\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/937556\nChange-Id: Ied4be4aee568e6d90e2f4264568f85ce20778eaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5576c7518e1e4d2cc7efe6762a260acef065e22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5576c7518e1e4d2cc7efe6762a260acef065e22"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0eb65255ae9b831947c790b433e2be42b0000493":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-12-12 09:00:08.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/59/925059/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/925059/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/925059/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/925059/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/925059/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/925059/4"}}},"commit":{"parents":[{"commit":"1372e0d4d66c0640f63be5176a1bd7a25eda39eb","subject":"Merge \"update constraint for python-manilaclient to new release 5.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1372e0d4d66c0640f63be5176a1bd7a25eda39eb"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-27 09:53:57.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibizer@gmail.com","date":"2024-12-12 09:00:08.000000000","tz":0},"subject":"Bump jsonschema to version 4.23.0","message":"Bump jsonschema to version 4.23.0\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/937556\nDepends-On: https://review.opendev.org/c/openstack/placement/+/937601\nChange-Id: Ied4be4aee568e6d90e2f4264568f85ce20778eaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0eb65255ae9b831947c790b433e2be42b0000493"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0eb65255ae9b831947c790b433e2be42b0000493"}]},"branch":"refs/heads/master","description":"Edit commit message"},"97a8cf245605515587517431fe458ca19834bd17":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-12-12 09:08:45.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/59/925059/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/925059/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/925059/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/925059/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/925059/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/925059/5"}}},"commit":{"parents":[{"commit":"1372e0d4d66c0640f63be5176a1bd7a25eda39eb","subject":"Merge \"update constraint for python-manilaclient to new release 5.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1372e0d4d66c0640f63be5176a1bd7a25eda39eb"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-27 09:53:57.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"gibizer@gmail.com","date":"2024-12-12 09:08:45.000000000","tz":0},"subject":"Bump jsonschema to version 4.23.0","message":"Bump jsonschema to version 4.23.0\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/937556\nDepends-On: https://review.opendev.org/c/openstack/placement/+/937601\nDepends-On: https://review.opendev.org/c/openstack/masakari/+/937602\nChange-Id: Ied4be4aee568e6d90e2f4264568f85ce20778eaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/97a8cf245605515587517431fe458ca19834bd17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/97a8cf245605515587517431fe458ca19834bd17"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
