)]}'
{"id":"openstack%2Frequirements~925198","triplet_id":"openstack%2Frequirements~master~I0e433b1ba9ddce1917b9aa654f6f6ec885ff308a","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-07-30 17:37:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0e433b1ba9ddce1917b9aa654f6f6ec885ff308a","subject":"update constraint for etcd3gw to new release 2.4.1","status":"MERGED","created":"2024-07-30 13:03:39.000000000","updated":"2024-07-30 17:38:22.000000000","submitted":"2024-07-30 17:37:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925198-new-release","meta_rev_id":"d65019850bd934e45fba444228dd883a188b35bb","_number":925198,"virtual_id_number":925198,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-30 17:37:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-07-30 13:40:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-07-30 15:59:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2024-07-30 13:09:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-07-30 15:59:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-30 13:09:11.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-07-30 13:09:11.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-07-30 13:40:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-07-30 14:37:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-30 15:59:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"cec0e40edd3b9a5975595dca64006f4defd72a4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-07-30 13:03:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ac6847abbe5a84d8ccde436425b8a74df07bfac","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-30 13:09:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b24bc2a970ce052509b63332b3d4a1f51cc80bbf","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-07-30 13:40:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0587243985dc4bbb32e262a04a45f02c59309830","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 14:37:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a28b9a2d076441ca95aa5942552114a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4248cf987f4b49f3a76ef20b7a0e03db : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2f643b6b62f4d1c804574cb8ad2d350 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eecd92bf5aed4caca3bc1f02b9e39b2f : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/37f28280971c42ada59543a5803c6c9e : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c00240234ccd4dd68fb8809b30be4c3f : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a516c0f3f3ea4ba696d18a66712250b4 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fe76d347eb3d4932a7a12144cb91fbb0 : SUCCESS in 2m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dee56a999be14fae8b9947c12838df33 : SUCCESS in 9m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e79b7561e99f483a9dfc54ee684c86d5 : SUCCESS in 13m 32s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b86f100222ac416cb1f996d9684e3d58 : SUCCESS in 9m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/067eb6ee2beb4cb48ea4d7200fceb84c : SUCCESS in 14m 37s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7aa163b8096340e982cc0982e3919065 : SUCCESS in 36m 16s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/14498e65478648aaa63dfc57f3d5d3bb : SUCCESS in 4m 42s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/5b1cceb476fe42ea944407a2f44e575b : SUCCESS in 5m 34s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/ea5b25fe80d14faea68a27c467d3ac50 : SUCCESS in 4m 52s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/4344cd91a5c344638ecd233f174bf139 : SUCCESS in 13m 35s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e969f6980284483e8f5b00a14c0fc154 : SUCCESS in 5m 48s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/107537b2bfab44429dae37d9865dc2bf : SUCCESS in 5m 52s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/03b06ac6f76f445f825f190d0ae05d4a : SUCCESS in 8m 17s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/932897103b374ae1a0d47b53039a8d68 : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6489951612a24238b726a0bec6d6bcf8 : SUCCESS in 7m 38s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/d57cb636a4e5460dbb1460268954df32 : SUCCESS in 8m 28s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/c2ca5fd030a642c88e899f462c54fd40 : SUCCESS in 11m 18s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/7d986f17d41242419d108679b5ee53aa : SUCCESS in 3m 43s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/78227fb0d932441784529640ca4a47d8 : SUCCESS in 9m 32s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/536d3739367d4392be129bc3263bbb7c : SUCCESS in 4m 44s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/195c065316ae49348fce5bb1e709cca6 : SUCCESS in 4m 59s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/142e87ef02e440f3be078c5b24df89ed : SUCCESS in 9m 13s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/77b567ea23b74eba8fff6acdf4ebf286 : SUCCESS in 20m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dfc183dfaba043d69bc8ce3b5bc2a6ae : SUCCESS in 7m 34s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/b619eeb7572942708d1b23f6f9356f81 : SUCCESS in 10m 22s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/ffbca0b280c74b378b0dbb7978f315c9 : SUCCESS in 4m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/988ff91e0f384218812a38b3823a0fd0 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9b634f35e60e40bf996ba471160927ba : SUCCESS in 24m 11s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/a9fd370278d64f08bff9fdb7542b09e0 : SUCCESS in 3m 41s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/36282199b4954de4ae4dc8e246f5a959 : SUCCESS in 4m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a39526b24cbb41678d718c844986a655 : SUCCESS in 5m 30s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/3a3c20a0d91a4fc08f5b991fe9d0f09b : SUCCESS in 17m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/edf7bc2319bf4f7faa3244de717966c7 : SUCCESS in 1h 07m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/019af57f561f48289b26e506c4769e93 : SUCCESS in 1h 32m 31s","accounts_in_message":[],"_revision_number":1},{"id":"a62055a04dcd4e0ba5099fe413d1370ab4de9cc2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-07-30 15:59:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8e56b02f91a8fd1b79bba6ac432940a2aeaebdb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 16:00:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1168e437ee4a6897c1d900aa6f5b8aa7555c6522","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 17:37:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5936836309264e62bbdf57011761f5b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3554b80ed8d410f87fb3a797c4ae656 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d71e438c0ba427db2b0085305dd2010 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd46e075e9b848f398430ed12525d671 : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e8fc1a70672742379e9d95a325fe35de : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3fdcdb85c5924279a6401f74cdb41c42 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/28862cd64a62456b8943fb901bacaa0a : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/48e64829589746b4addce2cd39b4063a : SUCCESS in 11m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/89fb03c3bf9b40ac99bf887f355aa8a3 : SUCCESS in 14m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6fa960768f6b41eeb3a9055b835d1167 : SUCCESS in 8m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9b1dafcf1ba9441e9502ffb26e633e6f : SUCCESS in 19m 12s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/93f3b549d5f340ada8c644e7631dd907 : SUCCESS in 11m 12s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/108477770eb44cf8bac7e8bbf560af3f : SUCCESS in 5m 25s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/a2bd46b6312445e4a5c067f8fbc7a1b5 : SUCCESS in 6m 57s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/f89794edae8f4f23b76dcc302683f2c9 : SUCCESS in 7m 49s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/c0ca5f4751614093ab79af27b06d54d1 : SUCCESS in 13m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/58099391b1154b46a810b90a51628af4 : SUCCESS in 7m 17s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/724c57502d694696b3eecf9e69f870db : SUCCESS in 9m 04s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/295b07b04c924d37937566b9f3cefe6d : SUCCESS in 12m 31s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/9a9e9789369c4fdc9fb17fde3af05465 : SUCCESS in 4m 01s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/6385c08ac94b46dcb1b5596c52e08d05 : SUCCESS in 9m 47s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/1155e98eaa0a4c6abe53d3225f82ee41 : SUCCESS in 20m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b59ad1f7a22d49e0b6378c42fcf4a6ce : SUCCESS in 8m 23s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/26e4625d5f094c7c89dc6a9c41ad63ca : SUCCESS in 11m 49s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/f915f55bc95a465fa74c4f88f00d78d2 : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b96a3c3abbd44ee68f1878c28b559878 : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8b7262f0f97b4e7a8c3e7837d0f9ce47 : SUCCESS in 22m 01s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/f323ee733a7f4cbead8344a68a86b1de : SUCCESS in 5m 24s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c0ed07bc64fc47f381ac5240fcfc16fa : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a2d99657972d4d9bad4b588649029f9c : SUCCESS in 6m 09s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/d6dcb4ac3535462abef1747ac18ae37f : SUCCESS in 13m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2fba718931a44ce7b9fa65c283c85dd6 : SUCCESS in 49m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8502883e6f7b4e3f8bf12783bb04846a : SUCCESS in 1h 18m 51s","accounts_in_message":[],"_revision_number":1},{"id":"cdaf0bf29a3b58162142ccf03407725da2591e4e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 17:37:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d65019850bd934e45fba444228dd883a188b35bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 17:38:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8703a36cbb5b4b92adeb5ff00c1f89ce\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f51f4ff319f4a82b2bbbf63d275f4dc : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c279bd469df43206693b27fe85565028346f042","revisions":{"2c279bd469df43206693b27fe85565028346f042":{"kind":"REWORK","_number":1,"created":"2024-07-30 13:03:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/925198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/925198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/925198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/925198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/925198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/925198/1"}}},"commit":{"parents":[{"commit":"d820df1067d03e5d911edfd6770a49ba36fb1f6b","subject":"Merge \"update constraint for os-traits to new release 3.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d820df1067d03e5d911edfd6770a49ba36fb1f6b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-30 13:03:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-30 13:03:38.000000000","tz":0},"subject":"update constraint for etcd3gw to new release 2.4.1","message":"update constraint for etcd3gw to new release 2.4.1\n\nmeta: version: 2.4.1\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I5528bdcbf8e2c43403ce7e9981a96b65a143ea8e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: I0e433b1ba9ddce1917b9aa654f6f6ec885ff308a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c279bd469df43206693b27fe85565028346f042"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c279bd469df43206693b27fe85565028346f042"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
