)]}'
{"id":"openstack%2Frequirements~925306","triplet_id":"openstack%2Frequirements~master~I15d3a4858383d7e0acc65e62b886c584bbc363ab","project":"openstack/requirements","branch":"master","topic":"add-importlib-resources","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2024-08-28 14:10:09.000000000","reason":"removed on reply"},"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2024-08-28 16:18:11.000000000","reason":"Change was submitted"},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-08-28 16:18:11.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-08-28 08:36:18.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2024-08-28 16:18:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I15d3a4858383d7e0acc65e62b886c584bbc363ab","subject":"add importlib-resources to replace pkg_resources usage","status":"MERGED","created":"2024-07-30 23:33:11.000000000","updated":"2024-08-28 16:20:07.000000000","submitted":"2024-08-28 16:18:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925306-add-importlib-resources","meta_rev_id":"7660ec2d5431ccf9b38d4a11a3f4e5a25b697fd7","_number":925306,"virtual_id_number":925306,"owner":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-28 16:18:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-08-01 14:45:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-08-28 14:10:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2024-08-28 08:36:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2024-08-05 16:15:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-08-28 14:10:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-31 00:37:44.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-07-31 01:19:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-01 14:45:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-08-05 16:15:26.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2024-08-06 18:26:13.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-08-08 05:27:52.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2024-08-08 05:27:52.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"a21de0b5f4559dbdccfa225b6688b3e949514a95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-07-30 23:33:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a64ac41fef1870756c4d7b2459fb7ae4204ca6b3","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-07-30 23:41:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb978409649b41f9c93028fae1e431b16b9c2b79","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-31 00:37:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f1cea6deb0ba5e965699f0b5000744fc557048a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 01:19:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9735b601706a4cccb8a1047b47e83f8d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41f36fb893cb4eb1abc478aaf22814a0 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a9b60a7f9c545539cb08b65042ef60a : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9facdc0156a844b78f6eee0efa08ecd6 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30ab5c52ab6e4b1d89321f65bb02d04e : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4bc349854fa245479ec27342ffdbe563 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cdbde4328d1448b0a5fd28366bf4e75e : FAILURE in 2m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b02f384ba1144b419a7f923dab008871 : SUCCESS in 59m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3475559b68024afd889272e7eb2e9583 : SUCCESS in 1h 38m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e1157bcc7fd497c69b8dab6943f120084b4e1f96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-07-31 01:35:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f36a78168755501bdda88c30c91e00e7a168d268","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 03:28:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/80e7d90cd8fc4a8583c09c0c80f69684\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da955859482a40c68f5e741df6fcf9e5 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc545db65fb6470a8deff79ba6a56285 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2990f04461bc486f82fcec140bb19bd3 : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/038685f59aff4bba8fb96683e1f916de : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e8fab8ff2c854fd7aa6870fd92d90ebe : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7b072050292a4e8aae9c47f31f287375 : FAILURE in 2m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d237111f2534c6ab1ed3bbe318bb5af : SUCCESS in 1h 12m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8e1c5cd714943f1ab8f824f43301fb8 : SUCCESS in 1h 46m 09s","accounts_in_message":[],"_revision_number":2},{"id":"aa1da05f1debd0780d1b386b95d958758483756d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-07-31 20:50:26.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1511b78cd5eb94aa9fd8e5e2479a2613377ac7ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 22:15:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34153313899249a2bcd4275f06f41002\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/984e9d81fba348cdac555dd437749daf : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8503184e700d45329a900192b75030e6 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a1feaccacf34211ba925bb505990fb6 : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0652651e83a045c4ab3b86fa42930a53 : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fbd7dd080de04f91b4bad65246bd45d7 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f692d24c3d364ed99cd2e85f4d17aa50 : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d0b341ef741f4dfea603fce3bccf08c6 : SUCCESS in 2m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8f6b8c70a8f04dc2ad851bdf124f81b4 : SUCCESS in 7m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1454ffe6fa6f4880905d62e3bb4d0286 : SUCCESS in 13m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/31824ac270fa474e8e0623966e22ff04 : SUCCESS in 10m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/60829b8db84146bb84f72c9ad2618b74 : SUCCESS in 14m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cb46c046dfd74002a97d0ad32b1cafe9 : SUCCESS in 32m 32s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/01743b9bdf114451928a8415efb0f029 : SUCCESS in 5m 39s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/6d5c6ba65af149e9b4e958275e46209e : SUCCESS in 6m 00s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/0a27cf7ad34d428dbfdf17ae17043eff : SUCCESS in 3m 32s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/ffa7d1c1cb5c45b38086b2a2f979de67 : SUCCESS in 10m 45s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/288623be3deb410ab7e5673e2793563f : SUCCESS in 5m 39s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/d9b86895068746ff875f5d8ce7a4589d : SUCCESS in 4m 01s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/ebe80ec24c024eee88195c910852a1a3 : SUCCESS in 6m 55s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/7934ecaac1ad43ceb3d522577e4d3b0f : SUCCESS in 13m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7f4ca692b528409f8b798c34994c5ef5 : SUCCESS in 6m 46s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/fe3b29ff05124743abaf282c0b8017d3 : SUCCESS in 7m 16s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/b51bbaef810c48fab0731a520d255157 : SUCCESS in 10m 51s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/aeee6f1513fc4d3ea3a638474c032a09 : SUCCESS in 3m 24s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/5552eff02ccd416eab5ac0e02a2e6468 : SUCCESS in 10m 25s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/81fd0c8ed5574a7d8847a2930330e1e3 : SUCCESS in 3m 46s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/9083252aee01424186039dd81cc9c188 : SUCCESS in 4m 25s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/53902afb96b24551b215299e30b9d8da : SUCCESS in 14m 23s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/2575f4d1308e49f6aa99acbefcba2c1a : SUCCESS in 20m 54s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7d522d2011de41e18ae8743f6d4f2734 : SUCCESS in 6m 40s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/e6e688d5a03649d381717d0d1c69a51f : SUCCESS in 12m 58s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/5bc32dbc577d4fd7867fa63e39a91b08 : SUCCESS in 3m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0a5e0d4368c0418f92aac6f372a2407b : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a59e1209bc9945628d4c30bd0ac2ae5e : SUCCESS in 20m 14s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/4d2c68690cb04f1bb6b640d21f330f13 : SUCCESS in 4m 09s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/7a4b5b9625a44ffbb18ef05ea66382be : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6c61ca9db01245a5978c527b05da6de4 : SUCCESS in 5m 26s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/92e555b281184a95b50b404599bf29b5 : SUCCESS in 10m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b6e4bb9c0434ba28dbe5ef2e7426337 : SUCCESS in 52m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e74a83976f24a88a8058899e2c89c45 : SUCCESS in 1h 21m 01s","accounts_in_message":[],"_revision_number":3},{"id":"068b7b1c2a27098e27c2800d214a6c87525508c3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-08-01 14:45:25.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"929af22f45f522a19d9640fe349a85f5990cbe9c","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2024-08-05 16:15:26.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4a063ef0ea496cb7bcedea9841826dd4af76d7c5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-06 18:26:13.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"66abef9bfc245ccf84dd9d7bc195aacf6570d1ca","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-08-06 18:35:58.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"af53852717d97cf58ed322449a5367a27a6d77e5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-08 05:27:52.000000000","message":"Patch Set 3: -Code-Review\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0738a4e1e83727ee8d3082839e2434601addcaae","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-08-09 17:48:07.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7623ee28f858579ccd633a901a72f24f5aec5dad","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-08-22 14:52:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9de2d248f191fe08e9a784dae00d78117d458584","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-08-22 15:28:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"12d6f9b9b4964ecdb467ea3795b6b0b582f7f4dd","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-08-26 15:25:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6c9006080bf11e0ef92d0be21f4e838fae70b1b2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-28 08:36:18.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0ebadfa054af5d4c652480921e892e9b31661bce","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-08-28 14:10:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a155869c275e708751ece3b0b946728839a0cf1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 14:11:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5a8c4adddece5728cc5da913df488788c2d96d1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 16:18:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73cdad549d2e435cbfa5a3cefb900b1a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d657bb448684097994432a8b9b581cb : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/667dae5f2e204dfa90aa03a0a194d671 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a448443f3384b6680e5c187af85e680 : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd8ee2f9b1d3449ca5d8b76be5db615e : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f25bd6b70451454d9cb9f8dd8a764067 : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4988e4dd353e4fdbb20b6b204372e2a3 : SUCCESS in 4m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/475a72f1458248fdb9cde582f0f9bd50 : SUCCESS in 8m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9840039273a54439a7ed5644d1a42b41 : SUCCESS in 13m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/bd0beb28a5964ee3a92f61481499aec6 : SUCCESS in 8m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6fa7459f12aa4057b41529d4d9341be5 : SUCCESS in 19m 31s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/55123d3dcbf84f3b9715f96dfdf449d5 : SUCCESS in 11m 14s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/71a258ea63394fb1b9af5e203ecdcb82 : SUCCESS in 5m 32s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/a39c0ff79bbb46198ae57589c4a8d7cb : SUCCESS in 5m 41s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/78d6b838fd9f49c490af6dc0c937c667 : SUCCESS in 7m 08s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/4966c017a73c4a0590a585f89cabb2f6 : SUCCESS in 12m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d876217a14eb464d94a913569bcde6c3 : SUCCESS in 9m 05s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/4db4ac86f534406bbc5c1653adf8842a : SUCCESS in 8m 30s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/fe4661a96b0c44d5bd488110a8b55665 : SUCCESS in 11m 40s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/9b9c06336dc84c4ab3de6bbbb841d5c7 : SUCCESS in 3m 42s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/f8086c2013f24c9293f5d02ee9dddf17 : SUCCESS in 8m 48s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/6e9af6be03564c858e2a76bdbdaaf315 : SUCCESS in 20m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f024bb28377b46f49f99b7086508d0a2 : SUCCESS in 7m 41s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/92fb5ded5a394743a3e18d3876977540 : SUCCESS in 12m 33s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/157f613b18ba431a935db38a0b944cee : SUCCESS in 4m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c7b88d69e7e144d69b9bb58dab456854 : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d203f85cd2094e969648c964ffe25d50 : SUCCESS in 26m 26s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/b7c23c42c4b645ab9b7163f7d1c8a8f4 : SUCCESS in 5m 07s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/b252dec37ee94cd38fee94c3cd64b9b5 : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/58db8e376991467c95b68e25fedbee78 : SUCCESS in 6m 05s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/67ce342c3ab243f0a02809ef45c48b34 : SUCCESS in 18m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/088a5309c0d1407484f7e18391166fd8 : SUCCESS in 1h 01m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d716d90d36de4572a168a5c55de69d14 : SUCCESS in 1h 29m 01s","accounts_in_message":[],"_revision_number":3},{"id":"dc835bade175444da45fe5e755e95f7e4acd7e3f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 16:18:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7660ec2d5431ccf9b38d4a11a3f4e5a25b697fd7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 16:20:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84dcc5b4f721403d9c085e451e9ca146\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e095466a9168446b9ce17675e49bb965 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ed046c41533a13e6acccbf8f4409ba23a71cfd6d","revisions":{"a8b08f66e4229a00da57dddc7c310da827b95b12":{"kind":"REWORK","_number":1,"created":"2024-07-30 23:33:11.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/06/925306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/925306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/925306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/925306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/925306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/925306/1"}}},"commit":{"parents":[{"commit":"2c279bd469df43206693b27fe85565028346f042","subject":"update constraint for etcd3gw to new release 2.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c279bd469df43206693b27fe85565028346f042"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2024-07-30 23:28:16.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2024-07-30 23:33:09.000000000","tz":-300},"subject":"add importlib-resources to replace pkg_resources usage","message":"add importlib-resources to replace pkg_resources usage\n\npkg_resources comes from setuptools which starting with Python 3.12 will\nnot be installed by default into virtualenvs which will break projects\nimplicitly assuming pkg_resources is available. Additionally setuptools\nhas deprecated pkg_resources so the best plan is to use the stdlib\nprovided solution, which is not available until Python 3.9 so we must\nuse the backport package for Python 3.8 support.\n\nChange-Id: I15d3a4858383d7e0acc65e62b886c584bbc363ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8b08f66e4229a00da57dddc7c310da827b95b12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8b08f66e4229a00da57dddc7c310da827b95b12"}]},"branch":"refs/heads/master"},"4e2d2afe7fa307eedaacf814644073d06bc0924f":{"kind":"REWORK","_number":2,"created":"2024-07-31 01:35:26.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/06/925306/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/925306/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/925306/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/925306/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/925306/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/925306/2"}}},"commit":{"parents":[{"commit":"2c279bd469df43206693b27fe85565028346f042","subject":"update constraint for etcd3gw to new release 2.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c279bd469df43206693b27fe85565028346f042"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2024-07-30 23:28:16.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2024-07-31 01:35:19.000000000","tz":-300},"subject":"add importlib-resources to replace pkg_resources usage","message":"add importlib-resources to replace pkg_resources usage\n\npkg_resources comes from setuptools which starting with Python 3.12 will\nnot be installed by default into virtualenvs which will break projects\nimplicitly assuming pkg_resources is available. Additionally setuptools\nhas deprecated pkg_resources so the best plan is to use the stdlib\nprovided solution, which is not available until Python 3.9 so we must\nuse the backport package for Python 3.8 support.\n\nChange-Id: I15d3a4858383d7e0acc65e62b886c584bbc363ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e2d2afe7fa307eedaacf814644073d06bc0924f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e2d2afe7fa307eedaacf814644073d06bc0924f"}]},"branch":"refs/heads/master"},"ed046c41533a13e6acccbf8f4409ba23a71cfd6d":{"kind":"REWORK","_number":3,"created":"2024-07-31 20:50:26.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/06/925306/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/925306/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/925306/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/925306/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/925306/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/925306/3"}}},"commit":{"parents":[{"commit":"2c279bd469df43206693b27fe85565028346f042","subject":"update constraint for etcd3gw to new release 2.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c279bd469df43206693b27fe85565028346f042"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2024-07-30 23:28:16.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2024-07-31 20:49:25.000000000","tz":-300},"subject":"add importlib-resources to replace pkg_resources usage","message":"add importlib-resources to replace pkg_resources usage\n\npkg_resources comes from setuptools which starting with Python 3.12 will\nnot be installed by default into virtualenvs which will break projects\nimplicitly assuming pkg_resources is available. Additionally setuptools\nhas deprecated pkg_resources so the best plan is to use the stdlib\nprovided solution, which is not available until Python 3.9 so we must\nuse the backport package for Python 3.8 support.\n\nChange-Id: I15d3a4858383d7e0acc65e62b886c584bbc363ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed046c41533a13e6acccbf8f4409ba23a71cfd6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed046c41533a13e6acccbf8f4409ba23a71cfd6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
