)]}'
{"id":"openstack%2Frequirements~925462","triplet_id":"openstack%2Frequirements~stable%2F2023.1~Idaa00004331bcba019b9ec05c28178a908737ce8","project":"openstack/requirements","branch":"stable/2023.1","topic":"new-release","hashtags":[],"change_id":"Idaa00004331bcba019b9ec05c28178a908737ce8","subject":"update constraint for python-cloudkittyclient to new release 4.7.1","status":"MERGED","created":"2024-08-01 10:37:31.000000000","updated":"2024-08-03 04:00:22.000000000","submitted":"2024-08-03 04:00:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925462-new-release","meta_rev_id":"0a3d9f1889001ccbb70ee1f8ad6453e606512c6f","_number":925462,"virtual_id_number":925462,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-03 04:00:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-08-01 14:46:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-08-03 02:18:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-08-03 02:18:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-01 11:33:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-01 11:52:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-01 14:46:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-08-03 02:18:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"7605d2975138e218dcccfd80a41c2382a83a9420","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-08-01 10:37:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f39ab5cc29ec9ded42b85bc7fac63dfb724dd58c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 11:33:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53f916566bee48b6b3f4427b4a6a4f0d\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d800ce2e5aff4f14aeed9ac90e9ae684 : FAILURE in 12m 09s","accounts_in_message":[],"_revision_number":1},{"id":"508b6f6ae70f98aafb5486986b738ed2ce6967c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 11:52:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2ff939491ad44c3a7f1ce26c6c0acbe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/619446e562dd48959c4f30423f279028 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/734fcb58514e415c80f86026352dd66c : SUCCESS in 5m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c52bee221ead4c028e198ed8167ba77e : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e84af7e6fb7a4b078cadefb1789daf76 : SUCCESS in 3m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/89ddb54f315d43af8c55bd070423e941 : SUCCESS in 9m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/41adecf299e24f58b649ef4b7b9c27da : SUCCESS in 17m 31s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/d2fa056f5f1441a1a431d3650abe2215 : SUCCESS in 6m 57s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/0735c5b7df344b96999905bf5386d7fa : SUCCESS in 5m 12s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6c8fb472ff4947829cb2fefa5809fa9e : SUCCESS in 4m 37s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/a74d378034064bad979028e5c726a713 : SUCCESS in 12m 10s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/0430ed87380f4d09be24034ef1e99c9f : SUCCESS in 3m 59s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/26e8c226eb604d61a34ae1ceca46ff61 : SUCCESS in 4m 54s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0c47efa1842945b39fd031d860332136 : SUCCESS in 7m 54s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/772b89c6694a4b8b857126e1502b6cfd : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/92e49ac87f5340009b87a56845a477a9 : SUCCESS in 8m 10s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a28450bb43254a899addf66f62275c82 : SUCCESS in 8m 51s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/7e9b78a370104d69ac1b4dbfc69b37f8 : SUCCESS in 18m 07s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d6a66b82029c42b68ab1c72cfa547037 : SUCCESS in 3m 35s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/4ad959bb7a8d41dcb1cb87284dcd67c4 : SUCCESS in 17m 07s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/a7bdfe9c1342413c9259b5a15f9fa987 : SUCCESS in 4m 00s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/3ee54a1171074441a87f1af58e13033e : SUCCESS in 9m 56s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/318577f7b04d43a0b6aa594ad8206408 : SUCCESS in 24m 15s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/5df0db594a694ed4aa77084b340d16e4 : SUCCESS in 15m 01s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/ad156bdbeaec44899f5389000d9cf48b : SUCCESS in 5m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9e9cd3af54d94a34a0f1e3ee30162f13 : SUCCESS in 2m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c1b14e261347486f98ca29317c9e0770 : SUCCESS in 28m 02s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/6272221399464db4a55d31757587493f : SUCCESS in 6m 37s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/a4dda086a4a24be88073003b63ffc818 : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c805c84a501142a5b6bf1e194d7f0dff : SUCCESS in 6m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/18d4c2de8f6c403797495e4f10be9204 : SUCCESS in 12m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5383c7ef93742d4a1bd251c76a45754 : FAILURE in 53m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5b0a168351242a69f8d5a82621fa780 : SUCCESS in 1h 13m 05s","accounts_in_message":[],"_revision_number":1},{"id":"9e8df16adebaaf435aee20e075a28d8b0ca2d0e0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-08-01 14:46:41.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e6dba03c6751e759b0f4afa3d8fb2ffbbab377c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 14:59:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d66e98533ae041259be138b748bc5f82\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6f7031944fd647cbb067a1e1662b5d78 : FAILURE in 9m 12s","accounts_in_message":[],"_revision_number":1},{"id":"8d6b2bb4298dde53407ecfa5f3ad97b0ae0bd070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 16:12:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/032289f29d654fd3b52e20eeaa085406\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9fd9311457d41c0ba130c6bda35d233 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ed49e17d6fe45b0a930f2658babda55 : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d1579918b364292a057a1f11ba6f3d3 : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/881cd9312c63469eaf358bd0fe1b8de5 : SUCCESS in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/02d57e7a050147a6baba8e2f0f0c3aee : SUCCESS in 9m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7bb159a53efc430b8b86dcd47af2777e : SUCCESS in 17m 16s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ce4aeab2dc294ca88ca348feb0b3eba1 : SUCCESS in 8m 11s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b1608ee767b84f2e95cc768b5fc29d4d : SUCCESS in 5m 15s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/347e6847be5f4932ba7fb2c2c5a74a0d : SUCCESS in 5m 52s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/9dfc5af0e7bd44fba075f0522336aafe : SUCCESS in 10m 29s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/88748643a31d49599f0c3ec3671fb444 : SUCCESS in 4m 20s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f2503111793f46ec9aa2019f65501b4f : SUCCESS in 4m 48s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a6b6c1b82794402cbddf4444817df3fb : SUCCESS in 9m 04s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a09fe1db5a2b4a0f95bcec51c17349e1 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5695914242f843d49d1d1a6ee95edf46 : SUCCESS in 8m 53s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c56bf52847bd46abbba5de7381784e1c : SUCCESS in 8m 25s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/a6c73a362f17475d967c428c5a50bb2b : SUCCESS in 11m 10s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/18bfa44f26c84d4a904a3f9d0fa16176 : SUCCESS in 3m 17s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/5b264a7c74ed4ede8247b0b5c4b18663 : SUCCESS in 10m 51s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/1191517d78ea437cbe1a8147655a3021 : SUCCESS in 4m 23s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/7fc120e8b2974c889708fcbaf7592c32 : SUCCESS in 11m 07s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4cbd80bc73794bfbb2bad28b4293c322 : SUCCESS in 22m 07s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/0ef55067e59b4656b9cdc59eaa417e97 : SUCCESS in 13m 39s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/cf0ac6bed8444c96a25bcdccdaf81382 : SUCCESS in 6m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/98f534a0e91f47279bd6a4a02defd45f : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fd256a9ae8ce46a0acdb577d593bc47f : SUCCESS in 25m 26s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/2d7dde7e6d3243958043167924242e60 : SUCCESS in 5m 13s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/877dc5dab9964c67a43808797633eb76 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8cdc48be3b4d462087ae33094050819c : SUCCESS in 5m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/984b9472c63940c3b7030179ae78cb43 : SUCCESS in 12m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c63d05f26d2445aa8425775c4f51d320 : SUCCESS in 1h 07m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/471c9bf85d564e0883c212cf0c409745 : SUCCESS in 1h 21m 35s","accounts_in_message":[],"_revision_number":1},{"id":"bb1eec8ee246d8911b3c79a9a70fc034eb64d961","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-08-03 02:18:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"61990b755fdb858dbc7274a555486739a8060553","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-03 02:19:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85daaf99d51de5d11295f83a4a37b28ac0e59554","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-03 04:00:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cf347eebe554bf9a5c46acb39e3cbc3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21ba6ae694ac482bb97f8e233b4dca7d : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44b437f1cb4b489794d1991aeb23385f : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ed6ccd82faf466e9097352f511a0995 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d0e8b8826e04a018996a0d858a78933 : SUCCESS in 3m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f9e947aa86764f3ca3781165c9d10be6 : SUCCESS in 12m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ba73df5e720a4338be37d41158112a29 : SUCCESS in 12m 17s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/8ce97160c3404edba6abc90325ca2152 : SUCCESS in 10m 42s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/71dcb10667cd4b009becc694bc4b02f7 : SUCCESS in 4m 58s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/87b0635e9c364f3aa1c6a80488d06758 : SUCCESS in 5m 33s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/eac553307a1f4a57ab20ffeeae96431b : SUCCESS in 8m 27s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/cd6ff7ffa3cf42ed82b77b49a1a676cd : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5371fe156ff44d72b5a91e96f1383f7b : SUCCESS in 7m 47s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/c860f7de70dd49fcaf8fa670006e18c7 : SUCCESS in 8m 43s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/308dd8c72bae48a3b40028ba451e8a71 : SUCCESS in 12m 54s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/d090c24decef443f9498d6cc84380df9 : SUCCESS in 4m 11s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/1ec3e38e53b04d3986d54cae2bf1f7c4 : SUCCESS in 9m 24s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/053660ae616a4d0ab86d44802a08c175 : SUCCESS in 6m 16s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f23848e289a444858644b799a04f697c : SUCCESS in 23m 32s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4d910bc4690c4d01ac61df5f3cf619c2 : SUCCESS in 13m 35s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/cb208e290f0d4d4abfc0eb1e03b8f1bc : SUCCESS in 5m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/881008b12904473ab9467fe2c3623fea : SUCCESS in 4m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b604aae836f243708eea9969a145b59b : SUCCESS in 25m 00s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ecd52e272b484693bcb15eb53166e53e : SUCCESS in 7m 56s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3cb16db15fdf4eb0bb51d258a13068c7 : SUCCESS in 6m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e6e175e4123d4ea4b33bce49c7fecca2 : SUCCESS in 5m 38s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/995c34e25fe44cf584eed29a088bf815 : SUCCESS in 13m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a4123b028b04185960223a6d0b76263 : SUCCESS in 1h 01m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b0e0f3e509a47e8bdfe8715c3cb05ea : SUCCESS in 1h 21m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0a3d9f1889001ccbb70ee1f8ad6453e606512c6f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-03 04:00:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f96b52c621750fa702c88738479ff66873869e71","revisions":{"f96b52c621750fa702c88738479ff66873869e71":{"kind":"REWORK","_number":1,"created":"2024-08-01 10:37:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/62/925462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/925462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/925462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/925462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/925462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/925462/1"}}},"commit":{"parents":[{"commit":"487812fc17be660c3b977464f6aacc40c215696b","subject":"Update packaging constraint to match 2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/487812fc17be660c3b977464f6aacc40c215696b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-01 10:37:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-01 10:37:27.000000000","tz":0},"subject":"update constraint for python-cloudkittyclient to new release 4.7.1","message":"update constraint for python-cloudkittyclient to new release 4.7.1\n\nmeta: version: 4.7.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: branch: stable/2023.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I389625d547c96c720f5939ae11146cd5193b3acd\nmeta: release:Code-Review+1: Rafael Weingartner \u003crafael@apache.org\u003e\nmeta: release:Code-Review+2: Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Idaa00004331bcba019b9ec05c28178a908737ce8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f96b52c621750fa702c88738479ff66873869e71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f96b52c621750fa702c88738479ff66873869e71"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
