)]}'
{"id":"openstack%2Frequirements~925464","triplet_id":"openstack%2Frequirements~master~I2632e5e1d277f520be14add37b8a201ad2f5daf2","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-09-03 18:55:58.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-08-22 02:36:06.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I2632e5e1d277f520be14add37b8a201ad2f5daf2","subject":"update constraint for oslo.policy to new release 4.4.0","status":"MERGED","created":"2024-08-01 11:11:26.000000000","updated":"2024-09-03 18:57:01.000000000","submitted":"2024-09-03 18:55:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925464-new-release","meta_rev_id":"6a468843bcea1a374decf3e6740d5bdd7d4f19d9","_number":925464,"virtual_id_number":925464,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-03 18:55:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-09-03 17:14:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-09-03 17:14:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-01 12:38:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-01 14:47:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2024-08-01 18:12:25.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-08-27 21:02:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2024-08-30 13:38:45.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-09-03 17:14:38.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"08b87a65d48ce9188a4bceb26d6d25e35cc41c7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-08-01 11:11:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6af07caeb0c9a4c691c17af6ea0a9e3a6d057fb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 12:38:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f58b24a1da9f481a9d6b6ad9c5403454\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1964259159774e63a8cc0a803799c10f : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f836d89b2da145e085747b61031ccbb3 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0a084ba2dc149e4a274410cc88ea3d3 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7cfef904784d47b3ae102d34a24d8253 : SUCCESS in 3m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3273396cce6b48b79a9e3c71524f0fda : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ae809475d134628a7acfb0170a7d7f1 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0094cdffc77048c48bda954767838baf : SUCCESS in 2m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/caa40f32639b44e596717329cdbb8eea : SUCCESS in 8m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8ff01040c02640cab68999e78a138f0a : SUCCESS in 14m 43s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b31b2163aa33467f83b15d2c25b86a9c : SUCCESS in 8m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/13f6ccbf943d4ac087cef12707c70948 : SUCCESS in 17m 26s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e5d9c3d6422842a4bb72c30f780a19ba : SUCCESS in 29m 40s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/b0e63b61ed2140e4b95d60afdf376ea6 : FAILURE in 4m 59s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/26e3d31b57a4450bbbf0c01f9aabce57 : FAILURE in 4m 35s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/0c6a76132ab0437ba57c2f274147b3dc : SUCCESS in 3m 47s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/a7cb59c7e2f048119bd234b1ad45f102 : FAILURE in 11m 09s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/8e4a44e4109c4673927de92a18840e84 : FAILURE in 5m 29s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/998224a93f4b4cc9a40ca53ddc7c4119 : SUCCESS in 5m 14s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/add36a7d24a648ed99b8a165b78417b3 : SUCCESS in 7m 03s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/7c0586b324294735b232b92e206824f8 : FAILURE in 12m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/10fc507eedcd42db9215db81ce9bb5c6 : SUCCESS in 6m 14s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/c2da04517ac84a3fb4a81da454a12624 : SUCCESS in 7m 51s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/1b384e50042343289de9842d9d2e5f3b : SUCCESS in 10m 19s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/a7af5e04301b46ba9439486d854d0a5f : SUCCESS in 3m 31s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/20b7677f37224ba89d5f28234e6ee3a5 : SUCCESS in 7m 37s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/56522cf55e5a42f5bb638f6adab1a456 : SUCCESS in 3m 58s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/c99e70ff3e7a4f10abef9acf4b45f25b : SUCCESS in 4m 21s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/f7ad77f7e9bd44afa48b68acb7520c25 : SUCCESS in 10m 15s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/3b70b3ab2b4941a78ec00eb5013b6b78 : SUCCESS in 18m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c94fe8c7e87c47c1be087810055c4c83 : SUCCESS in 6m 36s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/4104b20ca91c4e6e986855e8c23a54da : SUCCESS in 11m 00s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/5e543bafc4fc45c59fb1f7e4ff423691 : SUCCESS in 4m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ecbaf3bbb2b04939bc36cad0362b704e : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4378f8a1f0b24c0fa51feae1263572b6 : SUCCESS in 21m 37s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/1b321daab39b46f99ffac5953437ec06 : SUCCESS in 4m 31s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/e9b55eb051d246a4a430c43bc2eb43ec : FAILURE in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8dc60a7b0319458a8d9fd7fde78c7e9f : SUCCESS in 5m 39s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/09c09514326b4b4ba5c549b77ad96bea : SUCCESS in 15m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5ded06d92b34b568bf9861739629525 : SUCCESS in 52m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd9b9c01ef634629b9f832e72f705beb : SUCCESS in 1h 25m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a204e703bd57d13878d6e8a427dddb98ffe06ded","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-08-01 14:47:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d4aa83f4f8cc2e1e6ccb04e952ad0e35d3af6a9b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-01 18:12:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c487a2db17ba53246fcd450997f16464fbcf8a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-01 19:55:43.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a5842ec8c168669f6cd2daa5ea6b967152b20438","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-01 19:55:50.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"8ff0da2d1d014d698ecf97f3bf2536344a69d6d8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-01 20:12:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2da54c5624cf05dd20d334a533548522af42dbb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 20:52:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8af378d39894d1db910eab847e9772a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ad8c9ec97048a2b3fe5433a2d4f85f : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/588183e0c7fd479c9b20df11da869b89 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/309c5913c65b4effbdb576741709969d : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd1c65f5d02148dab6f38be673557ef6 : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c1d9050d06aa4087b9c70a65da950e65 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de8fd61997a34ba1926624878f00f136 : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/02e333b5784a4deea77108a3ca5bdd51 : SUCCESS in 3m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/05a5de41ba4e48f9a0332cbc85640a59 : SUCCESS in 9m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bb2d551d328c4f2c8ae54c52b4f9e09d : SUCCESS in 11m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c6af0e866382408aa006aba2c8f62117 : SUCCESS in 8m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d138999e035c4d0f81dffe81cda3a37f : SUCCESS in 16m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e8449a448fe84dfa855bc96331b57c4b : SUCCESS in 32m 57s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/32c1a1c8b8594224b6a4bd457a9538f1 : SUCCESS in 4m 45s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/38deba30f84e40c79f458a6ba7c35adb : FAILURE in 5m 20s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/f0ba5016baaa42da97e6701a2fbdcf59 : SUCCESS in 4m 16s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/5c5d59fb4a224158b06c71db4e7a207f : FAILURE in 22m 44s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/97ff9c577d3745f983b250c38f7d6a7b : SUCCESS in 4m 51s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/966f51f9403442e99ebe0c807a7fb39e : SUCCESS in 5m 34s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/091f469019aa457898458b394a0c90cc : SUCCESS in 6m 54s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/c21e4c11ad4241f59c4d2358e12c5456 : FAILURE in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b1b8be420b204b8f8ffededf727c48ed : SUCCESS in 6m 27s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/7c34124de86345d8b7bbb491399215d7 : SUCCESS in 8m 41s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/c69211db315f4692838f86a2cb8fbb80 : SUCCESS in 10m 43s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/4b8b5d0fb1914b398a0500af9c5a6577 : SUCCESS in 3m 54s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/ec7782b8a199403db153895c02990cfd : SUCCESS in 10m 51s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/2002a3da2b514a7293533094686bf488 : SUCCESS in 4m 38s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/a1187020f906439ab4e4b5eea886f334 : SUCCESS in 4m 13s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/78b2b286da2041f3a7c5d2081bc85b9b : SUCCESS in 10m 56s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/a087b2eabc064e3f99870121385bf257 : SUCCESS in 18m 11s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9c3181a674a24fc8af52aa5e296d5c8a : SUCCESS in 7m 22s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/5b0c8e67d02148408fb8b47080b3d53f : SUCCESS in 14m 56s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/d77e05c271f94e69be08236dec555e81 : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/80362d54f30443c287ee5c11d4b99fb2 : SUCCESS in 5m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f3bc567a76984847b5ad068a0a413d09 : SUCCESS in 23m 03s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/3d645499f8814711a179ffe8d8f178b7 : SUCCESS in 5m 39s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/55815ba2a0a743eb9d38b04e99abbbd9 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/56aeb9c246834774914593ad83170b33 : SUCCESS in 6m 42s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/747de58eb2b444b899020b49814db788 : SUCCESS in 11m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eac0975979ad426caa976e7ccfa4a03e : SUCCESS in 43m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96033d07a85b43ccb6962c30435911f3 : FAILURE in 55m 26s","accounts_in_message":[],"_revision_number":2},{"id":"a71dafd5a04177b99cfb8311cb3eeefeaf4f6759","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-02 17:07:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"399b63c6ba7cc950173eedfd565b60ae65e74f0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 18:12:34.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/925532 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"a7a339d70f02b7bafdae3be759843b13bcc550e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-02 18:14:40.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"40ffcd7c4b29cd5ad13f18b22976b4d5adb73247","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 19:30:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/515a5d5bcf8d409f913fd4a918e0a4dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2eac8c321ee427f94352fffc1adbc90 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a343b2b3d6694b48897885aa2364ae24 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b300fcddd91d4b788c7277e8d16a2d61 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7d28eaa66688453c8907c5c513162764 : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc88f3e2762f485e86d5d7e017880e55 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8d4e818efa514a48af66d7ff51c97950 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/389053ea26344ab697c7ea847ceea14c : SUCCESS in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/01787a22de374609a8e9220c1c6af637 : SUCCESS in 8m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1032a1707aba45b79b19bfcb91ba71d4 : SUCCESS in 15m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4f15c9c2e65c4758bdf6c80d564b1ad4 : SUCCESS in 9m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/bc0b837a450a4ee3b5d5c41bab8d1487 : SUCCESS in 13m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bc5d15d2acb043259c0d357ffc884d49 : SUCCESS in 37m 09s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/1e488e55eaac4f61bc1f293279d8fe5e : SUCCESS in 4m 42s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/86af8377b3a84027ba2548c4f2f4479f : FAILURE in 7m 53s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/65201fb806b64ffd9cc149b49f13b701 : SUCCESS in 4m 19s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/c1b97713a0ec431ba20f3c3e28a61668 : SUCCESS in 9m 21s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/40dd5ddf712149938ffd3a1977fc374d : SUCCESS in 5m 49s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/c40b94c4ebdb49329af0897b4556f70e : SUCCESS in 5m 05s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/58790827e9b8443d9b98932f165cf0cc : SUCCESS in 6m 33s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/c167c1cb7d534a29810ac5accdda91c0 : FAILURE in 11m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b0ae56ccaf794819b1a11f7660de3114 : SUCCESS in 7m 57s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/a56762b6f0fa472cadc637e895122a04 : SUCCESS in 7m 07s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/cacf7d77714242b895e120541a81fc45 : SUCCESS in 10m 44s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/3fa46ad0476f462f8867ae8a63d80abf : SUCCESS in 3m 41s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b4839540525842cdb5db602b7515b011 : SUCCESS in 9m 28s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/7a55256013f94c539effa14a79b2d32b : SUCCESS in 4m 12s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/663da7fefde94cf1bb08a30581f61222 : SUCCESS in 4m 34s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/e3c28f082e794265a341671bc6c79a73 : SUCCESS in 11m 25s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/8df9367ee0914014bff364ff39b1efc2 : SUCCESS in 26m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/42dc3dd6f75847c7864146fc86ef99e8 : SUCCESS in 5m 34s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/1c94ca672006437ab3e75cd983650a58 : SUCCESS in 10m 14s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/1c5e6db989044aa69ca6f5500bf898c5 : SUCCESS in 4m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0a67f399636848c5956340bb5ac64909 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/45988d4228364a43b5d3f769c4d6bf58 : SUCCESS in 26m 34s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/5ad4cdf6869b46089876dc5e04110781 : SUCCESS in 4m 49s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/e1248244e7c740b5a58c914a490f96c8 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/66fef54ba3a84baebffa4d3b5fe7e4d4 : SUCCESS in 6m 14s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/5f8b8685a5234689936c8eb7424d7987 : SUCCESS in 16m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dceb6b54c0884420a0f31dc8c2b9e831 : SUCCESS in 1h 01m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89ee17f10cef43508863156852c6e542 : SUCCESS in 1h 10m 17s","accounts_in_message":[],"_revision_number":3},{"id":"d6ff7ab090ceea2f009a80a689c5001f2de04b33","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-02 23:39:41.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2e983ef2d4553e7b061ff29647f91e28886b592d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-22 02:35:47.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"37b3bbe5ec06743bedcdf22d2fd411f02f5f75a6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-22 02:36:06.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3901526e5ff175409c2e0748949a1d19bcedd097","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 04:07:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dcd7f74bbd1542439ab57d00ca45b36f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfed5e78b9fd4e32a6b7b4b24749c82f : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eac7dd3e4d1b480cba076de08a18cfc4 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73f1191d05424cbdb7707274c365a977 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0bec1301e7124e5aafd4738609baa997 : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6fe4da49a7b4400f855457ceb97600bc : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f14cc2939a334a7191e1d89df1ef4145 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7cc946bf4f394a089256756b37fe6dda : SUCCESS in 3m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b4934f977933483d8c3f71a9fd707279 : SUCCESS in 9m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5291ce2ea546494d91f0d0cee2609c66 : SUCCESS in 16m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0da00c8a33ab42afbeffd9562c6267d5 : SUCCESS in 8m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/55340d08c6034686b1cf5ea88d479b98 : SUCCESS in 17m 13s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/da64046ace434dd1b4868b5b498d4598 : SUCCESS in 22m 41s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/6334d36909904d2b8f7c3ebcda9ca650 : SUCCESS in 4m 42s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/2d7379037c1f4aedb348ab5d7ea7a31d : SUCCESS in 7m 04s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/60d7b83df8e24d4ca1d9e35866278246 : SUCCESS in 4m 42s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/2aaa19dbb1424339b29921ad4007dff7 : SUCCESS in 11m 14s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/baeb2602abde44ee991bf6dd8971104f : SUCCESS in 5m 43s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/8587fabb6d7345e9a725cce9456d8deb : SUCCESS in 6m 19s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/95384d21c3af4fb3842c0f2b39f697c5 : SUCCESS in 7m 15s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/c23e02ae252e4a82838652668307daff : FAILURE in 12m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1790095509274ebfba0f11be4f5cc3c6 : SUCCESS in 7m 09s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/3b5bc1b15d3e454cb966942e9d2d73d8 : SUCCESS in 8m 17s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/baa414bbf59943f090d45c30b61a3b3d : SUCCESS in 11m 04s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/24d017ebf094463e8ff140ab2e4e8fed : SUCCESS in 3m 45s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/834f58dbeb3c41e1b2c7e358cac2987a : SUCCESS in 9m 18s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/4e378a1438714dcf93da2fcb639f9caa : SUCCESS in 4m 36s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/fde0e6e23b624a08bb6e7370780f395d : SUCCESS in 4m 23s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/7dcd7b8f19964caf8d85f89d38039b03 : SUCCESS in 11m 48s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/dcf8fea590f74a6f874f6b739737c928 : SUCCESS in 19m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/878029b693344f96be1c0529f79fe2a2 : SUCCESS in 6m 36s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/da98b35875694b1f8537fa698a8c4967 : SUCCESS in 13m 54s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/1a19831f80d54ad7bdea498ebbae48f0 : SUCCESS in 4m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2bcb11febf0c4b75995c61f4f1bfbcb8 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/10da2ca532b94a7da63ca57d8331a87e : SUCCESS in 22m 21s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/29eb355655194e0ba18b438a60cfb835 : SUCCESS in 4m 36s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/d41acd69c2ff4599a2e4e55eabeb8a5b : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2a395ba990284bcb85b19c1fe46a7ccb : SUCCESS in 5m 51s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/3c9a04e8ee924fa3a71d44da2f98f31b : SUCCESS in 15m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2699d2027b941569a11d974ced22ce3 : SUCCESS in 1h 02m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2005b6ac781403bb9357549b70356af : FAILURE in 1h 17m 38s","accounts_in_message":[],"_revision_number":4},{"id":"11a24daf3c1b8634547e16fa023742ae6f75d412","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-08-27 21:02:48.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e61da455375eedbca56f9eb5ba902ddf8c9e8985","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 22:59:20.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d21f552b78d04f08971ca46848c7531c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37463b4208b94ebdb11c3ab58126d863 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64c91f5f5d394ce180afba9c5d4af923 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/411ba937cab44a65baf3ab41d0e4eb33 : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/993915a1c3ea453dba33883aae457988 : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b9a6475d44a54fd4a1e12475f05d7b35 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a09a819034e84fc2b95a1c9af9511edf : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5d44172e76574d3a93fc9e5706a4325f : SUCCESS in 4m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bf497dd7eb6c45958b6aadaa86136a05 : SUCCESS in 9m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b0d70744787d4e73aeb17066377c6c64 : SUCCESS in 17m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/fac2a4a397544a2eb23c5e3dd427bdd9 : SUCCESS in 16m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6103b00fd29649df8ab06a830ee29e13 : SUCCESS in 17m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/788022f0da014a46bb0a308ec52cc384 : SUCCESS in 24m 26s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/8659180ede504456add8a6b6cf10af58 : SUCCESS in 6m 19s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/ef7a64c2d05c45ad866babc74e4b8d78 : SUCCESS in 9m 46s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/5234ed89a7d845f4abc8528752b8c37b : SUCCESS in 3m 59s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/90e0a938e09f4babb93dd9185e30c7b6 : SUCCESS in 9m 56s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/a150d90dc4514586afb553a25e1deb9a : FAILURE in 7m 15s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/4e54557f513a4bab83aaf6af716aca52 : SUCCESS in 5m 01s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/e0c63e46de5c445c8419b5e11754b52e : SUCCESS in 6m 45s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/dad4aacd73ae4eee8dc58fced1730732 : FAILURE in 17m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eeedab71e1124054814fe0dcaf2b4f67 : SUCCESS in 9m 54s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/e6f84db130c745e1afccccbfc5bbe083 : SUCCESS in 8m 35s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/78bbc071b4e3448a8020737e33d0d2ab : SUCCESS in 10m 58s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/9f631cfc552c45128b10937b010b21a7 : SUCCESS in 6m 38s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/7e6dd1effbb1448abe6ab482aad1f312 : SUCCESS in 12m 54s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/8493bf5ff2c14937902a5534d2bf4451 : SUCCESS in 5m 00s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/8e8e7b67863042229c397181f5f910a7 : SUCCESS in 5m 01s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/0abe3d48a9374141b7826bce6e7fe452 : SUCCESS in 10m 20s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/ceaa36307dc34b8db52bf27e9346d14a : SUCCESS in 26m 59s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a185ef685972483383904272ec82c84d : SUCCESS in 5m 37s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/8f74469ac6cc4685a6dbc89bd2b4fb1c : SUCCESS in 13m 08s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/917295ef9d874266bc8dd5a891c2d712 : SUCCESS in 3m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f0945d83808b42bb8217487adcd9f26d : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0fa41ef8a88349d2adc7d5c6b83d3cf0 : SUCCESS in 27m 11s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/44162061f8f544aeadf30f4709b7ed2f : SUCCESS in 4m 50s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/56ac0426dd954849af438c48b4a3e5fb : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f95c36aff9b7479eb5748b4c060f79e7 : SUCCESS in 5m 18s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/9d7edea2dd3c48569006ad995d80a3c9 : SUCCESS in 15m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71c481a11a9a4de2b1d6a8d1deac2f68 : SUCCESS in 1h 11m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1aefd318caf54ee2ba2ca46446af2a22 : SUCCESS in 1h 31m 04s","accounts_in_message":[],"_revision_number":4},{"id":"3d87313b06186cefd2b3de069388d257faf58f29","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-29 17:38:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"71ebd6235c1a48f04ea5c5372c102dbd7ace4722","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-30 07:52:16.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e0eba9d3f01429d5bd85211531bbc1583aeb6f17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 09:22:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0850b6f11e134fda9abb5e72e7d27d6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eea3fd93b56446b6aa75f7881a188701 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73968f64ee8a4904b2a8b1e5d0d34345 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65cb17b24c3d49128aed66d4dc02afdd : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65b03ce7afb84b0bbee52f63017833ef : SUCCESS in 3m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d21a4656aa9945c99d0f9d89675a0064 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3d2c8018f8ee423c893c1e9d6578abf9 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e5896bd944f74054a8ffb4b7851c0f34 : SUCCESS in 3m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/68525cd35ac14879a6bfdce84d11eefb : SUCCESS in 7m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dec59463a1b04de79c4b8f24d757229b : SUCCESS in 15m 47s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3a52775741434f9c9651dadeda0fb616 : SUCCESS in 8m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4c3ef48897014448ba390535427ec799 : SUCCESS in 18m 13s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/8aa886c678624b478467d3016b1973f1 : SUCCESS in 22m 04s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/be855ce2da8848e4afc93406e358592a : SUCCESS in 4m 43s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/5a5f0399007a497a88d1750c821a2d20 : SUCCESS in 4m 34s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/fdcd89d5e59d43faaa32db79fd097975 : SUCCESS in 4m 31s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/388d9f87317f41419f79b28be64cbc99 : SUCCESS in 9m 26s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e5875eacdffd4edda571498d1308d7a0 : SUCCESS in 4m 16s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/eabe16f582534a92a973f9de154f2f7c : SUCCESS in 5m 29s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/037d9135b0984151b5c2820a1af4dd13 : SUCCESS in 6m 59s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/57d228d49cbc427baa659cdc402a7eb2 : SUCCESS in 13m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/375aab8061284f0da39cd399ab40a5ec : SUCCESS in 7m 16s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/bcc69afb6360455db209d1c4e5b92845 : SUCCESS in 8m 19s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/d7495c785d5a431ca3532dea81d68d23 : SUCCESS in 11m 07s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/cce27846aad1491889e5fc3e8fb7b5c5 : SUCCESS in 3m 40s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b50be9008ff04e37be920017b44e74b5 : SUCCESS in 7m 41s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/ee19f7b5ab0747898e99b2410db5917b : SUCCESS in 4m 07s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/045044ef834844c5bd3b62a2ef9f7669 : SUCCESS in 4m 04s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/365a032a7ea944348103e8ae7c7db50b : SUCCESS in 9m 55s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/491ac4f06c684876b28f89b2d547e16d : SUCCESS in 18m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/766b81da3fef4839be654433007b883a : SUCCESS in 6m 34s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/baa96316431d408cb780c9c8f7996a9e : SUCCESS in 9m 52s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/47312a91bddf4e899e7c8ad992b6d6b4 : SUCCESS in 4m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f072e972616341bbbd9fb88b2c7f8d71 : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d610466fdd7943b4a97875a5dde283c1 : SUCCESS in 20m 59s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/d787b11310774898b3c387278d31f349 : SUCCESS in 4m 01s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/538c6fa0483747ebadbdd901332f9b3c : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/687d9c5448574410a6cf50e3bbff9be8 : SUCCESS in 10m 06s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/580a59954a284e99930e1733e53d21d3 : SUCCESS in 15m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd3571c2a31a4908bb0d1b560294fa78 : SUCCESS in 56m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3b48f68c2094f7d956eb132d7ece300 : SUCCESS in 1h 29m 36s","accounts_in_message":[],"_revision_number":5},{"id":"784955ca5aeb6e6e95eb8751e2e14fadddc6c535","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-30 13:38:45.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f1aa86a534e32d38c5985c78a6a893217261a108","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 13:38:59.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd84edc8174f4f40920e52b0e2fbcaf6\n\n\nWarning:\n  Change 927571 in project openstack/horizon does not share a change queue with 925464 in project openstack/requirements","accounts_in_message":[],"_revision_number":5},{"id":"3ed4200a36b92c6a8397a27e7e578091046f1fa0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-30 13:41:00.000000000","message":"Patch Set 5: -Code-Review -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7339305356c91476845bdc2adf6b1df5fb7d9a48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 15:14:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c66f03f8e60d4edebb3ba202a8d3b23c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0fa1ea6da2148ba9ab03bf4661a6e50 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/666c99f4ee0e43fbb3d20bfc2e8767f2 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/defb88aac5bb42fab4c2e035182316a7 : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e01ed367819e47f4b4c53102c67da0a2 : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/236b25bf58b74810b7bc7cdeb394ecd3 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffe6dd24dd3b4303ab36a5f7abf8b646 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/96721dff0ed34b24a77ea007f2c3c472 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/740ab1cc58944588a6ac21bf9a1f9b52 : SUCCESS in 8m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3f26e3fb30c440d0aefd38c594733789 : SUCCESS in 15m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5d6e349a075b40b4b3707f33fb728509 : SUCCESS in 10m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7871108081bb4b3cb8800d54e75fda04 : SUCCESS in 17m 20s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/de6ded3bc4fc44f9976fa090ad74e8f3 : SUCCESS in 25m 03s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/80b91627fc254217b6f4840e90c4a1b3 : SUCCESS in 4m 59s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/74e33888b2e24c13b3c1202c20921a9b : SUCCESS in 4m 51s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/7411103c9c3f4765b5a0c0d9b61accfb : SUCCESS in 3m 55s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/43225df31b0f45968ba0073dc27e13a7 : SUCCESS in 12m 01s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e310020ba72f4b9396b9c4bdcc0ef4eb : SUCCESS in 4m 48s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/5611cda7142f4fc28771afc90ddc4f6b : SUCCESS in 5m 59s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/62e2d1c6fef84103bc73abfa7c97427e : SUCCESS in 7m 41s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/5e20b0549c11403ea7c64d424fd6570d : SUCCESS in 13m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c360188cafeb4238acd5e426ba709d26 : SUCCESS in 9m 24s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/8a42629a4f8a4d8e9eb4f9ac93bc8d47 : SUCCESS in 7m 26s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/e3380a4a195b4db09e8d202013eba2cb : SUCCESS in 14m 04s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/f4837441bffc4e81b20aad5ad578c774 : SUCCESS in 4m 19s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b740e5a7767342a7a64ad8ce5a50b78b : SUCCESS in 9m 34s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/045a41ecb47b4e9f869f27e224691ce0 : SUCCESS in 4m 51s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/c24291cf1753484aab7e8e9f6c3cda7f : SUCCESS in 4m 28s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/8c2508887e0244cab158c23d84a568d6 : SUCCESS in 12m 41s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/6529ce7585cf444089375424cb317e67 : SUCCESS in 22m 02s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/885ffcb19c674e1faaae4cbfb9ba26af : SUCCESS in 5m 51s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/d4af08bbf18248eea20d9631d507ced1 : SUCCESS in 11m 58s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/902b15fb38fc4c1ea6e65b734c01d887 : SUCCESS in 4m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d4f9cb416e8c42cbb43aa40b419a19c5 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/260cd86b0e55451eb4697896eea88e06 : SUCCESS in 22m 20s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/1de34bb5e8ec4c918443491868c65fd3 : SUCCESS in 4m 15s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c185775df0c643e2ae85d69743c80d1e : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/763b9bcb9d134f1a8b719ee6909b8080 : SUCCESS in 5m 42s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/c19f1aafaddd4786b3de215086278f08 : SUCCESS in 15m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/663450b0853c4a3e821b868b27efbdbc : SUCCESS in 57m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5373faa0e25e4d158ef86f1a9d70a46b : SUCCESS in 1h 29m 43s","accounts_in_message":[],"_revision_number":5},{"id":"969d1687f8eb89e60c3fb6c5e34533a289bb138a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-30 17:09:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"376adf2b05d357d709870a285da0f5e8783ba75e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-30 17:09:56.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"b6f1a6a7e5b8fe43dbe84d90680df643a888ddb2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-30 17:43:48.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"957bbad81f44bda760b8925eb569eea0ebc91fa5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-30 17:48:35.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"08eedebcd1cb4f4de7798cea5d0c4869701d843f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-30 17:49:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c2805f0f71d614fea9bb5292061153e0b4a478ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 17:50:44.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/927571 is needed.","accounts_in_message":[],"_revision_number":7},{"id":"1f7197fc26f6a03689da1ccdd9989600364fc8ed","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-30 18:18:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3f323976e8e5b364bf3960fa6f91f71ff311e26e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 19:48:33.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eccf53a7deb140f0832390d9b4d8bd10\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af708dadfebb4e489bf425287719f797 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5bf8521348647ffb7452a667abdc320 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64a9757ec8944e539d72f6254a38a6c4 : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e62a41d8a44463fb9de8436e8b8ebce : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ac5f89c8c67429fa14e2f10688a7c69 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b4248d1c00747989f858ce8db2c290d : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/722346b6536d4f72b0166af41d28aa53 : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/348e7761471443a78b3f6ca59414f0d1 : SUCCESS in 8m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c13e622e54b14e098b36676aacc3d78f : SUCCESS in 14m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/99bea90d7900437fb1733c819bab7dcd : SUCCESS in 8m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b075326541f84e338fd0310567932687 : SUCCESS in 18m 58s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/25b8c8ff3d694c98a1dfb689aaead387 : SUCCESS in 26m 31s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/17e554e801614967a90a3775d523446c : SUCCESS in 7m 15s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/0098cc1fea0142c09fcd4ebff60bc5d9 : SUCCESS in 5m 17s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/c22bc92f41284afd9d34c20f99df04f4 : SUCCESS in 4m 04s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/5aed93353e914ee6b83dea1718462461 : SUCCESS in 11m 27s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/43989b859505406d848f67d8836f0954 : SUCCESS in 4m 45s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/dfbb185bcfc74eb2beea241ec09fc055 : SUCCESS in 5m 25s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/81545cd8c51940a0be3c5c9ab376a95d : SUCCESS in 6m 37s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/40c80f41f11547a89879bad1ab70e9ef : SUCCESS in 13m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a971afa4128648e2929bc436e07c9a08 : SUCCESS in 6m 14s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/0ce6de864be4435ba76aa183f79dcb30 : SUCCESS in 8m 09s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/1477c3bdc8614d2d83558a764c40af79 : SUCCESS in 10m 21s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/06b0e3b94f594bd09349415f1b3d4db2 : SUCCESS in 3m 48s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/3f316e1427344bedb346618445b3f37f : SUCCESS in 8m 24s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/b8bbbc3e214146d5bebdead6dc4007f5 : SUCCESS in 4m 29s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/bf2fffcae5d147c8b5168dfb1f746bf4 : SUCCESS in 4m 38s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/561abd1a353141b6a89133b68bfcc05f : SUCCESS in 10m 59s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/b77026a121f64a928a572ee05a4c6e5e : SUCCESS in 19m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/9fce348ffb624057b9a345d99eb9b2e0 : SUCCESS in 8m 47s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/45bd1a6f4271450287979447d5c1b3b9 : SUCCESS in 12m 40s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/6ff6ed8c37124335b68bc3554cb3104f : SUCCESS in 4m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/83733500c1774dabb7e8ceb3785716ac : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1865c1f334334620890e43a32555f65f : SUCCESS in 24m 34s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/3c8cf9095f2c4ef0b6b418d5fb2170b0 : SUCCESS in 4m 29s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/084f98a4612d4e3481694c6b120bb7a2 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7e28bfed705a4fdfa33479d7302036b8 : SUCCESS in 6m 40s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/7db461c8423942d0ba12692957c387a8 : SUCCESS in 16m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/750cab5bb0324c0abaef078805a62182 : SUCCESS in 59m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f657ae0733eb4a3e86a7f1d8e565f268 : FAILURE in 1h 13m 30s","accounts_in_message":[],"_revision_number":7},{"id":"b9dc549f6bd6735d84e4ad26cf74a3bfc102dfa9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-30 20:08:14.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e22170be28214a2bd3ca576c121f314a8e3ed168","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 21:51:57.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3deffec6db574a30b0ff60f6d44f25e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/042bb5652c4a449eb4acce319e3134a8 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/442905f977a7480bbb0a061ced8a0658 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2852027c07ac42249edcedd27edef11b : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/86b7a6c11e5d4c11ad434691ce3b5cf2 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/085b45f244074cd6ba6478ae51f884fd : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/252c92cebef14905b78ca2b21090356e : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/23e0ce2b082542b48fdeb152ab90df85 : SUCCESS in 2m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9f4620337c224682a9b1f30efaa9a8ca : SUCCESS in 8m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0dcf7761fe714afb9b6d710f48c016c6 : SUCCESS in 14m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/da2afc288a1749e5ac0ee30ee9ef985b : SUCCESS in 9m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7cb299f0fa3c4b56b615d896d60bfb7f : SUCCESS in 14m 39s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e41d35a4826d44b9822672d1cc1447ad : SUCCESS in 25m 27s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/25dcbea3eb3e4a7fbfe2050a8d16636f : SUCCESS in 4m 28s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/4aebd814d4f743b0aabf63b3a6e167cb : SUCCESS in 4m 22s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/2e70bed11c4d4061a97f11a3fc2f9191 : SUCCESS in 5m 23s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/b20152991f4b43a2a05983ac324546a9 : SUCCESS in 10m 37s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/68a24e01916b4a42b785fbf6b589194b : SUCCESS in 5m 51s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/0d98554952ba4e71a3dfbf8c0d045bb2 : SUCCESS in 5m 12s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/ae25fd8ba79a44a29dd2cd2adcbf37b5 : SUCCESS in 7m 00s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/f118669655954a1083ec5e0c4b9b9e07 : SUCCESS in 12m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/14e66931a5df45e89b2eeceab529b84e : SUCCESS in 6m 59s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/51a9ce49893d451591accd7540b6493b : SUCCESS in 6m 42s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/9f49947c877b4c1587a5c3f7e75a6681 : SUCCESS in 10m 17s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/1bada0e79d9b42188c57d034c0581a18 : SUCCESS in 3m 37s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/1b9d6e6adfb9481dadcbde5e87bac860 : SUCCESS in 8m 40s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/a03a179262ad4cd2b171d13544dd5db8 : SUCCESS in 3m 41s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/443a7887e6234ed9abfb0ef19695dc41 : SUCCESS in 3m 43s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/299da6ed008d4e9c8a8d9fac3c7d19bd : SUCCESS in 9m 05s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/df3570981dfe43e59978334959ab54cc : SUCCESS in 14m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2b99b3e9efc647029213943599cede33 : SUCCESS in 4m 43s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/faad411b524a49e38b1a654c2b6a4f7f : SUCCESS in 8m 34s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/4dc68e1e0c484173988d182797b03300 : SUCCESS in 3m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cfa5caae3d11425ebbad28951375a458 : SUCCESS in 2m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/302aaa9bf709400d84b09148d6b947ba : SUCCESS in 17m 12s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/d65814dd069b4f4bab67371e8ea49e7d : SUCCESS in 4m 06s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/88403fcc84274b73a4eb3e3c71fc8cd9 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0e42e5e386c748959a8e34fc81be0744 : SUCCESS in 4m 53s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/4af712761bb848f4b5b275397bbb6d55 : SUCCESS in 10m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c404bb8322a45f6862b2744d742dabe : SUCCESS in 56m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c548b239af434ba39ce8685110086539 : SUCCESS in 1h 42m 55s","accounts_in_message":[],"_revision_number":7},{"id":"699ebb26e1d00ae1912e8ef08e104f6bdd277ad9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-09-03 17:14:38.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"5984f71f6b1d0d2db652642d524312614e923792","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 17:15:24.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"21c092ed3dcc52ae7b133ea4561716f7bd9c8b83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 18:55:58.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4afe3ac2c2964797ab3075fe03203813\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d94489131884110b0b624b95b729d06 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48af0b75f53b4e3bb3254f55a5bec2b8 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb876bd7a28b4baeafe44792858032a7 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/719053192f884742b65656bd3d033687 : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0e02aec4a903438fa2438266e6013fa4 : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/35e5566676814b769f327652973a1f9c : SUCCESS in 3m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/05a2e1da90094ff0b33c3e6086e3d882 : SUCCESS in 10m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/09befb7f23124c1198e3c595862ccc37 : SUCCESS in 14m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/08eff82b5f4a4bb285200f8429c7dc1c : SUCCESS in 9m 05s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a3dd07a280e84efca73dab1796dae62a : SUCCESS in 16m 03s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/843f03fb6fb8400dba9e1bc5c84299e5 : SUCCESS in 10m 01s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/634d135d630146d3b3afe5204485b762 : SUCCESS in 4m 47s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/deff8fca2030434298328d50bb5e9c76 : SUCCESS in 5m 19s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/cea2555a66c1498b99099effdb1d3b4b : SUCCESS in 7m 23s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/71d7cb89c2204337b95ba58324aa6a75 : SUCCESS in 11m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9b55f4fa3b7d438b95f2f63dc9809dad : SUCCESS in 6m 39s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/35fb84525bf44ed4a9d9df8a6f17696a : SUCCESS in 8m 45s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/2135adc55a6c4366b632b510d423709d : SUCCESS in 10m 36s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/296b46f1d2e74b548532697dce4f5676 : SUCCESS in 3m 52s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/2b45f7ba0c3744a4a9dc0c4c0e58c222 : SUCCESS in 8m 14s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/cb41987d00504b7f9324b69c30d7ec7c : SUCCESS in 20m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b3f97111d7124b0280afb9804ad20940 : SUCCESS in 6m 16s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/ae998a6ea6604cba8e488eb86b529c5b : SUCCESS in 11m 45s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/12fe88c883e04a568f44cf7811db01ec : SUCCESS in 4m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ff3d16ee2f4f4695bae85b27c0c89f93 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8dd18d6e4f8b4b2ebf7ddac155ebd855 : SUCCESS in 23m 39s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/ec6cfdbc2c314db983e3f693630c474c : SUCCESS in 4m 25s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/08bd6d6f7a924e7696e65a67637e2561 : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6c27dc788c6241dfaab7eff4670c39ec : SUCCESS in 5m 49s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/b13c98cc96724bbbac534ec952930e0b : SUCCESS in 17m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7c922c40efc48259f314d087a2b97c9 : SUCCESS in 1h 02m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c63913a85174026825244aaf0195a5e : SUCCESS in 1h 34m 44s","accounts_in_message":[],"_revision_number":7},{"id":"385455e7ea1611c9e6904325ce3fdb12eb8cec43","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 18:55:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"6a468843bcea1a374decf3e6740d5bdd7d4f19d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 18:57:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f15ef0012604e099ccd670cbdbee544\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26a184bb66fc40a6ad9c114156dc1ac5 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"af724731878229d1071b371913833909f70ea91a","revisions":{"097b9cd12349ab88048a1b4d8fdc2588da658e32":{"kind":"REWORK","_number":1,"created":"2024-08-01 11:11:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/925464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/925464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/925464/1"}}},"commit":{"parents":[{"commit":"2c279bd469df43206693b27fe85565028346f042","subject":"update constraint for etcd3gw to new release 2.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c279bd469df43206693b27fe85565028346f042"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-01 11:11:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-01 11:11:22.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 4.4.0","message":"update constraint for oslo.policy to new release 4.4.0\n\nmeta: version: 4.4.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Commit: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Change-Id: I94c56727ff21fd0e4f2c2bae5b495fd386205b13\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I2632e5e1d277f520be14add37b8a201ad2f5daf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/097b9cd12349ab88048a1b4d8fdc2588da658e32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/097b9cd12349ab88048a1b4d8fdc2588da658e32"}]},"branch":"refs/heads/master"},"14f7bb0aa11acd71f44d53d4eb4e009d0dd1f94d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-08-01 19:55:43.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/925464/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/925464/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/925464/2"}}},"commit":{"parents":[{"commit":"2c279bd469df43206693b27fe85565028346f042","subject":"update constraint for etcd3gw to new release 2.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c279bd469df43206693b27fe85565028346f042"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-01 11:11:22.000000000","tz":0},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-08-01 19:55:43.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 4.4.0","message":"update constraint for oslo.policy to new release 4.4.0\n\nmeta: version: 4.4.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Commit: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Change-Id: I94c56727ff21fd0e4f2c2bae5b495fd386205b13\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/925526\nDepends-On: https://review.opendev.org/c/openstack/aodh/+/925529\nDepends-On: https://review.opendev.org/c/openstack/designate/+/925531\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/925532\nChange-Id: I2632e5e1d277f520be14add37b8a201ad2f5daf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/14f7bb0aa11acd71f44d53d4eb4e009d0dd1f94d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/14f7bb0aa11acd71f44d53d4eb4e009d0dd1f94d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b087d8a512dba994ef32de8c15a219bd920927bf":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-08-02 18:14:40.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/64/925464/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/925464/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/925464/3"}}},"commit":{"parents":[{"commit":"2c279bd469df43206693b27fe85565028346f042","subject":"update constraint for etcd3gw to new release 2.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c279bd469df43206693b27fe85565028346f042"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-01 11:11:22.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-02 18:14:40.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 4.4.0","message":"update constraint for oslo.policy to new release 4.4.0\n\nmeta: version: 4.4.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Commit: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Change-Id: I94c56727ff21fd0e4f2c2bae5b495fd386205b13\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/925526\nDepends-On: https://review.opendev.org/c/openstack/aodh/+/925529\nDepends-On: https://review.opendev.org/c/openstack/designate/+/925623\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/925532\nChange-Id: I2632e5e1d277f520be14add37b8a201ad2f5daf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b087d8a512dba994ef32de8c15a219bd920927bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b087d8a512dba994ef32de8c15a219bd920927bf"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1ff42616d6b7f841e938d29a4e5501629d3848a7":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-08-22 02:35:47.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/925464/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/925464/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/925464/4"}}},"commit":{"parents":[{"commit":"2c279bd469df43206693b27fe85565028346f042","subject":"update constraint for etcd3gw to new release 2.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c279bd469df43206693b27fe85565028346f042"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-01 11:11:22.000000000","tz":0},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-08-22 02:35:47.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 4.4.0","message":"update constraint for oslo.policy to new release 4.4.0\n\nmeta: version: 4.4.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Commit: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Change-Id: I94c56727ff21fd0e4f2c2bae5b495fd386205b13\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/925526\nDepends-On: https://review.opendev.org/c/openstack/aodh/+/925529\nDepends-On: https://review.opendev.org/c/openstack/designate/+/925623\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/925532\nDepends-On: https://review.opendev.org/c/openstack/barbican/+/926691\nChange-Id: I2632e5e1d277f520be14add37b8a201ad2f5daf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ff42616d6b7f841e938d29a4e5501629d3848a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ff42616d6b7f841e938d29a4e5501629d3848a7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3ffa47cf2fa5dd51733b62ed4daca655606f6818":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-08-30 07:52:16.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/925464/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/925464/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/925464/5"}}},"commit":{"parents":[{"commit":"2c279bd469df43206693b27fe85565028346f042","subject":"update constraint for etcd3gw to new release 2.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c279bd469df43206693b27fe85565028346f042"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-01 11:11:22.000000000","tz":0},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-08-30 07:52:16.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 4.4.0","message":"update constraint for oslo.policy to new release 4.4.0\n\nmeta: version: 4.4.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Commit: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Change-Id: I94c56727ff21fd0e4f2c2bae5b495fd386205b13\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/horizon/+/927571\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/925526\nDepends-On: https://review.opendev.org/c/openstack/aodh/+/925529\nDepends-On: https://review.opendev.org/c/openstack/designate/+/925623\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/925532\nDepends-On: https://review.opendev.org/c/openstack/barbican/+/926691\n\nChange-Id: I2632e5e1d277f520be14add37b8a201ad2f5daf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3ffa47cf2fa5dd51733b62ed4daca655606f6818"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3ffa47cf2fa5dd51733b62ed4daca655606f6818"}]},"branch":"refs/heads/master","description":"Edit commit message"},"64e3b9a6e587274e9b7140aea6760f41fd638c55":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2024-08-30 17:09:56.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/925464/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/925464/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/925464/6"}}},"commit":{"parents":[{"commit":"2c279bd469df43206693b27fe85565028346f042","subject":"update constraint for etcd3gw to new release 2.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c279bd469df43206693b27fe85565028346f042"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-01 11:11:22.000000000","tz":0},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-08-30 17:09:56.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 4.4.0","message":"update constraint for oslo.policy to new release 4.4.0\n\nmeta: version: 4.4.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Commit: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Change-Id: I94c56727ff21fd0e4f2c2bae5b495fd386205b13\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\n\nChange-Id: I2632e5e1d277f520be14add37b8a201ad2f5daf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64e3b9a6e587274e9b7140aea6760f41fd638c55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64e3b9a6e587274e9b7140aea6760f41fd638c55"}]},"branch":"refs/heads/master","description":"Edit commit message"},"af724731878229d1071b371913833909f70ea91a":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2024-08-30 17:48:35.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/64/925464/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/925464/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/925464/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/925464/7"}}},"commit":{"parents":[{"commit":"2c279bd469df43206693b27fe85565028346f042","subject":"update constraint for etcd3gw to new release 2.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c279bd469df43206693b27fe85565028346f042"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-01 11:11:22.000000000","tz":0},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-08-30 17:48:35.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 4.4.0","message":"update constraint for oslo.policy to new release 4.4.0\n\nmeta: version: 4.4.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Commit: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Change-Id: I94c56727ff21fd0e4f2c2bae5b495fd386205b13\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/horizon/+/927571\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/925526\nDepends-On: https://review.opendev.org/c/openstack/aodh/+/925529\nDepends-On: https://review.opendev.org/c/openstack/designate/+/925623\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/925532\nDepends-On: https://review.opendev.org/c/openstack/barbican/+/926691\n\nChange-Id: I2632e5e1d277f520be14add37b8a201ad2f5daf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af724731878229d1071b371913833909f70ea91a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af724731878229d1071b371913833909f70ea91a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
