)]}'
{"id":"openstack%2Frequirements~925489","triplet_id":"openstack%2Frequirements~stable%2F2023.1~Ic006649f210a249be5844158bbb133599188b426","project":"openstack/requirements","branch":"stable/2023.1","topic":"new-release","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2024-08-21 06:52:01.000000000","reason":"Change was submitted"},"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2024-08-21 06:52:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic006649f210a249be5844158bbb133599188b426","subject":"update constraint for neutron-lib to new release 3.4.3","status":"MERGED","created":"2024-08-01 14:32:25.000000000","updated":"2024-08-21 06:52:01.000000000","submitted":"2024-08-21 06:52:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925489-new-release","meta_rev_id":"78fbba0f2abf2a106276fbb644379f4d2ebbd482","_number":925489,"virtual_id_number":925489,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-21 06:52:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2024-08-20 18:12:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2024-08-06 08:29:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":1,"date":"2024-08-20 18:12:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-01 14:46:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-01 15:55:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-06 08:29:09.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-08-07 05:55:19.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2024-08-07 05:55:19.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-08-20 18:12:19.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"995fdaa1c84cc2905fdb8788598b2315b9d2a95d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-08-01 14:32:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67ab0c3ceb9cc5fd5e243a5fd5ea19de6977e1a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 14:46:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fea650229d92440f8911c62941ca1a46\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2a6cf8fca3a8445b96be60929488a0fe : FAILURE in 12m 45s","accounts_in_message":[],"_revision_number":1},{"id":"deacb805989e7dc5c5b580463a870c480d2e47fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 15:55:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf86618f63114967afd8ad28b40bad28\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddcb957cc69e4b328dad9fdc47af8616 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cbe787d6f6c42ff97bd2140ac201ba3 : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45aa4930e0c344d5b5f15bb6824dbbb1 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/60eb3a29bc2149609fa4943c83a306e2 : SUCCESS in 5m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9e213702b0dc47a9a77602b4f063d29e : FAILURE in 3m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2ba2dc7e2f8b41c8a61efc595d76185c : SUCCESS in 15m 06s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/0969a4af881b495099b4e2cf5c8445d9 : SUCCESS in 7m 37s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/0d847fab9d984e6f8351ffec5eed4fd8 : SUCCESS in 4m 52s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/852f6113d0d04c73998eb58dfc6e0643 : SUCCESS in 5m 05s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/7b2b25cf382d476ba855688f1444b075 : SUCCESS in 10m 52s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/5dd84a507da14cc5824c3920992f0a99 : SUCCESS in 4m 45s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a3d981e48a794f56b3110ad8a20bcef9 : SUCCESS in 4m 47s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/021173ce0ee94ee4b6c087832a19f2a4 : SUCCESS in 7m 39s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/76fec6ea8b734726af38bf287ba39cd2 : SUCCESS in 8m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9663c9aba0c942468c41e65d05b49bb0 : SUCCESS in 7m 07s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b4118762a11c45879f13f6739c7e748a : SUCCESS in 9m 29s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/9db034d536214040b2df1eb57cea5388 : SUCCESS in 10m 53s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/2ec762586eda458582d7c745e066c5fd : FAILURE in 2m 23s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/58d5a34c6c1641c48fed6eab84ad3028 : SUCCESS in 8m 04s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/8529a883110d458da46516cfcce05c1d : SUCCESS in 4m 12s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/0ecd59ecf53145238f2ff048bc23bf83 : SUCCESS in 9m 55s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b7f7cee048bb4d9196166a5d15198d52 : SUCCESS in 24m 48s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/1c63a39c2ec447f4b1329d00f40cac99 : SUCCESS in 14m 02s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/abe3cc9b3ea648138db2c17dcea2dbc5 : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/308fe61ab995415591a99c8e94b33a14 : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e3ddbe11fc246ea97f3284c10dd668e : SUCCESS in 25m 28s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/1bd605be5668406e9c1e658783df5c33 : SUCCESS in 5m 54s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/08e53c105d7a4cbbaf2b70cdb18823d5 : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5a7e286f1fd646ebb8e0fe423a9395d9 : SUCCESS in 6m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/130cb4063955413b8228fc70ec2c7401 : SUCCESS in 12m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6c94de3584c4159bc3d19e12705c39c : SUCCESS in 47m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e79d75ff6f94b80bef5a72a58bb2528 : SUCCESS in 1h 19m 16s","accounts_in_message":[],"_revision_number":1},{"id":"dd7a76ec2af65696d3e1493d472f08ce37aaf4f7","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-08-06 08:29:09.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab48ca73ff6127d95401a2994c90ae4f08608d16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 08:45:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25c5742b9fc048c780c0970225dc9049\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2ad16f504cf644249ccee5c1cc40e891 : FAILURE in 14m 09s","accounts_in_message":[],"_revision_number":1},{"id":"2540acd60f1b0d9ef663417e9eb15f005c99d128","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 09:41:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/222608a85e8d4a11a3867b512c15fd4a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26c9c906111c456e8542ccbc2590f544 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6373c6c2b79c4abdaaad6f7e479de90b : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c5bab231241e463382f18fa7155d924d : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c8056e47bebc45888b59db1934e74ad6 : SUCCESS in 4m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bb18a02e6a004f61b376cfcad5a68a23 : SUCCESS in 9m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a93149b469be4be3890f1ecd46f57340 : SUCCESS in 14m 18s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/37ceecb5545947b4adfc90a71bb2b01d : SUCCESS in 10m 58s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/6e326612c4d64696ab6603c29872df2f : SUCCESS in 5m 40s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/0d08b7081a194028903a7db08ac0e167 : SUCCESS in 5m 50s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ac43b6c734b64fbca9637deda03d8b85 : SUCCESS in 10m 58s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/71bb0c2668dd41df86cafd9834e7aae0 : SUCCESS in 6m 48s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a544731c245147038a76da2f69d02e55 : SUCCESS in 4m 53s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/6d6b93bcfb0d44dab3b2436644314639 : SUCCESS in 8m 13s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/1d334dbc405247d79dbb15a122fffe27 : SUCCESS in 11m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/33c8306b51b44d2a93763a5f74550897 : SUCCESS in 9m 31s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/988f6727b8c74337bd3f4a978dc7b737 : SUCCESS in 9m 28s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/5d6c7a29d4ee473cb7a0c834b5cb12bd : SUCCESS in 13m 45s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c5c37e463f074262a10d35fa43eb1790 : SUCCESS in 4m 37s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/89914de9bdf446d5bb8270dcc905d6ea : SUCCESS in 10m 48s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/be10ef3d69ec460e97885002f8f70a46 : SUCCESS in 6m 55s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/a2ac2bdeab3e47bc8ce8f2f8da02908d : SUCCESS in 10m 49s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4cfe10977cce46be902711421bc53421 : SUCCESS in 21m 54s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/4163fcb1b9194fb892b9b22c52edf2e2 : SUCCESS in 12m 33s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9ae3bdb05b7d4441ac62223693cf71d1 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7350ae329375448e9cc265d429a50867 : SUCCESS in 2m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7a2e70198f3442f5839ea2af5fa7614d : SUCCESS in 24m 46s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e05db029e0ac4ea08a01acded1f6ca5b : SUCCESS in 5m 00s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/319b6538f56e48f68e4625a4294a4027 : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8770b6d8364a4d0ea305347d3ed0c124 : SUCCESS in 5m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a91af2e793624ef9b3a65885bf5e9879 : SUCCESS in 11m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac629b5c18ff4d69ab472c824b9a6b30 : SUCCESS in 54m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83b37cb87d8e4ee5be8c7529ea33d94a : SUCCESS in 1h 08m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a21db783e95bde11d7873ce1fb7b2ebe94a8d0d4","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-08-20 18:12:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"046b2206788d30816feb1a101092b47fcd4c9422","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 18:12:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6e71a52996e3f7f861418b76a13689ca80c46dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 19:46:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3c6c46ae7d54f638a487009e35f3455\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eac4b109ed5b4d37b390fa9af18070eb : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dec9abcf48a44119ad741eb2e655bdc9 : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a59f41b52b18473d8ba666ede0937122 : SUCCESS in 3m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2fe9346aeb6e4c678611f2908988ffc4 : SUCCESS in 2m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d3313d0bf14e44089f1a88344c027ad8 : SUCCESS in 9m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/067d5a1cd696448ba83c954815f0f3ff : SUCCESS in 12m 03s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/92dfafd2213f43eb9d68b385d7b9e2f7 : SUCCESS in 10m 57s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/4673596e53cf4e39a4a18efd72f27cf5 : FAILURE in 4m 26s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/bf0c8880042a47d99b18e9f9af96b198 : SUCCESS in 4m 28s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/4287957846b041679016617bdcc26f95 : SUCCESS in 6m 38s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/3b5bd4b36e1648e79b25026002afb02b : SUCCESS in 9m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0f41c7421e204463a97b4d3719b257f4 : SUCCESS in 7m 13s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/0edbb8162d5e4c009196335dc07bdd53 : SUCCESS in 11m 13s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/bd41b128a2db43d59c3f6cc9df215f9e : SUCCESS in 13m 18s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/4abb9e25bce74035a9f5e0f605ff9421 : SUCCESS in 3m 36s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b10b9c6ac88d4fec8bc4edb34bd5b182 : SUCCESS in 9m 42s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/dfb6c4ae43e143b8a37e6ee2e974dae4 : SUCCESS in 5m 13s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/4e90673cad6c4771842f4f02448386f7 : SUCCESS in 22m 31s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/48d1966d3daa4385a2d02c68fb8b5cc4 : SUCCESS in 13m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/86af7198144447ecb0f5ede3077e06d4 : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e18236d99de84c158034003f473e2f08 : SUCCESS in 2m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a6446059cf9342d0a384d3bcf9581960 : SUCCESS in 23m 50s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/4e32de0d43bd4a28a83dbb1c83d999ef : SUCCESS in 4m 36s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/96b966f153234b61b33649f1ccbb3f69 : SUCCESS in 4m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3bf0196354f24f9a83754d1e5426985d : SUCCESS in 7m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/14a4a7d3af5b454c81b319e51f86fa36 : SUCCESS in 11m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba5aa15285c24aa79c0a9638cb0a5d1c : SUCCESS in 53m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11125b9a3dc5404ba284913978d4ccf2 : SUCCESS in 1h 27m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a3ff5bf59d55643a030c1ebd7c95b0c0e075ec09","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-08-21 04:23:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"752c9470eb62ac018099247fd02c604e0938cfb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 04:37:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4be2486fb3c647bcbd7de3e33e689f51\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3444efd5187848b78361c93df9d62f60 : FAILURE in 12m 08s","accounts_in_message":[],"_revision_number":1},{"id":"b8cf0018b0002a03c22b554898494d0c7b596abc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 05:41:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/061f21ced9f04b8a9c4327f3ce8a3dc2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63d14d9d02ed43c7af4107777a641043 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb3b51ccca4c452583bd5b24d19cd7c3 : SUCCESS in 3m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8731bb6c889347c2806dd276e924e0fc : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d829b9b7e8774ccc9440a953f17ac7de : SUCCESS in 3m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/034c2a564560426682b6cc7063c2ac66 : SUCCESS in 9m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3e3bcfe8dbde4c118e7d4e9af043b720 : SUCCESS in 11m 23s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/fe0ed9b5b66d4ad68b26badb663ba88e : SUCCESS in 7m 38s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f9f5626f28604f7ab8e4f25f14c9f50e : SUCCESS in 5m 12s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d9f7e9c4db3746b29a20a17b6b78b304 : SUCCESS in 4m 35s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/10a9c6011fdf42da899c7ebdc69fd6c1 : SUCCESS in 11m 20s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/86963d92727b4e7497fd200eee790fa6 : SUCCESS in 4m 44s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/01462a5882d64d2dafa02d5b8ae38178 : SUCCESS in 4m 58s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/bbe6cfabace7443fadf834caf7990089 : SUCCESS in 7m 28s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/decc4d1c252f4fc59af73f83a1036daa : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b53b940ec4f4483493f8577f89804f5a : SUCCESS in 7m 18s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/617fa4dd430a476cb4b5143171311d8a : SUCCESS in 7m 21s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/3c7e68657cf14bcd9d290a1ee99be8ee : SUCCESS in 10m 42s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/41733053d72d424da2950c02192877dd : SUCCESS in 3m 26s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ffd3417ed2514f0c80155c6f22b731f1 : SUCCESS in 9m 59s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3296959b914e4bb195dcfba1bcc50653 : SUCCESS in 4m 17s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2f2f965e211e485c864f4812cd3626cb : SUCCESS in 10m 22s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/26c805597a7c4f53a64f023e28c9e9c2 : SUCCESS in 25m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/bf296dc7c19e439d91f1ce7d3006e037 : SUCCESS in 12m 57s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c7b6d0f4333a4503b3220c72701b519e : SUCCESS in 4m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e35ad7e3fa9546dc8f7a84923c69822e : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bd52bf65e28e4ca19d1cf772e8922685 : SUCCESS in 25m 04s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/028966a0a9e541a6bfa19cfbc3855b90 : SUCCESS in 5m 06s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/0217f59b912748c5af6daecc99407844 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6a028275763f4c0b881f58e28850dfb4 : SUCCESS in 5m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6b7fe856bcb746a49d48ff2838ec70ad : SUCCESS in 10m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4394c2631404f7abdac328388798bd3 : SUCCESS in 1h 04m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/48fa622f3a574442be8ec76cedfe9238 : SUCCESS in 1h 04m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f3b16122bd62db26783d70dab87793378ff1fa66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 05:42:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"272234e557f00fbf91822b5242f061eceb6a636b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 06:52:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/061b8f1193c84b33b06a8d5ec61c6ad1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a382e07af9464eada2bdd873f7e03ebc : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/779d5db0de6148e7ad308948de822fd2 : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffec2d61ad3f4f19a740baede6d5750a : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc680922e85a48d5871d7f75e154575f : SUCCESS in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6f70adb51f92438ba29304c097db82df : SUCCESS in 9m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0d7a75865fbc4f9e97942fd6d9cae631 : SUCCESS in 10m 45s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/b6bf8227f1ef4294887c76c7e9698940 : SUCCESS in 10m 12s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/8466ea91446e4e8baf5ff237412c5c5c : SUCCESS in 4m 13s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/a40ce9ac53c04daaa15f4193e9f9b47b : SUCCESS in 5m 24s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/d07b60d6153a4bef9f9adf38c9e126d5 : SUCCESS in 7m 41s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/c0de3ddcedfc419f918153197c637ec1 : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/400551480eba4316a262f272521ac383 : SUCCESS in 8m 19s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ba36dc40e17a4162b3d46f7cfffa6349 : SUCCESS in 8m 02s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d2d1fce65d7b41ddacbe91c92d998e80 : SUCCESS in 10m 52s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f6f30cc5c7764f7594786f6fe4d28815 : SUCCESS in 5m 14s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/2496415a51734dc7939b3963522fc3e0 : SUCCESS in 9m 08s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ad7e6dd53dac4aea8e6ff599e502d475 : SUCCESS in 3m 55s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/9537a4269ebf4b908e5ecc21d72316df : SUCCESS in 21m 40s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/feb266681182472881c1381e5b8a9081 : SUCCESS in 13m 54s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/497b0bad2fa64f38be374373016627ec : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/414a8785a72b4814b079096e7a24ed63 : SUCCESS in 2m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ed9ddb2bbe2b455e9a1b3d979299db90 : SUCCESS in 23m 23s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/0247b0e6577249c1a0282497e6b6fc71 : SUCCESS in 5m 31s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/1fd6687113f74e7b831e4d471ce878d2 : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8cec5c60335f4af59880d82e86b0ea46 : SUCCESS in 5m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0ad0c4618be84656896852c358d9c699 : SUCCESS in 10m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5dddbb8791a94277b480601d856ec4b3 : SUCCESS in 43m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/07895468ce334a07ac487596f856713c : SUCCESS in 1h 09m 33s","accounts_in_message":[],"_revision_number":1},{"id":"78fbba0f2abf2a106276fbb644379f4d2ebbd482","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 06:52:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8074b407b23f7489bf01bab316dad7c510bf0a9e","revisions":{"8074b407b23f7489bf01bab316dad7c510bf0a9e":{"kind":"REWORK","_number":1,"created":"2024-08-01 14:32:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/925489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/925489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/925489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/925489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/925489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/925489/1"}}},"commit":{"parents":[{"commit":"487812fc17be660c3b977464f6aacc40c215696b","subject":"Update packaging constraint to match 2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/487812fc17be660c3b977464f6aacc40c215696b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-01 14:32:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-01 14:32:23.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 3.4.3","message":"update constraint for neutron-lib to new release 3.4.3\n\nmeta: version: 3.4.3\nmeta: diff-start: -\nmeta: series: antelope\nmeta: branch: stable/2023.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ib98cf12d3aaacf3b2529dced8c8bf5ac4ea6945f\nmeta: release:Code-Review+1: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ic006649f210a249be5844158bbb133599188b426\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8074b407b23f7489bf01bab316dad7c510bf0a9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8074b407b23f7489bf01bab316dad7c510bf0a9e"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
