)]}'
{"id":"openstack%2Frequirements~925494","triplet_id":"openstack%2Frequirements~stable%2F2023.1~I65daf8279bf62d6415f1e4d74dfc03e0eae9fec7","project":"openstack/requirements","branch":"stable/2023.1","topic":"new-release","hashtags":[],"change_id":"I65daf8279bf62d6415f1e4d74dfc03e0eae9fec7","subject":"update constraint for oslo.db to new release 12.3.3","status":"MERGED","created":"2024-08-01 14:48:41.000000000","updated":"2024-11-01 10:13:12.000000000","submitted":"2024-11-01 10:13:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925494-new-release","meta_rev_id":"7cf55d0a5171521fbc3315cab83fb5249055ca99","_number":925494,"virtual_id_number":925494,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-01 10:13:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-11-01 09:01:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-11-01 09:01:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-01 14:59:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-01 16:32:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-09 03:18:20.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-11-01 09:01:45.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"acb69a7a5e30ab468f0eecb9f9fcc05d8d563e21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-08-01 14:48:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c71be7186615ab7147a86bfbb06fa6ffcdc00ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 14:59:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c98c125deb146e5b31b0e2f49a2c137\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8477b3e8af204293b33b393f5fecfa49 : FAILURE in 8m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4a41b68f17051f98f13fc2274485848578fcfe96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 16:32:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8336b6a767b64e2bb3b437e782a56c62\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45fc63bae53c4ddc89a1492c51af11fa : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44f2727d7ed84fd5ba2a5712194fc85e : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2bbfb7993cfe46c686066cd51425942f : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f4f9b9c2a0b74316965ebc828171da9f : SUCCESS in 3m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c865264fed864afd85ee1058347ed8a1 : SUCCESS in 8m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b0bb93f3ac0949e78aa191ec77f6401a : SUCCESS in 13m 25s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/202fd10b9e3040bc9f5d82c36c1154e3 : SUCCESS in 7m 11s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/d3284e551de2476bb913566421d300c2 : SUCCESS in 5m 07s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/f8d0da66783f427f8156b93fde9f1c15 : SUCCESS in 5m 27s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/44c8604c9353486fbcb08374b55ba6ff : SUCCESS in 10m 44s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/afd913cc8e904cfbb0afa7da93b397ff : FAILURE in 2m 06s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/370e7972cd0b434b85f1a136a1bdfa8e : SUCCESS in 5m 28s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/8af3baf3357344a2aa9fc3139f5c2a26 : SUCCESS in 8m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/51e7836beda646f0bd1fb3b00ed73d34 : SUCCESS in 13m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8d91d6454e684fcc9b09a9b71c1c75b2 : SUCCESS in 7m 20s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/65aaf91582834b65aa435d2cd3027c93 : SUCCESS in 11m 17s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4b33f5e2a2234ae0900da65667c6d41b : SUCCESS in 15m 33s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/48933d3d241f486ea61abf4dd0d3a6fb : SUCCESS in 3m 48s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/b5a09fd5a0ec40c5a92edcdb674ec199 : SUCCESS in 9m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/4892a9cca8f94f7ea161342bc7ef414d : SUCCESS in 4m 21s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/19e1034f2fc848bb81a5f6adad748c3e : SUCCESS in 9m 48s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/166243a1c7f943ffa903079fe0778e58 : SUCCESS in 23m 26s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/e3be4bc57fb6468ebe07b68663204588 : SUCCESS in 18m 39s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/be659283c4c94023bb16f09203666fc3 : SUCCESS in 4m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cf5a0fd134b84dee9102aed3e528f6ce : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/66070ba0daab4bf1a02185bc46b7dad2 : SUCCESS in 25m 11s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ee795a445c7e4b5d8f2190b136af27c1 : SUCCESS in 4m 55s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7df4b6d6d7e84a84afb970e79ae5ec40 : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a808c8f49b0049e9b46f856c70a88a02 : SUCCESS in 6m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/babffab4e31b4497a02ec03baa261bd3 : SUCCESS in 9m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10e9659e9b9643ec84c71057ea1fda32 : SUCCESS in 49m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d76121ef699248ebaa07bd9f17481743 : SUCCESS in 1h 36m 47s","accounts_in_message":[],"_revision_number":1},{"id":"1fee48e03c58e7412f447799f5ba3bf4a6702e94","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-09 03:18:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b4c8a5bfc2f189d63c49d0f93d942c43d66375b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 03:35:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f05a5f2161f472fb7f00ba64954fb10\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4ddaf5d4fe6e498ba2d77507f883b0ae : FAILURE in 14m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ea23d17398eb88a89f362c36325243d7b0d3535f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 05:01:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e64a55cf98847cd93282c1940a25696\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d02d45f4bc06437b88f1dbe18b538285 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ad8a9240d664fc89e837b6056bad585 : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/846d683b66b8484991eeef149d679228 : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c9759aeb712645718201fe59d79bb512 : SUCCESS in 4m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/93def78dcbd241c89762656f31a5872a : SUCCESS in 10m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ba8aa6eba93843ab8f7634b8e726b4dd : SUCCESS in 11m 38s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/154426879b1a4e6fa890450f3a1d6190 : SUCCESS in 6m 37s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/f7ce2606d1d841a09891335d7791e2b3 : SUCCESS in 4m 41s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/6a08968810bb4f47ad3f3b0f224b3dfd : SUCCESS in 6m 00s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/70f1fb2c2c6a4547a808672c4783fe32 : SUCCESS in 9m 38s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d7c5b543c8784db6977573daf024db7b : SUCCESS in 4m 03s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/3fa065b8521d478faee67b7177b531da : SUCCESS in 6m 01s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a4e3f55f15784a1f967c99b4ccb0def8 : SUCCESS in 7m 14s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4acdb54550b2490d88a6bd30b9bd3156 : SUCCESS in 11m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d0442b0e07e84e90ba6c1c58e607a5df : SUCCESS in 6m 46s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/85f5e3dbecf044efb2ee682bc95d0cda : SUCCESS in 7m 53s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/6d6d62a7ea734e94a8509fda5cf74637 : SUCCESS in 10m 35s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/3cddb013b66a40c4a891d2f85f62ffc1 : SUCCESS in 3m 54s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/ebc765870c6640a8bb8f8e1d05338889 : SUCCESS in 10m 22s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/ceec90c2eddf4c64a7e7ee54d560bd68 : SUCCESS in 4m 01s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/06a6d41a6ced41b3b68d10211418b163 : SUCCESS in 10m 21s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/70003b0274084464b171eefc0982179e : SUCCESS in 28m 14s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/65497ce14dc744cc9842b94164e321ec : SUCCESS in 14m 36s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/57bbcfa207674393af48225fa867f653 : SUCCESS in 4m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/31189188e4af436a807d30b6a10bcc5a : SUCCESS in 3m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cdbc385a5118495eaaf51fa3233e24fb : SUCCESS in 24m 56s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/60594880151c46e38a99c7a89bcb3598 : SUCCESS in 5m 21s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/3c41cb84d1b94763959e1d24963503d0 : SUCCESS in 4m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3106b99aea8e448c997a1ed10fe37f56 : SUCCESS in 5m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9bc81547f76245f78d224c2ed1eb9296 : SUCCESS in 9m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/229a16806ec34a6b9e3bf64940b0eba5 : SUCCESS in 1h 02m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a093f54a479415a8ab47aebe59c4500 : SUCCESS in 1h 30m 38s","accounts_in_message":[],"_revision_number":1},{"id":"6b1324bb75245e22802fadeff1c1fcef70f34bf6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-01 09:01:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a441b51e734fbe51fc852de7c1b8e319bcc3a17f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 09:02:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d4723f89d0ad08c5a0e247043fc73ddad083684","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 10:13:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b928ebb4b04b4e368e2ece72ef9b1cc6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4dc9e5f5695413d98e22015c6ae701a : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75a5a77df5704db5b508d9d9808f07d2 : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/79923065c5ce46da9e6754beaccee7cb : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1a2cd4a3fd2e487c92d77c56b8405750 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dcfe6900b14f4a06b6c8431045b52d1c : SUCCESS in 10m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/11a51d15399943739a2d55948ead5a01 : SUCCESS in 14m 50s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/637dcca5ec594fef8f68242dffde1637 : SUCCESS in 10m 21s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/2e41f9cfeb9e42118bc263524e7e01b5 : SUCCESS in 4m 23s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/78afc27f4683428dac0d8385df427a53 : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ac22c81269204d84833a5533e2159e6b : SUCCESS in 6m 23s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/0e9f9c8c4a5047889e00d3864f842c2d : SUCCESS in 10m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/840bb599a6d54f66b9507c590ea66abb : SUCCESS in 9m 12s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/95d14943d54a47368a89a9691de36374 : SUCCESS in 7m 59s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/d0ae7a6fc48c4c39be5b71520e3dfa9d : SUCCESS in 10m 45s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/020cc176a409491c9a0c535429abefb3 : SUCCESS in 3m 30s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c03f4662c7bf47b08e1f42830925be7a : SUCCESS in 8m 14s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/8082aba5544f4b6584aca2a0cf4ca74a : SUCCESS in 3m 52s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/ec822210811f44bbb7bf844416332976 : SUCCESS in 17m 43s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c1e6c2828208449ca59cdd1c5a2157f3 : SUCCESS in 10m 06s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/16ad510f343849abb92661966d8960b7 : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e9c0cfa30df44206ab1f02046a7975b3 : SUCCESS in 2m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/888ae1fde03e4da9a012cf74a5a22da9 : SUCCESS in 25m 02s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/c61037aab7064413806aa983fcae62a1 : SUCCESS in 4m 00s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/72ae9e2c61604ac480b32ac9f6c60f30 : SUCCESS in 4m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f07fb01aa39f421fa22781bd417df3a9 : SUCCESS in 6m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b6eb75db70194c15a633ef4ca007a98f : SUCCESS in 14m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f562bd1f38204bb8a5844c53b369b0e0 : SUCCESS in 1h 03m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/12bad9ff96bb426095224f6c5f87a697 : SUCCESS in 51m 35s","accounts_in_message":[],"_revision_number":1},{"id":"7cf55d0a5171521fbc3315cab83fb5249055ca99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 10:13:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b25c6884e5425c32d4093efafda923cbd21839e9","revisions":{"b25c6884e5425c32d4093efafda923cbd21839e9":{"kind":"REWORK","_number":1,"created":"2024-08-01 14:48:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/94/925494/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/925494/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/925494/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/925494/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/925494/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/925494/1"}}},"commit":{"parents":[{"commit":"487812fc17be660c3b977464f6aacc40c215696b","subject":"Update packaging constraint to match 2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/487812fc17be660c3b977464f6aacc40c215696b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-01 14:48:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-01 14:48:38.000000000","tz":0},"subject":"update constraint for oslo.db to new release 12.3.3","message":"update constraint for oslo.db to new release 12.3.3\n\nmeta: version: 12.3.3\nmeta: diff-start: -\nmeta: series: antelope\nmeta: branch: stable/2023.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I075f3ef47ba5c3dec1341bb5cdf49d24ef22700f\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nChange-Id: I65daf8279bf62d6415f1e4d74dfc03e0eae9fec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b25c6884e5425c32d4093efafda923cbd21839e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b25c6884e5425c32d4093efafda923cbd21839e9"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
