)]}'
{"id":"openstack%2Frequirements~925763","triplet_id":"openstack%2Frequirements~master~I385e25cde1f84a68b67e97e8bfd14cd1530204ee","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I385e25cde1f84a68b67e97e8bfd14cd1530204ee","subject":"update constraint for python-openstackclient to new release 7.0.0","status":"ABANDONED","created":"2024-08-06 12:27:18.000000000","updated":"2024-08-28 16:56:30.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"32c998894bdfb23aefd9cf1e1d256700b3703d30","_number":925763,"virtual_id_number":925763,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-08-06 13:54:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2024-08-07 06:56:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-06 12:31:40.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-08-06 13:54:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-27 21:02:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"}],"messages":[{"id":"bf73748ccdf9d391e2c2250bd131e0433cd4bf4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-08-06 12:27:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0626988224727a024a52607e2fe4db329685f88b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-06 12:31:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e79c17b120c632601b62d2ff0730d86356abbcf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 13:54:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/32f9d02560834ac88ae33f9eb60d996e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24fd85b0247f4ae885762db15765de31 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1199bebc0faf4cadbb4b0d0b120fc3b0 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e50cb012e22c4364a9db87b65529e291 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4eaba9477fec4cb3bc00f61b0092757d : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/28574cd9c65141028ff6b4557ee5c872 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f3385a8840c44b6688054e8fd29fe42a : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e69bb2e36a0c4c20b97d4e26f561261c : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d92c8497c02e4b5f85ef84c233db8fc7 : SUCCESS in 8m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/be58f6b296384aa4a460826116ed4070 : SUCCESS in 14m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7141cb45314742e88342c48c693a1496 : SUCCESS in 8m 29s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4bcbbe4f8aff43df8c7da974e8dea4e0 : SUCCESS in 17m 07s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6a392b0ff24c440f9e2df4900328b55d : SUCCESS in 31m 23s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/c9820587f64649f89e98e126f958e382 : SUCCESS in 5m 08s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/3dcd5b1c214540e880b32775138c74af : SUCCESS in 5m 06s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/f92b4c0d5db245d198fb36c431b5252d : SUCCESS in 3m 59s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/fc00a97191e8478d924057ef99df6ee0 : SUCCESS in 10m 27s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/ff9e91ee1b9b45929fb2a784a51787b5 : SUCCESS in 4m 52s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/d6f8c68cde7343e69405d33cff918c1b : SUCCESS in 4m 42s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/78126e1c0963400eb7e64793c8641670 : SUCCESS in 7m 22s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/612d0ecc9d23490ea830025a4a42597d : SUCCESS in 10m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/52ca10ae3d5747d3860392260e6c2258 : SUCCESS in 7m 13s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/abea3855e89d408fa11abba49bb860f4 : SUCCESS in 8m 00s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/92f3969fe8d54d1faeb54998889cb24b : SUCCESS in 10m 08s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/06c7c4793fc74211b775dbb43532c86e : SUCCESS in 3m 29s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/ad888314f7e740c097d7c048087f58e6 : SUCCESS in 8m 13s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/9b8965dac7b9416bbb4ce82e38727fe3 : SUCCESS in 4m 17s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/d56f62fe86ac408faee7b06b95c836e9 : SUCCESS in 4m 39s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/e5c6d5d484cf4040b9b3ab68284c02e5 : SUCCESS in 12m 06s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/eb10701e84b7473da37fbc51f7fdce61 : SUCCESS in 16m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a6d430c9904e4b49a44b3578b0208db3 : SUCCESS in 5m 50s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/c6629b46345d479f94988c2912240ff8 : SUCCESS in 11m 50s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/db39805d46b84735a0b6e34e6bd4fa38 : SUCCESS in 4m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5f72d56869ec4c8faf74e53587c5b629 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b69af9fd71e143c4afa128c2ee06f963 : SUCCESS in 19m 59s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/aff53fb09fcd494a90a91c071a1167dc : SUCCESS in 4m 05s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/b2104290c94a42ad8a97ccb95f01c217 : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b10eda30501a4d44a3aa716b22cde6e9 : SUCCESS in 6m 32s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/40addc81bc4240e18d465dcad0b354e1 : SUCCESS in 20m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce0ea76a7c394b35aaeec5ee02d9914d : POST_FAILURE in 17m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6d5273edda4473b9cb3f5a1820ee294 : SUCCESS in 1h 26m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f19f357105dbc1de78ad60f50b4c59298e4fabd4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-07 06:56:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0a8c6960a94a5a9796c6cbae017e1146c1b2dfe","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-08-27 21:02:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5d046baae1ebce5ad267ee3854f22b18321df7c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 22:55:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c63acc5b571499a9af0d8d7d037e66f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bef668f518b640b6920a97724346dc22 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e27dd457ca7f4903a5833ac223641bc5 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c41f7e26eec49cd8cb16c19ff47bb5a : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/498fb2d1891041c08dc094b0ac0c4c5e : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/02c5c2587cde460a89ba9ae95030c49a : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c01bfca652b346048582d5815dac158c : SUCCESS in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aca018fbbcd34fac8790c1d08e628f8a : SUCCESS in 10m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9b3bd42d83d14b96a67de919cb7d1650 : SUCCESS in 10m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3a5a4cbb30d34f08a43be8fadfdafc73 : SUCCESS in 13m 19s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f1c4b4d956d24491998529130894aeba : SUCCESS in 8m 27s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/86f832b904a64500befa99d499e7eac1 : SUCCESS in 22m 48s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/133a4ef59a2b461fa03af67ee8d6f06c : SUCCESS in 21m 43s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/d0fcacf80c27482cb1bcce2b883716d2 : SUCCESS in 5m 23s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/36fcf000155b49da93e525815f003015 : SUCCESS in 5m 02s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/1292df4f12d34ee4a348a244fac092a5 : SUCCESS in 7m 48s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/fe17a26b9c7e4d8e9cb4dfe148f67a67 : SUCCESS in 11m 26s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/5bd3df507ca7453d80b79bec48212f72 : SUCCESS in 4m 59s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/d340708414994f2bac46ac1c6f74f392 : SUCCESS in 7m 44s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/5fe3ee4de02c44edae5d9ee36069c4e9 : SUCCESS in 7m 52s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/75e5e1fa8b8940058db9c8541c42ead9 : SUCCESS in 11m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ffa145c640b54970b509e766172d47cd : SUCCESS in 6m 29s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/6dd9678acb864950b1c2d23ebd730737 : SUCCESS in 9m 02s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/b596a1ff84b44f7b94b7aec59becd88e : SUCCESS in 11m 44s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/05958cf396834a5580a282f706bc9339 : SUCCESS in 4m 03s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/967d0224b71d471b9f73c5556c89673a : SUCCESS in 9m 34s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/db50c3aaeab7401c8b0dc44e0e964fb9 : SUCCESS in 4m 28s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/e7d02fd9cccb4360b8048fa24e131884 : SUCCESS in 4m 41s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/25c582ee5e1c41caa99eac97a57a4983 : SUCCESS in 12m 22s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/da7eae65581c448d86f01138c3789c76 : SUCCESS in 28m 08s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/10f4e1dea8224d84b63f6426bc0eafa6 : SUCCESS in 8m 00s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/b3020bcbe25c486490c30e5246a06382 : SUCCESS in 12m 17s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/b2b414b27d024f6b95db7e21ab690f84 : SUCCESS in 4m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/84171873f83543d6b57458948dc43e88 : SUCCESS in 6m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6860b485e94d4ab69190505533a76656 : SUCCESS in 24m 40s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/e4ed9237d7764ffb935e5d2da3874761 : SUCCESS in 4m 42s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/d0bb25d31854405f9c63c6f93c0c1968 : SUCCESS in 9m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b5ba70edd2214daa9f57b3a5fbebeb2b : SUCCESS in 6m 45s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/530afe06f82b4c3bb191997fa67ca87e : SUCCESS in 15m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90d282a4e07d4a93a7372de56c50440c : POST_FAILURE in 28m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70b5a339a0994f1ba69dc4c70531eddb : SUCCESS in 1h 43m 43s","accounts_in_message":[],"_revision_number":1},{"id":"771c4daf50f38c700f39882be8485a4999022a7a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-08-28 16:54:06.000000000","message":"Patch Set 1:\n\nabandon this and wait til their commit makes it into the next release","accounts_in_message":[],"_revision_number":1},{"id":"32c998894bdfb23aefd9cf1e1d256700b3703d30","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-08-28 16:56:30.000000000","message":"Abandoned\n\nwaiting for release with the updated commit Jens linked to","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"df163f8c7b68947a01599ed66caa379603e83551","revisions":{"df163f8c7b68947a01599ed66caa379603e83551":{"kind":"REWORK","_number":1,"created":"2024-08-06 12:27:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/63/925763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/925763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/925763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/925763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/925763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/925763/1"}}},"commit":{"parents":[{"commit":"31a9e6a6798373d6d4c534ca4872ea053f73721e","subject":"Remove lextudio packages that don\u0027t exist anymore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31a9e6a6798373d6d4c534ca4872ea053f73721e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-06 12:27:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-06 12:27:13.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 7.0.0","message":"update constraint for python-openstackclient to new release 7.0.0\n\nmeta: version: 7.0.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Change-Id: Ie0cea236828e8a01b61064e996d4cf57555c7562\nmeta: release:Code-Review+2: Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: I385e25cde1f84a68b67e97e8bfd14cd1530204ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df163f8c7b68947a01599ed66caa379603e83551"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df163f8c7b68947a01599ed66caa379603e83551"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
