)]}'
{"id":"openstack%2Frequirements~925786","triplet_id":"openstack%2Frequirements~stable%2F2024.1~I9c269d65a6b30a319b3e3ab5d103c5af0f2e2d06","project":"openstack/requirements","branch":"stable/2024.1","topic":"new-release","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2024-08-07 17:07:23.000000000","reason":"Change was submitted"},"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2024-08-07 15:37:45.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I9c269d65a6b30a319b3e3ab5d103c5af0f2e2d06","subject":"update constraint for neutron-lib to new release 3.11.1","status":"MERGED","created":"2024-08-06 16:59:46.000000000","updated":"2024-08-07 17:07:23.000000000","submitted":"2024-08-07 17:07:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925786-new-release","meta_rev_id":"7a250b50aa01f7ee086f005f19455eace994194a","_number":925786,"virtual_id_number":925786,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-07 17:07:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-08-07 15:37:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2024-08-07 05:52:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-08-07 15:37:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-06 18:24:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-07 05:52:02.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-08-07 05:55:31.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2024-08-07 05:55:31.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"725a1684cb8dcb988f7965b792f6dcca7361d379","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-08-06 16:59:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"901b4db8b6146fbabbf18fff1d034dae7dffa461","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 18:24:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52a2e2aba37b4e3dbf4d016f77cc1692\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7227e3eb1fe4434f913831e4d78fa9a1 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f985ca4613cd450591d31e0b1c41c889 : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d202fb990d8142b6b2c82b6cc6d533a5 : SUCCESS in 3m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3aa250d6e0b04246a357fe7901fbdb20 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/232092c544014249876d9a93a072eaf4 : SUCCESS in 2m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ff877e971413425abf6bd8c891fab19c : SUCCESS in 10m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6ba6ba82ed1e47b0a733aa777b5cda23 : SUCCESS in 15m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/221aacff41db4e4ab4f99cd9d92f1eb4 : SUCCESS in 6m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3225838d1e4443b8aad3d8780d6412b0 : SUCCESS in 16m 15s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/bc347d42b0a84967a0c01626e19c6241 : SUCCESS in 3m 58s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/bc6ed7126f254a788c9c6dc47bf1fa84 : SUCCESS in 4m 49s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/828cbadbba3a4c6d949846886cccdcc7 : SUCCESS in 6m 51s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/abb659bba6d24a35987241bcc84716f7 : SUCCESS in 10m 45s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/84994a945b1d4f65916638de883fc76b : SUCCESS in 6m 02s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/847d494dcbb74603bbb568870eacc2f0 : SUCCESS in 5m 45s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/f10e72c148f54447b0f8a7dc9d6e9409 : SUCCESS in 8m 48s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/be174f98ea3c44f68b8b622afebfa10a : SUCCESS in 12m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cfc81e7071f143a4b673a68300e6f5c4 : SUCCESS in 9m 04s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/1d6540a93555421e9b084f42aaa11cd6 : SUCCESS in 9m 09s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/f275ab5f5255401b876f212c8362d19f : SUCCESS in 10m 36s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/4c4186ee29464a92bf34d861dbf4c4a9 : SUCCESS in 5m 15s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/c7a800dea26f48ed92fc46585d04abd2 : SUCCESS in 9m 33s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/c9e7b21e29c544fca8e302389f9c70d7 : SUCCESS in 5m 01s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/4a30db25ab73432cbdbf3648c1e25a59 : SUCCESS in 6m 41s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/99c4e9499dd142a7998be11bc1185740 : SUCCESS in 11m 09s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/a1d699eec350463d9f1dd2858b6722f2 : SUCCESS in 24m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/455a9d7e136847f98cfce826d2d63c31 : SUCCESS in 6m 18s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/284b6024eb3e4c3ab88b8a93979f168d : SUCCESS in 10m 32s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/2b68df6e528f4290af84413f5f8206d2 : SUCCESS in 4m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bec97cb29b0e43af80b483257b35885a : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/180831d096404af28a8702686d7a51fb : SUCCESS in 22m 46s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/0f446268996948a8b6f12df332608a58 : SUCCESS in 4m 04s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/b941cf7c8c4c44b081f28d7172766ca7 : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b9ceee43919544588b1ae9a868802865 : SUCCESS in 5m 24s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/da42e65150db472d95519052c07f25b1 : SUCCESS in 14m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa8422b27e9042399a82d8a021ff5aa9 : SUCCESS in 55m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f40141a750524e92aa6406791017df9e : SUCCESS in 1h 23m 47s","accounts_in_message":[],"_revision_number":1},{"id":"21a1ce45b0643c090c60935a31e2cda9a4a74d06","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-08-07 05:52:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9a3ecb0141de8599679bec14efeee5d416d91d93","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-08-07 15:37:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3d7b7779856a517773a0502915f4094130dff270","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 15:38:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"75574f92e7396f46a348b08fa934c1b8f12eddf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 17:07:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76b07fe8bfff47549eec8c482927d63e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e71fbaa8754d4227b44344406134e0c2 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c98c7cf651f942d8b84dbe1b9c069431 : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/80cbc2aad28e4339aa6c1041e6721b8e : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7e1425607d934c5d829ca05d460f4d10 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/40d0267b1d794c138d7c1eabc945e760 : SUCCESS in 3m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/016a9e56a4524af1ac3684743af5469e : SUCCESS in 8m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8a7d47de714e4b039a424b338534a6dc : SUCCESS in 13m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/985c3d92474545a1b0cbfdc9011deacc : SUCCESS in 9m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e5f4278504d44315b49031d4d2421899 : SUCCESS in 16m 00s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/55441d2ee723463ea2b6b5b905e9fefb : SUCCESS in 10m 20s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/9509de6db58c4bbfa4b0bba5506ea2a5 : SUCCESS in 4m 32s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/ece3c72d878046308293f3bf2f5f9d31 : SUCCESS in 5m 23s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/e0b0a660eb2349edaed0a1390bea4d77 : SUCCESS in 6m 24s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/fb5169f8efc44924ad325fe3f6b584ec : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/23b92029a5cb42b5bd0dbce07da0c4db : SUCCESS in 6m 47s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/a903dddd7c2749b7aeaafb1289b7138d : SUCCESS in 8m 03s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/0e47f266cad34e2daf7931dc874d12ae : SUCCESS in 11m 15s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/051d69c38bcd493e99a4145cd7759c7c : SUCCESS in 3m 23s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/26b99d07b6194606b7e5b7dfcb0f1195 : SUCCESS in 8m 15s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/ba5b42ff7acd49f7b6314ffc184fbc41 : SUCCESS in 20m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b518090708484a78822bbdc48c923006 : SUCCESS in 6m 03s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/69450958b9234628a67435ca3811de42 : SUCCESS in 12m 38s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/f262eaf3bee44ccc9fd73e7d5ad2b23f : SUCCESS in 4m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/668a3063364d473fb73f76e00c7ba6b5 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5a9059b125544b2ba4deade10fdec850 : SUCCESS in 21m 00s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/ec01d87e1fe94672923a25eee7f5e8ef : SUCCESS in 4m 20s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/da011ff5385b4e35b7680b876531ba92 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ec0feff7c7ec4df1a7385b4be33d4812 : SUCCESS in 6m 08s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/78a3343a66a04b07ab93c4aea9b327b5 : SUCCESS in 12m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/002cc42f87f94791a83b4dca7faac664 : SUCCESS in 1h 04m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb820520ac974f0e8972080cc014a062 : SUCCESS in 1h 28m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7a250b50aa01f7ee086f005f19455eace994194a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 17:07:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c412c5a7949f5e4134b79b4629245cbe95545ad","revisions":{"9c412c5a7949f5e4134b79b4629245cbe95545ad":{"kind":"REWORK","_number":1,"created":"2024-08-06 16:59:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/86/925786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/925786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/925786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/925786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/925786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/925786/1"}}},"commit":{"parents":[{"commit":"3c03e2335bce4dba3a126ef679da4710cab3ada3","subject":"update constraint for python-openstackclient to new release 6.6.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c03e2335bce4dba3a126ef679da4710cab3ada3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-06 16:59:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-06 16:59:45.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 3.11.1","message":"update constraint for neutron-lib to new release 3.11.1\n\nmeta: version: 3.11.1\nmeta: diff-start: -\nmeta: series: caracal\nmeta: branch: stable/2024.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: yatinkarel \u003cykarel@redhat.com\u003e\nmeta: release:Commit: yatinkarel \u003cykarel@redhat.com\u003e\nmeta: release:Change-Id: I9f7991e51a1b8fdd7910a567ad62c2cfdb7ad71b\nmeta: release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nChange-Id: I9c269d65a6b30a319b3e3ab5d103c5af0f2e2d06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c412c5a7949f5e4134b79b4629245cbe95545ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c412c5a7949f5e4134b79b4629245cbe95545ad"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
