)]}'
{"id":"openstack%2Frequirements~927101","triplet_id":"openstack%2Frequirements~master~Ie9a4296489f27e5bc020955ab67cd3d1d39ea9c2","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie9a4296489f27e5bc020955ab67cd3d1d39ea9c2","subject":"update constraint for os_vif to new release 3.7.0","status":"MERGED","created":"2024-08-26 10:47:18.000000000","updated":"2024-08-27 11:23:52.000000000","submitted":"2024-08-27 11:22:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927101-new-release","meta_rev_id":"c4a044f51b6464bf8c77fb52a265d9f0d32b4098","_number":927101,"virtual_id_number":927101,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-27 11:22:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-08-26 15:25:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-08-26 15:25:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-26 12:17:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-26 15:25:54.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-08-27 08:07:19.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"}],"messages":[{"id":"d04321e68168b49015d317714b2648c4ab33d8af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-08-26 10:47:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d99d9f9f7fd9ea91361cfa5fd771292d3e1b91ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 12:17:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91cc95846e9d46dbb0ce0bafa2bb496f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28115d2ea7c9449582526ac9e5da25e7 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce0509a6ca23434aabb8536c7c65384c : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/019af227934b478285f0670815363da5 : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eaa9fa40ed884e8388793fac5d81c499 : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/890e6aa539504849b04511f66c531e55 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd14262da15043b1984be2977a8bd97b : SUCCESS in 2m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c546e3b86544dd3aaaadc49cb39bc0e : SUCCESS in 3m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a10c490eff494b22baf59de6f2b9678e : SUCCESS in 9m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6e44ddf6057943eeba0e3428606fd2f9 : SUCCESS in 15m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/799146ba5f0f4aca8cd620c41a6195f1 : SUCCESS in 9m 19s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7c36e9b1fbef46a2baa681b2e2e6ecea : SUCCESS in 18m 55s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bd31027db4c24058aca6f305cfd8d735 : SUCCESS in 30m 18s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/7945393088844fbaa92f6c92696c91d6 : SUCCESS in 5m 14s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/a4222cdf1cbb4ed7bbfdc4b93e30ddfa : SUCCESS in 5m 34s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/3a1eef3c93b244098b798a27a2bbb229 : SUCCESS in 4m 10s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/1f87d25c708944fdb2ff911164409621 : SUCCESS in 11m 21s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/f82a74b31b14410d933d23d9bfb1886b : SUCCESS in 4m 40s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/84fda0cdd1074856aa2518dd6e204057 : SUCCESS in 4m 57s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/7a0f1890b00f4035bf15181d6f6ce79e : SUCCESS in 8m 12s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/6a9d6d7091094f7fae2d5881c1ef679e : SUCCESS in 12m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2410f901ecc0495ea3ba7936eac97cd1 : SUCCESS in 6m 46s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/66ef4a26e0e04624ba6b1d9d674e9e4a : SUCCESS in 7m 25s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/42ab157927b549f49978682217cef3ff : SUCCESS in 10m 06s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/8ad9e4eb7dc24b8f8cb8c06e223842f4 : SUCCESS in 3m 20s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/7c7672d2550446cfb7fe8bfd5d95475e : SUCCESS in 11m 04s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/f2b2c61299204c638e96d7f796286fec : SUCCESS in 4m 28s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/e7480db228de4271ad5d8e3751c3671c : SUCCESS in 3m 58s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/3d10273bbcd34921bdd4e93fc60cccf8 : SUCCESS in 10m 39s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/39be6c52060e49ef988d33ba6da19841 : SUCCESS in 19m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7585988d44da4b52a25f301c8f9c5464 : SUCCESS in 6m 42s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/7bdaafa52e984db283b525d3c2e3a88d : SUCCESS in 11m 15s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/e15e6f4fbb464d1aa17ea8b4158ccbc7 : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a1ecc8400e604e91a468a039914edc35 : SUCCESS in 3m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/acbf478d8a51437e8a4524ea8187f6bf : SUCCESS in 24m 20s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/63369dcddbbe43128d886a1562d85eea : SUCCESS in 4m 42s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/96f1cccde40f44c2b2772e4eb2258a7c : SUCCESS in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/15989f97d42f4d3386445f97a6342fc1 : SUCCESS in 5m 50s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/435c1d584d474a47af15668210cd77cc : SUCCESS in 12m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bff1840f65b413583ad5cd7954f6ca4 : SUCCESS in 55m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13234784e5e6491c8a9e6809ee2d709a : SUCCESS in 1h 26m 00s","accounts_in_message":[],"_revision_number":1},{"id":"59e8b4e651618b9b5463a7c82277153cfc308b2d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-26 15:25:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"60bbb187f739e9a09675ed62babcf8263d5bddf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 15:26:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"322061ca571a08c84fae255f05863bac8d5339d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 19:14:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/13d6111032cb4fbc9f6ad62b5429c796\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc1ba2cb9a63447886f3f765b5f23af2 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9b7cbab04bb4220be80dd81217df536 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/641e700e95a245128fd22bfc0c15b470 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/264c0924dda7495386c932f5306ca32b : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/37ea1db5287943308a6e9e2930549a97 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/19299c27dba14227aab755a3e70c9260 : SUCCESS in 3m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/766a376903764609afa12f8ca49e39ca : SUCCESS in 7m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/17ded5c54c7c4c79a075e5e41747a9e8 : SUCCESS in 18m 08s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ce818b98da684fc78006981c7045195f : SUCCESS in 9m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8da8e5535fb041eb845bd31d10dee4df : SUCCESS in 15m 56s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/f4598d4fc2bc430e9326094a31ad83f8 : SUCCESS in 11m 11s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/f023f70ce648405791bec84af6cc01ed : SUCCESS in 4m 51s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/f6ae97431f2b4d11a0d0e3719afda351 : SUCCESS in 4m 53s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/d25654bc852e4d4da4dc588d5a93c76d : SUCCESS in 8m 51s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/b24d4096fef649a087e60fb357678605 : SUCCESS in 13m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/59d82f7ed6e740499aee0034cf6cf88f : SUCCESS in 6m 34s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/666d98e2b21e4bac83c693872dae7d09 : SUCCESS in 7m 27s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/e8d76fb5d233461b8a1b32aa7bd002e8 : SUCCESS in 10m 14s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/79a18e987ef34d1ea88cc3f5e374db96 : SUCCESS in 4m 00s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/4c2b7c7198e547c1b97b8df660b1e370 : SUCCESS in 9m 02s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/8b2e2ab98f7e40178614e0121637a8fa : SUCCESS in 22m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/199a64696834497094b3d534ba163ff1 : SUCCESS in 6m 25s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/3ed5344bcef64becb85d7b29ffe26f1a : SUCCESS in 16m 24s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/06f924fec829499ebd98b6de69b4554d : SUCCESS in 4m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d11cfbe77c8a49fb9d8b1ec99cba6956 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c5acfffa0a0c48c4b32d8cd55da0b588 : SUCCESS in 23m 39s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/5d3e396ec3dd433fb86f76139fb64aa5 : SUCCESS in 4m 16s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/0f098d55bb6f4934a549ea027c92b0d7 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/79b37d5748524b93ae05e413fb0f3228 : SUCCESS in 6m 50s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/07443248018843309f4d4b326c91c056 : SUCCESS in 20m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f7c57745b1e495e86b9a2fe2578f6d6 : SUCCESS in 1h 00m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/67c0eb453611423abb52b9683840874f : FAILURE in 1h 24m 33s","accounts_in_message":[],"_revision_number":1},{"id":"91200877d9cead880739a7a180346bc0ea6488aa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-27 08:07:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bed9fd04289751ddef2cf5ff950ce4884cd2b382","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 09:50:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6b3480e425e425a913acb4b1ac6b7b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33951f588f9945ac919cc3949042fa9f : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3274cbbc608e494e935e4d978bc3f71a : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/480fa130bf9a45d7b9666a203cef6d15 : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/834e789b94cf42f4ac6142ac3b5eab00 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/835eae4af856420f9ba804db1530cf41 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4276a6ec430f45978a62ea36aec8bdf9 : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e605bffc9d5747e58d66e50e93f919a3 : SUCCESS in 2m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4a86b76c1ef44bf88540df2ca69ce135 : SUCCESS in 8m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/447f700a5f6f4e5bbd86a8883eab0f17 : SUCCESS in 14m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/773f3863c0174e6bba1f98d1535f46d0 : SUCCESS in 7m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/80509b6454ad4dd68dcb31e517cef154 : SUCCESS in 15m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/54b12f629d84474a9cb76ae19b595e55 : SUCCESS in 25m 25s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/f0d6264a7e5d4cab95c16719ff999ef3 : SUCCESS in 5m 46s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/7ab721e46459481e9fad9f743c918af2 : SUCCESS in 6m 04s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/8b8544f629b94de5b664a73e5cdca237 : SUCCESS in 3m 50s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/db318e200b99456094046faa56654931 : SUCCESS in 11m 18s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/1217198fe5a94eca9e05109cca985a46 : SUCCESS in 4m 13s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/0da5790c324641d181aaeac2a1091e8e : SUCCESS in 6m 13s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/ed383830626842a19ed81bdce37d7e9e : SUCCESS in 6m 33s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/5fb2b0f4c80d41daa617778ff7eae63b : SUCCESS in 13m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/74101754c3104f76b340e18522de60da : SUCCESS in 6m 48s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/a33fc5ad19b24ba992a0a46fa204c189 : SUCCESS in 9m 25s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/a44d09a4532746f5903743ac326c1477 : SUCCESS in 9m 56s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/77e1f3a3a38747ca93720fe10ff8a6cc : SUCCESS in 5m 05s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/c4cd1099ffef4d7caa97c621c6ed69e2 : SUCCESS in 8m 58s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/7eaa7205189441df81ae34fc9578ae3c : SUCCESS in 4m 18s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/6fc3fe42c43b48ae99aad6d8583c9140 : SUCCESS in 4m 20s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/54a8a5e1cf0e4579b2ba7b45d62470e6 : SUCCESS in 10m 27s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/bf142d8f66d54f4b85897355045afe6d : SUCCESS in 19m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/65155d952d274e5d867983f402a630fa : SUCCESS in 5m 59s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/6bc4675dfe984c7db3c4d6219d152633 : SUCCESS in 13m 00s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/ad69842926374697a4db4e9041812ab6 : SUCCESS in 4m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/79906cc3024d4c97923b539550109bf9 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a389d26aae9b418dbdfc55302204a2b3 : SUCCESS in 19m 07s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/25c6ea7b980d4b78a1b5718364960584 : SUCCESS in 5m 23s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c945e6ada8444ba7951aefd0d34f8c04 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ba5c9936655e40128f355ba482769af4 : SUCCESS in 5m 08s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/fc5739af59cf48fcba535d8a371efda2 : SUCCESS in 14m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b24b716daed45fca3437791154ae1fd : SUCCESS in 49m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9a85095370b48249f7e3c083f9ff8c9 : SUCCESS in 1h 37m 08s","accounts_in_message":[],"_revision_number":1},{"id":"37ddbcce2d804a9382817257aa46b6dd37cf5b7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 09:50:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"174827594b8f1d3ff317b837532013851d2df3de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 11:22:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd4ba47bd60a4815a8bc0b93a080f849\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83aa389cbf2b432585e8460979bf90d5 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6167ade67d98436ab5dfe2bf3dd9c401 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e19f892e3320405da2d80ab4a64546cf : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9820ec8f864f405fa45486c53d532897 : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/55da9416480c44b6a7b14b3fc67ecf81 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8d8f88c4a91e4e8a9337191587b55e58 : SUCCESS in 2m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f49da3a7f5084c2ca7670fe4dc7f7dd8 : SUCCESS in 8m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ec928fe52aa04960ba7233f0523deaa8 : SUCCESS in 14m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f13a0cf3d7464f288b21f8e1b17c8196 : SUCCESS in 8m 17s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/48e9cfa48ca14a959a3cd77f61c3bafe : SUCCESS in 15m 08s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/ce8093a7525649aeb45cdbb0f8196521 : SUCCESS in 10m 08s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/bef885d1417045d19889b2826003fa1b : SUCCESS in 4m 49s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/353f6b33d7fd4ab888ee999ecc6e0ffc : SUCCESS in 5m 03s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/c1476b9517c848e381a4d7e8041b4173 : SUCCESS in 6m 42s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/86cc468b2c8940a9990685002bea4d11 : SUCCESS in 11m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/395165d27bf949c4b0deb12e5e3292b4 : SUCCESS in 7m 50s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/5437b24fae1043c0a9a6ca17f065f1ae : SUCCESS in 10m 12s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/93c4b25cdb9f4226a03205c038383267 : SUCCESS in 11m 21s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/814b5795230e4c79bf1a49a29579e03e : SUCCESS in 4m 11s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/49dedc772a294bbf99a57cff521443e7 : SUCCESS in 8m 33s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/f51d1254590b4aa1a8f5cd6cf5b2e82c : SUCCESS in 25m 25s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f4d6fa56bcad4c1bbff85932b58e4cec : SUCCESS in 7m 14s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/ccc559382e6a447a90bae9efc4a1467c : SUCCESS in 11m 18s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/035cb20fdf5c4691b159721b27146cc6 : SUCCESS in 4m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1c87567b0ae34a939a969acadb203b81 : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/42f3b6c4f3a54d0db3ae06ea651d68d4 : SUCCESS in 22m 16s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/8d29c8e8ed9e44298b042055baee5954 : SUCCESS in 4m 03s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/4b3e5daed1b84f06878406112c1ced4a : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/faacb16a4759452cb1336a3b55af9a36 : SUCCESS in 5m 25s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/c6014bc425d64108b147cab1602de532 : SUCCESS in 21m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d49dbff19654f2280b22c98229636e5 : SUCCESS in 57m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0640501ff7ed4d799911de19ce9ea32e : SUCCESS in 1h 27m 51s","accounts_in_message":[],"_revision_number":1},{"id":"5d44092217a1a6e5a502684e2ecf8962ab7fbced","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 11:22:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c4a044f51b6464bf8c77fb52a265d9f0d32b4098","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 11:23:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e322825a523c4583a0f795cd7bcbb36a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40dd5b5f484140768b8704eb5f16c30c : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7142a0f22c18d11913546499aa7640ada6c522e2","revisions":{"7142a0f22c18d11913546499aa7640ada6c522e2":{"kind":"REWORK","_number":1,"created":"2024-08-26 10:47:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/927101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/927101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/927101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/927101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/927101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/927101/1"}}},"commit":{"parents":[{"commit":"bdab288f3c4fa57dcc66e52416789a0c7d49d7ac","subject":"Merge \"update constraint for sushy to new release 5.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bdab288f3c4fa57dcc66e52416789a0c7d49d7ac"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-26 10:47:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-08-26 10:47:17.000000000","tz":0},"subject":"update constraint for os_vif to new release 3.7.0","message":"update constraint for os_vif to new release 3.7.0\n\nmeta: version: 3.7.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nmeta: release:Commit: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nmeta: release:Change-Id: Ic034923d8d74ca2358e093915adfccda7ef2e922\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: Sylvain Bauza \u003csbauza@redhat.com\u003e\nChange-Id: Ie9a4296489f27e5bc020955ab67cd3d1d39ea9c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7142a0f22c18d11913546499aa7640ada6c522e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7142a0f22c18d11913546499aa7640ada6c522e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
