)]}'
{"id":"openstack%2Frequirements~927751","triplet_id":"openstack%2Frequirements~master~Ib9152266608613ab3e58863abb025618f5d50625","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib9152266608613ab3e58863abb025618f5d50625","subject":"update constraint for openstacksdk to new release 4.0.0","status":"MERGED","created":"2024-09-02 17:17:08.000000000","updated":"2024-09-04 09:41:49.000000000","submitted":"2024-09-04 09:40:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927751-new-release","meta_rev_id":"807f4afc18fed24af810368c4f05e28c47727d85","_number":927751,"virtual_id_number":927751,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-04 09:40:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-09-04 07:04:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-09-04 07:04:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-02 18:42:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-04 07:04:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"24b2b4fcd92f21e02b072abb8817e1a70e80b4f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-09-02 17:17:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51194910f97cd8b3faf16bf559873bc92feaa290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 18:42:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b36376ce0cca4444930773aa4b1b4515\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11f5a673873a4ce580870dbbf43a0f10 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a18194a438f449a959802eba3150118 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3482a3dda5204d688eb239d6a2ed23d4 : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c38c5444e8b400d838615fd3bd98e46 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7704115b710245deaf96a2cb959751c8 : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7eb574d3b23e4227bb60100cf554791c : SUCCESS in 2m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b7d662daa0b2425587398435000e6e49 : SUCCESS in 2m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4b2ef3aaa545461c8de9d19b9162a715 : SUCCESS in 8m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2397430339634d9c9894cae4ee92419c : SUCCESS in 14m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1fc892d96d1540d299bcc84ae06f23d6 : SUCCESS in 8m 11s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b8976d12890e4645a8c3f1bbc8ea44e3 : SUCCESS in 16m 26s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a8e547b3b80f49dbb91ca489b1b11788 : SUCCESS in 26m 07s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/8b4164b3c4294eb686da02513c5caf9b : SUCCESS in 4m 36s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/71d3b16f0f654464ae5f4206a3f4af1c : SUCCESS in 4m 14s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/5e60a84eafa2415fa7e4d52d6a333c0c : SUCCESS in 3m 42s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/9f359cefa19f4f498ab823ea71343e7c : SUCCESS in 11m 11s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/056489cc61f3427e8bf431e4597fdb2c : SUCCESS in 3m 55s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/07ec864e259247e99b58bf21ceef317c : SUCCESS in 5m 19s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/aff9c651d0ec48208fa99fbb2d8bef3e : SUCCESS in 7m 10s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/9e2015ccff1f455fae905ea0082c1b85 : SUCCESS in 9m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/253f6fdc7863462c8b7ea3db62ba57f1 : SUCCESS in 5m 31s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/a49a29082ce04077ad44207792876a8e : SUCCESS in 8m 40s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/6266b57223f64fa6a093008377d90a5d : SUCCESS in 10m 45s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/21ba7e865adb4815a46f804310bcef39 : SUCCESS in 3m 07s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/aca309b9795544caa654d9e27481b197 : SUCCESS in 8m 12s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/ce5038e17d1b4c23bc0946adb341a7fb : SUCCESS in 4m 12s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/39f3cacb5e9c49cea6743196ba902151 : SUCCESS in 3m 42s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/cff75a3ef4cb4c6d9efdecc486db7b16 : SUCCESS in 8m 36s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/9792e5e22bac4b75bbb770ad83b489f9 : SUCCESS in 18m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/64f921b47bd44bfd93f67576f03fe2da : SUCCESS in 6m 55s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/eca1cdca8dba4cc7819d6cc192c1aaac : SUCCESS in 11m 16s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/abb75fe70001463d96b9fd1c2e190dff : SUCCESS in 3m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3e90daab9ec243e98894cd3458841ce1 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a794c0176ee48a680e867a6459de5b5 : SUCCESS in 27m 38s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/e8d518ee455d453a8db0ba799e18abd4 : SUCCESS in 4m 09s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/36e7343d0e394acfa8e29993126ee586 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/32c711d4f85f4c7ea943b4181443a161 : SUCCESS in 4m 53s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/e3454d1bd3ac40d39ee24c318202e316 : SUCCESS in 11m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e410ca050a084777a00e12e327d27db3 : SUCCESS in 46m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/595fe3c63f754f98885a1c33ed050ee0 : SUCCESS in 1h 22m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ca5cbe1f2cc67656f88295c5cc374f4241130ac4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-09-04 07:04:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"93dbb8306a9f800b3fa51461f89e1d46c7e33409","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 07:04:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2414adb67da0690f7f940472ff5414027f20c60b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 09:40:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/544225f65aa04f6eb35019075e67002f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/724bdc060fb140c4bab447236093094b : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/618002e1483643988dd1c1559ecab900 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7b0749c1ca24376bcb008d1bcea9bc6 : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/292d874e7cb2443996b75e9a7fc75f2b : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/13b944d13faa400eb255ec0a2270fd8b : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/72b5c3cf8a704c67ba340dce76df712a : SUCCESS in 2m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/29427b394e7a47c5b7df77a082d2b9fc : SUCCESS in 8m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8884fcdd5a934dd6951bccd846c63b66 : SUCCESS in 14m 31s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/eda1b7a945c949d8ad3c23589b3b4d45 : SUCCESS in 8m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/05d794e88f004c9e920ef1d1784a40ba : SUCCESS in 20m 20s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/0ac7de9b473441e5877233565e548bf6 : SUCCESS in 10m 33s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/639d9c2b71014bd08795a4d88d55cc33 : SUCCESS in 5m 23s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/246f036b9c0b4880a93e8b12e711ef75 : SUCCESS in 5m 10s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/dc7df4fae84943cdb0e714e3ee8b567d : SUCCESS in 6m 55s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/072849d8b69944f5bf865e7de23fa4bf : SUCCESS in 12m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0f70a31e0094452b94fc27701a2e1d15 : SUCCESS in 7m 26s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/cdc821bb8ae24b6fb89343551c73f288 : SUCCESS in 8m 34s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/3ebb6eaf10cf4aa1a72b46a349e75bcd : SUCCESS in 13m 10s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/61912cc0ebef46fab3565b03a68bd3b0 : SUCCESS in 4m 04s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/258c75cd92bf4af7af57550ee56b908f : SUCCESS in 11m 24s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/d68ecf773c334c1caf0716ef2098ecd6 : SUCCESS in 16m 16s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3cf325d3707243c6a95739ab1ef71706 : SUCCESS in 7m 42s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/66d2e3c5085e46fe9433f331afc75ed0 : SUCCESS in 10m 51s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/c82ccd79e0134528b51f9c362928aad9 : SUCCESS in 6m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/133e1e5b3a2a437f940e1cb34801beb5 : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/01d6bd966ef14ad1bdb20e6c59504565 : SUCCESS in 21m 38s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/589548a5098849988742e3819a955ca8 : SUCCESS in 5m 36s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/e85289bd574e417397ae3e8172ed8a3a : SUCCESS in 6m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/490cd111281f4332a17553eda3fcb4bd : SUCCESS in 6m 42s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/0b0366b5e6de4a3391bfe4a5dbdc206f : SUCCESS in 20m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/178e5625e20f408b83cb4a067b943be4 : SUCCESS in 55m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/252ea86d59c84de5aad6d85425f46692 : SUCCESS in 1h 45m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e50dfecb251b744d666686ed8f63f41906fbb153","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 09:40:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"807f4afc18fed24af810368c4f05e28c47727d85","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 09:41:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b32b2a68e959438ab3f33fe39c8e6e65\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9b886319d654e7cb87cfbfea64aa9d5 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0c0c84f3fdfbcee8fc21697f1275863b30e3f20","revisions":{"b0c0c84f3fdfbcee8fc21697f1275863b30e3f20":{"kind":"REWORK","_number":1,"created":"2024-09-02 17:17:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/51/927751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/927751/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/927751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/927751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/927751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/927751/1"}}},"commit":{"parents":[{"commit":"f39baa8538e0776e61391694df722b0faeb6cd84","subject":"Merge \"update constraint for keystonemiddleware to new release 10.7.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f39baa8538e0776e61391694df722b0faeb6cd84"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-09-02 17:17:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-09-02 17:17:06.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 4.0.0","message":"update constraint for openstacksdk to new release 4.0.0\n\nmeta: version: 4.0.0\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:1d9596288 pre-commit: Bump versions\nmeta: release:Change-Id: I39894cb4866c479dba5c19eb4ae9fe1a41cb975c\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Workflow+1: Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: foundjem \u003cfoundjem@ieee.org\u003e\nmeta: release:Code-Review+2: Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: Ib9152266608613ab3e58863abb025618f5d50625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0c0c84f3fdfbcee8fc21697f1275863b30e3f20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0c0c84f3fdfbcee8fc21697f1275863b30e3f20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
