)]}'
{"id":"openstack%2Frequirements~928092","triplet_id":"openstack%2Frequirements~stable%2F2024.1~Ia8ed993858d4c8bf5af4795fb1668ee9a8544747","project":"openstack/requirements","branch":"stable/2024.1","topic":"pin_bcrypt-stable/2024.1","attention_set":{},"removed_from_attention_set":{"36670":{"account":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"last_update":"2024-09-26 04:24:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia8ed993858d4c8bf5af4795fb1668ee9a8544747","subject":"Pin bcrypt until passlib is fixed/removed","status":"MERGED","created":"2024-09-04 21:21:34.000000000","updated":"2024-09-26 04:24:29.000000000","submitted":"2024-09-26 04:24:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928092-pin_bcrypt-stable/2024.1","cherry_pick_of_change":921873,"cherry_pick_of_patch_set":3,"meta_rev_id":"ba9804cbd27f2bf7a6bd5017bd610dd7e36d519c","_number":928092,"virtual_id_number":928092,"owner":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-26 04:24:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"recommended":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"all":[{"value":1,"date":"2024-09-04 22:08:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":2,"date":"2024-09-26 02:45:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2024-09-09 14:14:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":1,"date":"2024-09-26 02:45:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-04 21:21:34.000000000","updated_by":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"CC"},{"updated":"2024-09-04 21:45:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-04 22:08:39.000000000","updated_by":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"reviewer":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"state":"REVIEWER"},{"updated":"2024-09-09 14:14:02.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-09-26 02:45:10.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"d73aad9f215472f5a6eaa61b79ac0a6191516daf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"date":"2024-09-04 21:21:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* global-requirements.txt\n* upper-constraints.txt","accounts_in_message":[],"_revision_number":1},{"id":"cee2b77adbe8ccd47167db24bc38a46049ef207a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 21:45:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1c7489c7eb24793a40ea3a6c2f6537f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d2f1c20f34346398d2a3db386c1f363 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ade5817f6e2d46748d71bddffdeeda53 : FAILURE in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bfba30bcf79747f5a6b9f5d91e998c08 : FAILURE in 2m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b0016c99db0341fb9dd5bc5d399d5d01 : FAILURE in 2m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6a19656908114b4a8b4494348cc8ca81 : SUCCESS in 2m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7edbdbea586349989e94c547bbcc0de5 : FAILURE in 2m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4c3502c2a30443e3a4d7d53831a26c7f : FAILURE in 2m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/6b64a2a346c148fb9bb161c9a6e5691a : FAILURE in 2m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/06f7f1a5028e4c3d8c638f1038fb3d4a : FAILURE in 2m 42s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/74a25d1b2c0849dabd569a51d154aeb0 : FAILURE in 2m 50s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/6216a407ce494d10831261bc8c994e16 : FAILURE in 2m 56s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/b616f64d35eb401899dd68192a25c5c9 : FAILURE in 2m 29s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/2ac40668073a46fda316d1ff497de25a : FAILURE in 3m 10s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/76cb2ee5fc144cb09aa81ec759eca308 : FAILURE in 2m 12s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/1f3d5bc4e87247969bde8d5a4ff311a3 : FAILURE in 2m 39s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/15ecda8c645f4971a25f7bfa5dc5f9d8 : FAILURE in 2m 43s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/196a310c21c947e2a2fee64e6da9b82d : FAILURE in 1m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e3867f73395343d88cd287680abaa7f9 : FAILURE in 5m 07s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/90bf3550c8394f31916e42be92e28346 : FAILURE in 2m 51s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/8975fedd22894df48681d3f3603de953 : FAILURE in 2m 49s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/95df5566c283430e9cb929fdf91c9616 : FAILURE in 1m 56s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/bf58860809194e71b1da0127991cf0e5 : FAILURE in 2m 59s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/dbee8dde63cc484f8e98bb9f5f63ad5c : FAILURE in 2m 10s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/74a69bbbc6ea4f97aefd2a78a8637c88 : FAILURE in 2m 24s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/de7fd1dcaa0547578b79924e964fc717 : FAILURE in 2m 31s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/a6e701f6048e459cb97f8ad9f62d511e : FAILURE in 2m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/10e7672e99924580bcdc2dfa8b2c0750 : FAILURE in 2m 52s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/4f20183cf84645dcaa8244fff7cde6b1 : FAILURE in 2m 44s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/ad09c2ac3d73474d983018cb5816e6b6 : FAILURE in 2m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b6b43c3ef9d241c4a2fd45767a5c16e1 : FAILURE in 2m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b0994bcbaf6546a8b332eac89b829ccf : FAILURE in 2m 37s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/23df155ddf4d434cac1ec1deec16daa0 : FAILURE in 2m 39s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/7373747fa08046edbbca5eef5893a827 : FAILURE in 2m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1373f140e1c14d0e9a4ce5deb5acb3b7 : FAILURE in 2m 26s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/c73c3f01a70e4af09cbb20766705d353 : FAILURE in 2m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f22e42634cab4c96b4a33174caea2b45 : POST_FAILURE in 5m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a72cb5720e984010a761e70d0e17185b : FAILURE in 17m 10s","accounts_in_message":[],"_revision_number":1},{"id":"87adc50b39411b2e4ddca9ddc4923f88ac50c32b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"date":"2024-09-04 21:47:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8e480937dcd0bc87b97c0e6409c989fe0c4848a8","author":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"date":"2024-09-04 21:48:09.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"04825bd60cb0ce6e235b4aa94dcb028d634918f8","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2024-09-04 22:08:39.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e51284d241f81497167ca26ad83260e4812d404d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 23:33:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5c2fc20cac2404c95357fa934f07f01\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6753b69f36b9443e8a2d3908a2ceb12e : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3c458300d7b47d4a9eb471b69c5e2f5 : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10ff2025f90a48d78ab15bb80dc848b6 : SUCCESS in 3m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e9d0e78850b4aeeab9a6162a11cf61f : SUCCESS in 2m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a74ede537e2848c498ae6b95369cb85a : SUCCESS in 2m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ea747b5ecff7471c97c7ccc06862e903 : SUCCESS in 8m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ca1de6efa7734667929057acb2c461ab : SUCCESS in 11m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f460c269e0e34fddad5fcc9a6d367750 : SUCCESS in 8m 23s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/56c584325b014be68e05cb02b022a7fc : SUCCESS in 16m 31s\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/b84816bd4d404d799bcc8d8c2b89170a : SUCCESS in 3m 51s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/eb6b3df9cdbd431d8bcb30c42db4eae7 : SUCCESS in 3m 57s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/009df9e9a23242bf9dfd4b7c3ec23f73 : SUCCESS in 3m 24s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/a47c6dc0b12c40cdb20fb2d7fec42d95 : SUCCESS in 14m 17s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/44020023d77249f9a814532fd0c9d392 : SUCCESS in 6m 47s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/64257ed1b6664b61a7f8a158b08154a8 : SUCCESS in 7m 09s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/25b6ccb142694f6dbe2aa2a04641a6d8 : SUCCESS in 10m 31s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/089183105a2e4fa8b187255c8958f5ca : SUCCESS in 10m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9e6c891f6fc04d7f9eeee64595f17720 : SUCCESS in 7m 02s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/99a39437586a43a8838a1e8c917b843e : SUCCESS in 9m 21s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/ae1efe66876443c7ada03f6c8ed6b2e9 : SUCCESS in 11m 02s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/3e7f38e55c794b369f28592c2398a54e : SUCCESS in 3m 56s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/9eb9e3e547be413a9eae35e48952b25c : SUCCESS in 9m 37s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/f37da419affc4aa1ba3bf193847f3145 : SUCCESS in 4m 34s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/bc9847d1779d455eb4c7cfda95494dae : SUCCESS in 4m 47s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/252ef63147554a2eb4c8e596d6b6931a : SUCCESS in 10m 46s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/287930f43b3049168de9ce091670f1e2 : SUCCESS in 22m 23s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b68ac87c5ee44465b5cd287bbf690601 : SUCCESS in 6m 30s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/a69291a499b642539104b4c9a58998f7 : SUCCESS in 13m 48s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/5b198b50b1c246919dec1ee8269ecc52 : SUCCESS in 3m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/25f76ede44384f92a9b612cb91b46352 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a1f9c16ebf034ee1bce9a51a1deaeef2 : SUCCESS in 22m 53s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/84a7fc50af294f83a2a90599178fb3ed : SUCCESS in 4m 16s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/03e4db15a03b4b3f945b7d320794ad41 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/75a789d9effb41fdacca74a52041d75e : SUCCESS in 5m 33s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/a6fcac5a732c448f9feae5b50de7dcf0 : SUCCESS in 9m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f5bc925d44e460e9644e6cfa3aa094f : SUCCESS in 59m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/857fa6b6e9b849c9a22b1b0ce3470aa5 : SUCCESS in 1h 39m 22s","accounts_in_message":[],"_revision_number":2},{"id":"3ff0678c0e052595489b5dae5ee18735e7a279da","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-09-09 14:14:02.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"59234424414e49b09239c7d4102da75fef4b4043","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-09-26 02:45:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1b7ef30f725a2a70f59529c6bb38534c7141da39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 02:45:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"04d2ddd5e4b368bd3a66a4902a4200451b4f9787","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 04:24:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/611240db0b174f2e99bafbb1dfb341bc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d4535605b9b45e1a5b432d019d182cc : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95ec278d43794c6cb340fa5436ac9fae : SUCCESS in 5m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/27dd009dcbd249d784115e44af530085 : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a41360412e546dd817d2957324de5b4 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/97240e5415634036837b80e30951aceb : SUCCESS in 5m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ac3fd79d65f04071be56b338681bb752 : SUCCESS in 14m 20s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/70cf347e38194ce2b8a21e616359e2b7 : SUCCESS in 15m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3f0099ab00a644b2a96f1a60b8f54e28 : SUCCESS in 9m 45s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/23f7ad1f75f3466ca66e26fa3a99bdf8 : SUCCESS in 20m 15s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/267b99946de246f5beb8367190733412 : SUCCESS in 10m 37s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/a4abf35e648340d1aa50846537f0d2cd : SUCCESS in 4m 33s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/f567fd6ba9ff48e4a222b32764297843 : SUCCESS in 5m 45s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/6a7a5b9670a14bf7a5b865c382a1a063 : SUCCESS in 9m 39s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/7d13b3b6b3e54f1a8ce14a45dd98a2c0 : SUCCESS in 11m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dff23db8a0eb423f96a5d661114d50c0 : SUCCESS in 7m 52s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/b2042edf1628445f87da35326b336e16 : SUCCESS in 10m 22s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/b845568a0793411a96ef5b32a05ee61d : SUCCESS in 15m 18s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/71e8878e591b48a7b47a8128f4fe21b2 : SUCCESS in 4m 16s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/c7d84bab74d1476fb62f905a640521ec : SUCCESS in 10m 56s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/d6e32d7b34d541f594bf24a96b021bc7 : SUCCESS in 30m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ca10c270ec854685b036e70ffb05e61f : SUCCESS in 6m 56s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/a24e0e11dfd246c1b6e02c92625893cc : SUCCESS in 14m 40s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/994af4d88a3348d7ba15323547700b70 : SUCCESS in 6m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9235f6020cdc42529119e24f3cb708d6 : SUCCESS in 5m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1112d5dd33954c8caaded37d68731a71 : SUCCESS in 28m 49s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/5225ae89528241acb6c9f54a76f30e12 : SUCCESS in 6m 40s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/3df6ed7d6d544a97bd5fdb1ce7adeb3c : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a57f87c0802641f28be07030103bf6fd : SUCCESS in 4m 49s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/ba97fc3a52f74e29b7d6f320ca8f0890 : SUCCESS in 14m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d77c9fbe7eef476ea7d9ee7b57ec5806 : SUCCESS in 45m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e239f3df37e4caa84c9bb019bf42720 : SUCCESS in 1h 31m 52s","accounts_in_message":[],"_revision_number":2},{"id":"ba9804cbd27f2bf7a6bd5017bd610dd7e36d519c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 04:24:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2f72094e9b8f690a3e3cc61d3d91305eb234d0a1","revisions":{"e78acaf16bf6f0c58b08d0c584b944c9ae65cdec":{"kind":"REWORK","_number":1,"created":"2024-09-04 21:21:34.000000000","uploader":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"ref":"refs/changes/92/928092/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/928092/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/928092/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/928092/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/928092/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/928092/1"}}},"commit":{"parents":[{"commit":"4ecf0a84f11bbeda6c6cd64c1d2fe7ba412ec656","subject":"update constraint for oslo.serialization to new release 5.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4ecf0a84f11bbeda6c6cd64c1d2fe7ba412ec656"}]}],"author":{"name":"Dave Wilde","email":"dwilde@redhat.com","date":"2024-06-12 14:59:05.000000000","tz":-300},"committer":{"name":"Jay Jahns","email":"jayjahns@gmail.com","date":"2024-09-04 21:21:34.000000000","tz":0},"subject":"Pin bcrypt until passlib is fixed/removed","message":"Pin bcrypt until passlib is fixed/removed\n\nCurrently there is an incompatibility with the latest version of passlib\nand the version of bcrypt we have specificed in the upper-constraints\n[1]. Passlib also apears to be unmaintained [2] at this point and we\u0027re\ninvestigating the removal of the library from Keystone. For the time\nbeing this pins bcrypt to a version that is known to work with the\nlatest passlib that is available.\n\n[1]: https://foss.heptapod.net/python-libs/passlib/-/issues/190\n[2]: https://foss.heptapod.net/python-libs/passlib/-/issues/187\n\nCloses-Bug: #2068879\nChange-Id: Ia8ed993858d4c8bf5af4795fb1668ee9a8544747\n(cherry picked from commit b9c903e926087fe2f29bd92c9465a3564ccb2a1f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e78acaf16bf6f0c58b08d0c584b944c9ae65cdec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e78acaf16bf6f0c58b08d0c584b944c9ae65cdec"}]},"branch":"refs/heads/stable/2024.1"},"2f72094e9b8f690a3e3cc61d3d91305eb234d0a1":{"kind":"REWORK","_number":2,"created":"2024-09-04 21:47:52.000000000","uploader":{"_account_id":36670,"name":"Jay Jahns","email":"jayjahns@gmail.com","username":"jayjahns"},"ref":"refs/changes/92/928092/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/928092/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/928092/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/928092/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/928092/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/928092/2"}}},"commit":{"parents":[{"commit":"4ecf0a84f11bbeda6c6cd64c1d2fe7ba412ec656","subject":"update constraint for oslo.serialization to new release 5.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4ecf0a84f11bbeda6c6cd64c1d2fe7ba412ec656"}]}],"author":{"name":"Dave Wilde","email":"dwilde@redhat.com","date":"2024-06-12 14:59:05.000000000","tz":-300},"committer":{"name":"jayjahns","email":"jayjahns@gmail.com","date":"2024-09-04 21:47:50.000000000","tz":-300},"subject":"Pin bcrypt until passlib is fixed/removed","message":"Pin bcrypt until passlib is fixed/removed\n\nCurrently there is an incompatibility with the latest version of passlib\nand the version of bcrypt we have specificed in the upper-constraints\n[1]. Passlib also apears to be unmaintained [2] at this point and we\u0027re\ninvestigating the removal of the library from Keystone. For the time\nbeing this pins bcrypt to a version that is known to work with the\nlatest passlib that is available.\n\n[1]: https://foss.heptapod.net/python-libs/passlib/-/issues/190\n[2]: https://foss.heptapod.net/python-libs/passlib/-/issues/187\n\nCloses-Bug: #2068879\nChange-Id: Ia8ed993858d4c8bf5af4795fb1668ee9a8544747\n(cherry picked from commit b9c903e926087fe2f29bd92c9465a3564ccb2a1f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f72094e9b8f690a3e3cc61d3d91305eb234d0a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2f72094e9b8f690a3e3cc61d3d91305eb234d0a1"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
