)]}'
{"id":"openstack%2Frequirements~928348","triplet_id":"openstack%2Frequirements~stable%2F2023.1~I95f2974a49585e84c0f57cfb343fee7afbe58209","project":"openstack/requirements","branch":"stable/2023.1","topic":"new-release","hashtags":[],"change_id":"I95f2974a49585e84c0f57cfb343fee7afbe58209","subject":"update constraint for octavia-lib to new release 3.2.1","status":"MERGED","created":"2024-09-06 12:13:07.000000000","updated":"2024-09-07 00:08:49.000000000","submitted":"2024-09-07 00:08:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928348-new-release","meta_rev_id":"ce107b6ba1b99e28955419ed903498dd1fcd4b1d","_number":928348,"virtual_id_number":928348,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-07 00:08:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-09-06 22:45:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-09-06 22:45:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-06 12:26:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-06 13:44:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-06 22:45:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9bfd06a8e6d619936132cad2b8ede3419456ab2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-09-06 12:13:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7cd4b53c9d0e1a0c5675804b82a3df20b2f2fddc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 12:26:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2f798bb306b460e9eadbbfdfca09d36\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/707b985f3f604c948833f0883681dc4e : FAILURE in 12m 02s","accounts_in_message":[],"_revision_number":1},{"id":"83f9ae349a9b04093b803635111c9174a2abd070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 13:44:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5255f6b8d154b798a601c9f1bfac1b6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4262227706894c4386221655eaecf6a8 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74d8d54de5a84ce78d218eb096bb9e44 : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dccd2d416d3947e6bbe9a99b7f2120a2 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3b4d42e707f348799f2b4950273ff620 : SUCCESS in 3m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/223f40524e6443e8b36f41601f31a378 : SUCCESS in 13m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4a2564f873594dd29c242538ae96ad44 : SUCCESS in 15m 38s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/b3dbee836b7f47c092f9753dc757790c : SUCCESS in 7m 46s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e9ed4a53be114cf496417ba0ac32ded0 : SUCCESS in 5m 39s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/d6f9ecf0a9584b489954457c81af05dd : SUCCESS in 4m 39s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/882b1caddb464640a52c55bb000d0665 : SUCCESS in 12m 42s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/95aee71b484349cebc1daa248c21853b : SUCCESS in 5m 26s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/bec00447cce446c382072b5952c67956 : SUCCESS in 5m 09s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c77208ce22644a9e98ebad0dfdd0c3af : SUCCESS in 7m 48s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e57dbb3b5b144bc7a45a8b33b674b913 : SUCCESS in 10m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c89cf351e8d542b5b87425b0cc235a24 : SUCCESS in 9m 03s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/4c903ee6a0c148989da1ff4f7ac7406d : SUCCESS in 8m 13s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/df430d07fa5d41ada8467e0baea957e6 : SUCCESS in 11m 37s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/fbb382e2a39d4c0dbe3daaa8c58ea9ca : SUCCESS in 4m 01s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/c64cd1093fb347c4836a6d12c1e5b4a6 : SUCCESS in 9m 10s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/bde7adfa561a4e4eb79bbf334e4f0b31 : SUCCESS in 5m 11s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/312e9c9adb2e42c3b35d32bb528ac793 : SUCCESS in 9m 04s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/674cfbd3fa47494d89997b73bb0d6776 : SUCCESS in 22m 32s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/97911eb87bab40b5961515749740da27 : SUCCESS in 13m 54s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c64975ee7d27448a92b5b67ba764d369 : SUCCESS in 4m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a135c6ba51ff4fea9ec9c443d3e4d58c : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d48ee8c5b18f42bd9c617c23fa3763fb : SUCCESS in 26m 00s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/bd72887e265643858ba726cfbf57218c : SUCCESS in 5m 16s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/136b0b59a8734ae3bee9180023b053f3 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a754fc3d45234d39ae072d2d5bea6e73 : SUCCESS in 5m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/95ba1db3fac7420e955a1b0d93307e3a : SUCCESS in 12m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f13dcb95a9ce48e7a4539cef3dba3120 : SUCCESS in 50m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ed5e86e2de84d41a15d7ba151a02a6a : SUCCESS in 1h 27m 02s","accounts_in_message":[],"_revision_number":1},{"id":"15c942c023fd33dee7677d6bbd7d100082f74c51","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-09-06 22:45:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7a1ad8a9ee3c652003c4d2f0f2eb11547d76e4bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 22:46:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37650e92ec622ba5db7800ad3209c74aa830529a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-07 00:08:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab1b477f07f54022abf62ba67b0e132e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feb85d473ac14add99300e53cff49156 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f7c5f158c9c4669ae016ffc436dfa79 : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/142229dc319d49c286f20373172d040d : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d0861d4ef2574f0d8bc9d2688e9d2e1b : SUCCESS in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c7e4b004759a439f95022b2ea0118859 : SUCCESS in 9m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7c5286fefe1948bab1ce1afe475d028c : SUCCESS in 13m 36s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/cff303d58e5743c2bb5a5ba0bec912d8 : SUCCESS in 10m 28s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a59ce1017b3f4486a243cb9391ba3cc5 : SUCCESS in 4m 49s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2c09262a92a04870bf991c4ca39b8098 : SUCCESS in 5m 06s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c71131a2efd7445ba24576198fd25606 : SUCCESS in 8m 03s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/e4a9f7bfd1d843eab964938fbd507b93 : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e5b3bf07b4ed4b3dac2aeeed26ab3aad : SUCCESS in 13m 32s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a6ff2ae6734d4f8eadda9fb1ce99a334 : SUCCESS in 9m 28s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4e21f05d2042497d87a6d965f2311332 : SUCCESS in 11m 31s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/dd070b28141c488cb6183ea93c9adcbf : SUCCESS in 3m 27s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/3caa3609fdd746f2bb5e0ec4b9853688 : SUCCESS in 10m 03s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/1277a28a6c7444d0b73c46ccfc1b751f : SUCCESS in 3m 19s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6e0558b138db4fdf84c1bd789e47f952 : SUCCESS in 20m 24s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ac51eb8cab9649048d2aa01cbc3d734a : SUCCESS in 13m 53s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/8b22653072c7441d9a858f3950e41afd : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7b20f9af0e8f474a8e31c2343ed631b2 : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/432fb72d8bf745d1ba12405defca4e96 : SUCCESS in 29m 28s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e5577378dded461894fc2f1b285f3256 : SUCCESS in 5m 20s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7e73bd275b8143b1b9d3c89542da0f2e : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/09e2eea46fd440c8ad0e8df1cb7c0c4d : SUCCESS in 5m 49s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9a52a52024a443a88288c2abc5a07d06 : SUCCESS in 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45ed1dbfabeb42c3991870672ef414d1 : SUCCESS in 42m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b432535873ca4f879ab1445f526e0a2c : SUCCESS in 1h 14m 30s","accounts_in_message":[],"_revision_number":1},{"id":"ce107b6ba1b99e28955419ed903498dd1fcd4b1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-07 00:08:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"939b69461ef0225645791f7d7c50d36cd99677fe","revisions":{"939b69461ef0225645791f7d7c50d36cd99677fe":{"kind":"REWORK","_number":1,"created":"2024-09-06 12:13:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/928348/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/928348/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/928348/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/928348/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/928348/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/928348/1"}}},"commit":{"parents":[{"commit":"bcf6e847eb53621ef833acc512ca532aac20bb2d","subject":"update constraint for oslo.serialization to new release 5.1.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bcf6e847eb53621ef833acc512ca532aac20bb2d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-09-06 12:13:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-09-06 12:13:06.000000000","tz":0},"subject":"update constraint for octavia-lib to new release 3.2.1","message":"update constraint for octavia-lib to new release 3.2.1\n\nmeta: version: 3.2.1\nmeta: diff-start: -\nmeta: series: antelope\nmeta: branch: stable/2023.1\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I9e6fd11891f8834cc64d287325e9887b10247842\nmeta: release:Code-Review+1: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I95f2974a49585e84c0f57cfb343fee7afbe58209\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/939b69461ef0225645791f7d7c50d36cd99677fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/939b69461ef0225645791f7d7c50d36cd99677fe"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
