)]}'
{"id":"openstack%2Frequirements~929504","triplet_id":"openstack%2Frequirements~master~I9dfaf51401eb7e61e85fdfdb4f81938b21b1772e","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-09-17 15:45:05.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2024-09-17 15:45:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9dfaf51401eb7e61e85fdfdb4f81938b21b1772e","subject":"update constraint for python-openstackclient to new release 7.1.1","status":"MERGED","created":"2024-09-16 13:42:01.000000000","updated":"2024-09-17 15:46:03.000000000","submitted":"2024-09-17 15:45:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929504-new-release","meta_rev_id":"121a5521b273d988267c51de3cc7954cba719565","_number":929504,"virtual_id_number":929504,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-17 15:45:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2024-09-16 15:27:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2024-09-17 14:03:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2024-09-17 13:35:17.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2024-09-16 15:13:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2024-09-17 14:03:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-16 15:13:43.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2024-09-16 15:15:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-16 15:27:50.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2024-09-16 15:31:26.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2024-09-16 15:34:11.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"CC"},{"updated":"2024-09-17 08:38:27.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2024-09-17 08:38:27.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"500d303bd1b71d18844dab11176d240be0dc3d68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-09-16 13:42:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c74f863df6fc548aac1a718f2cc1cb62af68495","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-09-16 15:13:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"86c1acfbeaf740419abd38ed16c24e7b942e81a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 15:15:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae1cb71c4f7c45b0b8e7b08e91554ff2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a3a3dd476804c1fb2b5493da5375e5a : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04a795a306884600b1bf7aa36e4682fa : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43239335d30442c3ba83368a96496645 : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/263125aa9699418e82d2ec10ae2fb82a : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/42cdcc833c3345388b8d32ee928dac85 : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f39a942c5080430fbbd143542659d00f : SUCCESS in 2m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a1cbc9c7cc9d4d31bef84728d4854c76 : SUCCESS in 3m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/902db5d74dd345f592c9a4de0a989342 : SUCCESS in 8m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7d664063ecea41dbb02691a948166d19 : SUCCESS in 9m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/57f27b34a72e4e8bb727ef0efbd294a7 : SUCCESS in 8m 52s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fb32e420da724e3fab5ea432dddf12bc : SUCCESS in 10m 03s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/03c1157e6a074c8bad3fce02719b9aab : SUCCESS in 30m 15s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/c7e5419ca1b447a88aa510263003d183 : SUCCESS in 5m 01s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/8ac1291eca1a470894ff0912ad1a3bc6 : SUCCESS in 7m 08s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/8226a448f0a34bc08e961693be11604f : SUCCESS in 3m 52s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/28672970d21540839d2dfebccbef4476 : SUCCESS in 16m 01s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/1178c7814c1f4e3db0fec34bed3c2cf6 : SUCCESS in 4m 25s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/4550876d93654b09b3dfea2c674fdc79 : SUCCESS in 9m 32s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/785a67e6169c4516a26c36756b3b4403 : SUCCESS in 11m 05s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/13c941e033e1480ebd0ddb9dd40e4aea : SUCCESS in 14m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ed73a211815d470e9b44d5c1adc1f536 : SUCCESS in 13m 43s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/67f2470bde434bd0a4fb34b4531cf367 : SUCCESS in 7m 24s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/e8b62be39fdd44aa8d01ad5d52a4cb3f : SUCCESS in 10m 05s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/2d3d76f6f9224e4fb69069ea0357af8a : SUCCESS in 4m 12s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/de95d0ed283b44c98b5d37e9a8d456f2 : SUCCESS in 7m 16s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/bb7a26c4796a4c7ea9a1b2511c6db3a9 : SUCCESS in 4m 35s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/510927ea2a924729aab936e0c8fb1ecc : SUCCESS in 3m 43s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/fc2317fbe3de48c19e9b031bfe564f50 : SUCCESS in 11m 37s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/0c94d1c9b75c4445b7566ae8fc3b0395 : SUCCESS in 20m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/324a6268cd694aaf934fdfaecbcfda9e : SUCCESS in 5m 00s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/94d2f3866f2a4f78b9eb7722b493e834 : SUCCESS in 10m 33s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/4320cd2f11a547b2be4da11b0736c224 : SUCCESS in 4m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1a5a913ba45b4ea2ba936dc0803c3ca5 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/55051f2cea4040acaeaedc8ce1fb763f : SUCCESS in 19m 42s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/b309b1652b0c4bf8a5960bfcac0dbf60 : SUCCESS in 3m 59s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/e771ccd2d701400987715bde06deed37 : SUCCESS in 3m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3aec92dfa0b8474a995237181e2f9818 : SUCCESS in 6m 15s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/74ff562d49a54bee8fbcf81e08c79eea : SUCCESS in 18m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3797ba759d94c099b36450db29b0e58 : SUCCESS in 21m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb432b7a9553446687110aa1bb2c27c0 : SUCCESS in 1h 32m 57s","accounts_in_message":[],"_revision_number":1},{"id":"2a98a7b97a9bbc35107037430faa9e0a1f8c2ed3","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-09-16 15:27:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7d23b9003973aa51a07d3b55b13824544a87bd6d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-09-16 15:31:26.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"182266beb61a83963cf9157e7d64191d41f9b315","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2024-09-16 15:34:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2eb915a3f08cd16d36abf6f98b16fd4bcbdda5e5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-09-16 16:22:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a79a13f3e2e3731ddf641bc75493540ff31b026","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-09-16 17:04:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"25d485b6db5b4c197ba4190b6a1fb32f136019d3","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2024-09-17 07:49:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df6fea4875614b79258ac3c5026b7091c6f9327e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-17 08:38:27.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9b7d20254ddb2d2761b07f4b2d2bacc8f3d8671","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-09-17 13:35:17.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"851f6379badf7f65c1cf80f923020fff89244d2d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-17 14:03:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"754cb36509586959f36f8c3af3edd6cf158ffb8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 14:03:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"83842db6520eb72bda8a661719528963867bf949","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 15:45:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04d06922a3de4d28aa5eaafb8f8d5e36\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a615c09694b49dfbb48529d48b6ddcf : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12ab74b36fb24e03b75b52870215ab43 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f2f9e6f7e364987a66116941f280586 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/68d7fe51370349fe9fff16456160d316 : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f664c4f06e7f4f76b71f4beb8c8de97f : SUCCESS in 2m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c7e634c8e0f2469182fa22bade4b7445 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0825be25821b4011abc1972a034751e4 : SUCCESS in 7m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bf69983af53a4d0d9575c0bd732474f6 : SUCCESS in 16m 17s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/04c6afc0d75b48e8b85eb0718656b4c5 : SUCCESS in 5m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/935eff53fda546429d950d856dc952e3 : SUCCESS in 18m 43s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/85723ea59a0149b68cc4a0e7cb587019 : SUCCESS in 12m 13s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/4e7bad97107349bcbf16fcaa76845808 : SUCCESS in 4m 47s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/8425d784778243b2b7bb7a90643ce2db : SUCCESS in 6m 27s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/d32f471cf0024199bee4076f541f1842 : SUCCESS in 6m 51s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/5dd96e519de44ae2a5972483fbb544b0 : SUCCESS in 14m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/03b19df83de246b5943895090d3320a2 : SUCCESS in 7m 04s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/c2b47e8908304ace892ce516fa97cd84 : SUCCESS in 8m 41s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/ad167cfeaab541daa99d9cad70d3c3d3 : SUCCESS in 9m 24s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/28462fc15bc046e883bcaa4cc03c30ad : SUCCESS in 3m 52s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/f8357897b68447bea1219e8ba9490f8b : SUCCESS in 8m 05s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/b59aec3bbe8f4d6da9194dac06399508 : SUCCESS in 17m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e798907aa3124c00a11a1fb65cddbb6a : SUCCESS in 7m 23s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/fff74a093fa740c9a5abac526c11597c : SUCCESS in 11m 36s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/ac6747c9c3ed4ae4a13a71f30c8b0de4 : SUCCESS in 4m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a574c7f8b4874d53be059630017ca41a : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8ce17b20b3bf4ff7bc66bc751f8102e5 : SUCCESS in 23m 19s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/2bcaa8930e014428bfb4989567251463 : SUCCESS in 5m 53s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/b27c000aa50e45318c6105035ea4674b : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0483c9ee482945de930e54db3501d613 : SUCCESS in 4m 15s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/09db55c1488d421db5afd07fa7f06316 : SUCCESS in 15m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c9f26ee66ff45cea8868a94496ee4f2 : SUCCESS in 35m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eeb76ca562e74e8e814e6795008c2e98 : SUCCESS in 1h 36m 38s","accounts_in_message":[],"_revision_number":1},{"id":"cf8063d093cd106d3b56cbcd0ef10dc904d31a4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 15:45:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"121a5521b273d988267c51de3cc7954cba719565","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 15:46:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc3baafed483464d81adf9fd0522247e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/625758f3b83c49a799d6e2717457f23b : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"139b63c77cf4f1343298699c4648157f0a1fbf8a","revisions":{"139b63c77cf4f1343298699c4648157f0a1fbf8a":{"kind":"REWORK","_number":1,"created":"2024-09-16 13:42:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/04/929504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/929504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/929504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/929504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/929504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/929504/1"}}},"commit":{"parents":[{"commit":"59ddc5f862e46c065337467bad673f3d716dc77a","subject":"update constraint for python-openstackclient to new release 7.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59ddc5f862e46c065337467bad673f3d716dc77a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-09-16 13:41:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-09-16 13:41:58.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 7.1.1","message":"update constraint for python-openstackclient to new release 7.1.1\n\nmeta: version: 7.1.1\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Ifa6cbc5f017cd5074fc269e21fde63666672cfd9\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Balazs Gibizer \u003cgibizer@gmail.com\u003e\nChange-Id: I9dfaf51401eb7e61e85fdfdb4f81938b21b1772e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/139b63c77cf4f1343298699c4648157f0a1fbf8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/139b63c77cf4f1343298699c4648157f0a1fbf8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
