)]}'
{"id":"openstack%2Frequirements~931262","triplet_id":"openstack%2Frequirements~master~Id8ad5dfadf7bfeb1ebd44d353d9b2ca1d1219d30","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id8ad5dfadf7bfeb1ebd44d353d9b2ca1d1219d30","subject":"update constraint for os-traits to new release 3.2.0","status":"MERGED","created":"2024-10-03 10:52:42.000000000","updated":"2024-10-04 14:29:59.000000000","submitted":"2024-10-04 14:28:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931262-new-release","meta_rev_id":"e2a17d12cca449be6b145fe5a6a8d32acb2e6adb","_number":931262,"virtual_id_number":931262,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-04 14:28:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-10-04 00:25:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-10-04 00:25:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-03 12:18:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-04 00:25:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2024-10-04 11:29:39.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"}],"messages":[{"id":"90ed29955a198002725a95d33fee08e90b3b1113","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-10-03 10:52:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f8dfd57438aa608e29a114b24bf20ca819d8d0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 12:18:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8dff25f476f404eb462636491c2702f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e575a295af474c81bb21a8ea28fd1374 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af16ff30b80b48f685dc9d481e23a843 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61da3605338d4527b2ed63a424958daa : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a9ee0f862a14723a673563ee26253da : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a3048f1e1f0449d88ec5f3ec1d2c4eb8 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/80d363e7b58d4a9d9d31afb0e1108c4c : SUCCESS in 3m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d8c1f70f8d744df99f61a18126445593 : SUCCESS in 10m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2aebe6f1ad4c431a89bf9106a4a863bc : SUCCESS in 27m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/548039cfe35042d8819850354932fa62 : SUCCESS in 9m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7eaf7b3ce11c4453a581c68e8c9c75db : SUCCESS in 13m 38s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/21f08ad2f7ed4e0083c3b7f2a8b3a81d : SUCCESS in 24m 04s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/db3c1c795a634d5f81904c47f70f9710 : SUCCESS in 4m 44s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/3e6c119167c3405ab74239e2324e9354 : SUCCESS in 4m 38s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/1527408b23d94d95a9f13a9b811ae19b : SUCCESS in 3m 20s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/016a8dc77b8c41fcb4678bf237e2786d : SUCCESS in 10m 04s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/9d9c30b56fc14af0ba28e81e036741e2 : SUCCESS in 4m 39s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/5a259abcc98447bb9eaad31780168707 : SUCCESS in 4m 10s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/8456370c733b4dc6bae04d50f0cca2f8 : SUCCESS in 5m 57s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/09c2f2b184344a91aba6a506ed006b18 : SUCCESS in 11m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eae97993a9714595a5e240de4d287bfd : SUCCESS in 8m 38s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/56a9fa33e8b4416a9a35c6ae14968c83 : SUCCESS in 7m 21s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/16fcc93320264df1b3d6f2ae8c2d52ac : SUCCESS in 9m 00s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/3b457c02a2044184a127a12a1522a681 : SUCCESS in 3m 48s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b0e7762db65d4e0a8f31b7cd53d3e86a : SUCCESS in 11m 12s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/b69274131f6e4198956087e06ed5a9f5 : SUCCESS in 4m 08s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/28ec75791b5f4fbab5b17f89afd4571c : SUCCESS in 4m 50s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/236d2aa0107c40c69f5a06f0acfaf338 : SUCCESS in 11m 18s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/f8453479b8db48cf88f794ba11da7f4d : SUCCESS in 19m 18s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c5054eb753e54793b20eccd5563b6ea2 : SUCCESS in 7m 28s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/84980c6d61c043a884bfca6e97df9613 : SUCCESS in 13m 49s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/3fc84705af8742b7b0aa8b69a1e7dd23 : SUCCESS in 4m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/19f0a22ef1ba434cbe4c7a72aa43454d : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/69a7f311ca7149a99d7c6dd6fcdeef93 : SUCCESS in 21m 01s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/0b5bb8e1bddb4baaaaec0c6829b6dbb4 : SUCCESS in 5m 55s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/93c40ceebc4c4adeafc8f07bd44ae329 : SUCCESS in 4m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/132a16cab5934169a04d533201f8fdf6 : SUCCESS in 5m 53s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/f92fc3f014554c77b7198ac8e301985f : SUCCESS in 8m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0addfeba435d4ceb92e95a6b142c40bf : SUCCESS in 38m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/551c8e96f05542158ba8ab772691e625 : SUCCESS in 1h 25m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e1f1abd536c00c7068515264789f7b5651a2e867","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-10-04 00:25:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"50aa067910f067865ae4cb1bc180de80b3056403","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 00:26:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab7075415b8a0e2d775b5e5e3a5c58e503d92a8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 02:16:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0af3054304e94251ab297ffe2400d842\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5e2f149e5b841339dfd1ec9294783e0 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c34941f83c824300933d77749d683ab3 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abcb9676c0b146649d43152a30f2baac : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/be42925bd6204de4a737908eaa2ecd3a : SUCCESS in 2m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/03e24e78ad184c37b168c37ea5c30a97 : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3224357fd3ea4eb88bf6e0d4781beb96 : SUCCESS in 3m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/00620754917b42a7a1fbbada1b03d9d2 : SUCCESS in 7m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fa382aa87ea74f98885e6c9c41e22fbb : SUCCESS in 12m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/abbb6f86e3ce4a81961eb853483507dd : SUCCESS in 5m 31s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/3334f005da9743ac8c0a3ff26c90ac3d : SUCCESS in 9m 32s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/9012e4c68a8546c6af800e8a033b3445 : SUCCESS in 10m 11s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/ed11a3fedf664abfab53a0e4714670cd : SUCCESS in 4m 00s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/e031423676d542579195f8b8988758f8 : SUCCESS in 4m 56s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/1f064ae448c34e758913a96635bc41f2 : SUCCESS in 6m 48s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/e6e4ced65afb426a8206a15b0d1f4617 : SUCCESS in 11m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/024a455107c34ee694b472f1e33c99fc : SUCCESS in 7m 21s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/e43a9f013ac24d18ab474285f3d645fa : SUCCESS in 7m 09s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/e15ff3f85ccf41f38429e6d64e035ad4 : SUCCESS in 8m 57s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/e8bd29e2202d47bb8ac16aecb67b6bef : SUCCESS in 3m 50s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/d014e9696a8a4be79af88b32e91d9693 : SUCCESS in 8m 54s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/17e031251e7b448e8d0471f55f157db3 : SUCCESS in 19m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fe78dc38ef6a440e83dbc7159456dc80 : SUCCESS in 6m 16s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/5123c62823df4e29a58ea2daa5c16263 : SUCCESS in 9m 36s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/3b723cab25f945859284ca8e5e66d970 : SUCCESS in 4m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/01410969b47b45769eac8a089e6395df : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bb14435f22cc4e03a1978c8d783e038c : SUCCESS in 26m 23s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/87a77a8c52e44bfeb6e96f5ca0d522db : SUCCESS in 4m 24s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/8bd8c1f5210149218570546f7fec9ff7 : SUCCESS in 4m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ff3f920a98d14ea6b022add3aff5595c : SUCCESS in 5m 51s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/7962c4157c1a45c98cb8034d7bb88282 : SUCCESS in 11m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e44f8c0024a4569aab33b4ef3b7a6bb : FAILURE in 37m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05d2cf58762b4c458ecf2861c026cd36 : SUCCESS in 1h 45m 03s","accounts_in_message":[],"_revision_number":1},{"id":"9cad1117a23e810f96d54df83fc34531d613b818","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2024-10-04 11:29:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f34497069db6ff0c961f5f872e0642e31fe3edc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 13:02:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8275210f175448938e41e31afa8d8d49\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ac0e9d6ad4d4c449e2fbb1ac1db3895 : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb4c49cf273a4dcd8f3f1249b8ccb873 : SUCCESS in 2m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e90023ac92a94292913912aff40a44d7 : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1db6266ff3fe4d8a8ae0ca7c27fab3e0 : SUCCESS in 2m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0506811666db4992a38d1ff835504f2d : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6e921e3df90e40cfb97767fa4ec83699 : SUCCESS in 2m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1302ef20fbef49a18fc324d91b14d1ac : SUCCESS in 9m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7a86eac8c2bc461dacc86c38a867cf60 : SUCCESS in 14m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9aa3bf3b60694582b6e2f58dfe77006b : SUCCESS in 8m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/616f2a74dc3641e5aa5d8bc0ed01eef7 : SUCCESS in 17m 12s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f6fc32d305204f53bd7cf4b85bda033d : SUCCESS in 24m 03s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/6e4a1462d2fb44a5b59d0085538f3704 : SUCCESS in 6m 31s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/45f3430680134e35b30028f057fb3add : SUCCESS in 6m 18s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/b5307634b85c4643842a4e761504f1c8 : SUCCESS in 6m 20s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/423f307daeae4a4582e7a1d65c096841 : SUCCESS in 12m 43s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/62b10c4fcc144b538e38939f0c4d5996 : SUCCESS in 4m 18s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/be81d7870c0e46ffa901905e2b4f634c : SUCCESS in 6m 18s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/48d8edd3610248d89446be7bd21b5cdb : SUCCESS in 6m 36s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/260fafdced504d1cad1c76b3f623845e : SUCCESS in 12m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6c3564769ca14ec282851d20adf54f0f : SUCCESS in 6m 23s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/23464637dd754a59b03a048fd3ee45c2 : SUCCESS in 9m 21s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/5b99ca00d1344d57828e04e7a9b1e42b : SUCCESS in 11m 31s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/af0ad0fa69694c89b240e4ebeff4a31a : SUCCESS in 3m 40s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/89529963cd36476fa6ca422ea98071f0 : SUCCESS in 9m 39s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/7e12bcb128994a95b373138024124b69 : SUCCESS in 3m 04s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/71fb8d8658f945fdb7d1a87cb8932947 : SUCCESS in 3m 20s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/a8d7050e64d84ff3bb12fc407e71b555 : SUCCESS in 9m 14s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/a656a7498fd94abd9453d5f129c9fa6e : SUCCESS in 19m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/50004c61faaa48c494c81ed62c8ca393 : SUCCESS in 6m 18s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/ddd9d89ae0824837acaeab5949cd9228 : SUCCESS in 10m 15s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/85d2f020eec54478a859bad2f6bc7b6b : SUCCESS in 4m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7b2375d2fada4e6d94f2560566ab6266 : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0f2427a29fa64e198c0a26323c68c05e : SUCCESS in 18m 19s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/25dfadee5e69450abeef5fbae0d04727 : SUCCESS in 4m 37s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/9fe136b9d6b0400dbf5a380db0d047b2 : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2c854fd0c49e4f0a90bf2e2f289eb3a6 : SUCCESS in 6m 40s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/cbe274a50ade4e8e87513d4cd95146b5 : SUCCESS in 16m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b172fd60a444f7cbce00b03cc0e47e9 : SUCCESS in 59m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77d8b420ad7649a1aee620cf64de8f93 : SUCCESS in 1h 31m 44s","accounts_in_message":[],"_revision_number":1},{"id":"106efeb0db67acb71b5872cd9cdbc6c927d191d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 13:02:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe877b1f2eda612d21f795870b1858628bc142b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 14:28:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4d829b5ab814b62adfbe728355fa36a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f25a135badd48bda6919fd8a8f77d9b : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e02e5d2dac5c47edbf758aab4c0bc3e2 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a229edba83504d118693a23cda848055 : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8023e90afbaa40429f9d157a2c7bb346 : SUCCESS in 3m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6f7926f96d314b28a196c08ae7c8f8fb : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b75906d0c1e4465191e2153b14fbf06a : SUCCESS in 6m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ba672627fcd340558b03465039cf4e35 : SUCCESS in 5m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3b35c2d8bdb04cd5aa0cb0e34d1868d6 : SUCCESS in 14m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/99fb1ca729dc4d8892fb1afbb36e6de5 : SUCCESS in 8m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e4042a806da24cf7afed875e26c566f5 : SUCCESS in 15m 46s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/bd24d2675d014d2aa22f22eae3a31700 : SUCCESS in 10m 15s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e9d1301a19254308a56e9597d3bfc5d8 : SUCCESS in 4m 18s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/3b3373fd3f2346cba657425bdbd6d74b : SUCCESS in 4m 24s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/e29bf55506354162a7f17e5eb83111f0 : SUCCESS in 6m 39s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/b180bfbf097c4b47b5a348abb86f8ba6 : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/99719ecba89e421c9ef610b1abd152d5 : SUCCESS in 6m 30s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/cbb738bc513d4773aa43142940e45efd : SUCCESS in 8m 26s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/908d60aa68384b57a2494ecec1f99dc1 : SUCCESS in 11m 11s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/c4980b405dbf448cb58ad10838fd0074 : SUCCESS in 3m 35s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/47f5ea8e70c848f092e05caec9867f4f : SUCCESS in 8m 34s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/4582e33f539144f5b7c06aec537a2227 : SUCCESS in 18m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0fee5675feba4e50b008db3335c759a5 : SUCCESS in 6m 26s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/a846f441ffea434eb3afa3351ce59372 : SUCCESS in 10m 26s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/58f8e6ae11f040baa2ad59af36212485 : SUCCESS in 4m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/58d061705a3248d2ba49f46071fd8950 : SUCCESS in 2m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/42186191619845e78b3cc4f08ee8c67d : SUCCESS in 22m 46s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/d25dbbc0c5bb439b9f2b037c06eee66c : SUCCESS in 3m 32s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/4ea24d210d9a485eb64eaa318098c912 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/84359ba0298644d694a0f5590c8e3827 : SUCCESS in 5m 39s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/d842e934359c4d2bae121153bb0cbd08 : SUCCESS in 12m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/860b35aab3e94ad8b6d9116cce0dd3c5 : SUCCESS in 47m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8bcbd32419014327b89161221540ca3a : SUCCESS in 1h 25m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c93a756ab724984a40255599fcd85ecb92a9b128","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 14:28:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e2a17d12cca449be6b145fe5a6a8d32acb2e6adb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 14:29:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62e8059470f84245bd92f0ac8004421e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74cc9964c51e4dda861fd811c5655081 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b73ff32f098faef2d3c50175b37b928df25fc09","revisions":{"3b73ff32f098faef2d3c50175b37b928df25fc09":{"kind":"REWORK","_number":1,"created":"2024-10-03 10:52:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/62/931262/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/931262/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/931262/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/931262/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/931262/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/931262/1"}}},"commit":{"parents":[{"commit":"b4f18c29aa8cfc7a5cdfe5a95a51791858502379","subject":"Merge \"update constraint for ceilometer to new release 23.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b4f18c29aa8cfc7a5cdfe5a95a51791858502379"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-10-03 10:52:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-10-03 10:52:39.000000000","tz":0},"subject":"update constraint for os-traits to new release 3.2.0","message":"update constraint for os-traits to new release 3.2.0\n\nmeta: version: 3.2.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Balazs Gibizer \u003cgibi@redhat.com\u003e\nmeta: release:Commit: Balazs Gibizer \u003cgibi@redhat.com\u003e\nmeta: release:Change-Id: Ib833d387126fd31aea9239bded98cc6ff6452126\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Sylvain Bauza \u003csbauza@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Id8ad5dfadf7bfeb1ebd44d353d9b2ca1d1219d30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b73ff32f098faef2d3c50175b37b928df25fc09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b73ff32f098faef2d3c50175b37b928df25fc09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
