)]}'
{"id":"openstack%2Frequirements~932687","triplet_id":"openstack%2Frequirements~master~Ie408cd552101f68c6d29e0ed39bf2953eaa64615","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-11-01 08:49:06.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"Ie408cd552101f68c6d29e0ed39bf2953eaa64615","subject":"update constraint for openstacksdk to new release 4.1.0","status":"MERGED","created":"2024-10-18 10:30:59.000000000","updated":"2024-11-01 09:53:49.000000000","submitted":"2024-11-01 09:52:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932687-new-release","meta_rev_id":"7082406f95f9b9bc365a2a55a66b31d4375ccff4","_number":932687,"virtual_id_number":932687,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-01 09:52:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-11-01 08:49:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-11-01 08:49:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-18 12:40:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-25 07:37:40.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-10-28 10:00:29.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-10-31 12:56:10.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"CC"}],"messages":[{"id":"a023c82aeb702420a279c20f4cea907ad8ffeee1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-10-18 10:30:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94e992c02c5030c3d46f7e86308d4bf3cc7c5648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 12:40:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aeff71706204438282f3ba646f097d31\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0c94fafefb74606ab2ba2d14f7a54ec : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3e7e44867ff400498ac706a87eb8479 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed457ed534bc4e6db129c917d10c5141 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/54cdc18891b9412ebccf9edf704c4445 : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/02ce093e00b345e791804c333f910c95 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f53b107328e34c18b90d56e4e83edcf4 : SUCCESS in 3m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bfacffa05ce4414f80f48ab062b651b0 : FAILURE in 3m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7632462a86b5416caef5440f28a99f14 : SUCCESS in 15m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/eae64457d4e4404c8b5214a3948412ef : SUCCESS in 8m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7cf2574a29df47878a4e973a63d03858 : SUCCESS in 16m 38s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/160c93aca76e455286ff93aecc37b72e : SUCCESS in 28m 29s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/fd3a8cdaebc945ea9314d73dbd851275 : SUCCESS in 4m 48s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/e103eff0d9d34c958222723ab23f7477 : SUCCESS in 4m 39s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/e94a44853bf841f6950683b49f6a3143 : SUCCESS in 3m 40s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/af69a12e485347c0820da2c381643565 : SUCCESS in 10m 08s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/44e81d4477c5420b92928b83edcd420a : SUCCESS in 4m 46s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/570d7e8a0d5d45f7a8a7371132861782 : SUCCESS in 4m 56s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/1d0451946c8c42b28e171f0ae20d7cb5 : SUCCESS in 6m 41s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/2d204b3f9267495e8b22c5f04032d2b2 : SUCCESS in 12m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bbdc6932275c4159b0e8d1fb8e4e8071 : SUCCESS in 6m 26s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/3155ff44ccf04700af25a551a8a873a8 : SUCCESS in 7m 08s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/f603760aee264e92a75d1a698fc37506 : SUCCESS in 11m 14s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/cfbd50a5669449719d7afa1dc0933d40 : SUCCESS in 2m 38s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/965ba791bfeb470e9e6b12482052b6ee : SUCCESS in 9m 04s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/b73a23674d334b689a77c8068e081de3 : SUCCESS in 3m 56s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/7d27747a3bc345e8897a0bc4c547a2d1 : SUCCESS in 4m 24s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/f5f97467ff4c47c0be10ec91b4bb6ecb : SUCCESS in 11m 44s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/016c6d78fdf5487195dfebbe6188dbcd : SUCCESS in 20m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ea4fa4a8898d4ed7b115b5266e3de3da : SUCCESS in 5m 52s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/4e45eddac7bf4397ab840d8975647d3a : SUCCESS in 11m 05s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/8fe187b257234c0ebf1075690f74d12f : SUCCESS in 4m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/46af81aef3b1434e8f84e7b16e085128 : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82d51e95784f41df987dda090eb0857c : SUCCESS in 23m 50s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/5aaf7ea3bcc647988ca260f91d21cc39 : SUCCESS in 4m 35s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c35c2242f03043568f2a60be825be121 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/182dab866bd54a4aace5578289cbef54 : SUCCESS in 5m 50s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/1b3478e739d7454a8a9fd87b13be4c46 : SUCCESS in 21m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93afa70d0cf845f6b32a6b9cf07d1f20 : SUCCESS in 1h 00m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83cfe3b0e6234546b1d90938938add67 : SUCCESS in 1h 57m 17s","accounts_in_message":[],"_revision_number":1},{"id":"1d58dfb13cc9ffa2a326465dff4cfd4b24cdcca0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-25 07:37:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0faa5f2fa0a63344d4b7b7876e25f526e9bc237","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 09:21:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ae6a332f66c4a12993757ce5b4b118a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0148643766e4390a13ce8db96259426 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32b1b41eb1fb4c48b6d25a5166c46b8d : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db0f3470240b4406994e562d0a84548e : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/75d08dce1ca541f2800ea08082faedcb : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6bca58c0037444e4b6b977aa0519a1bf : SUCCESS in 2m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9f790b8f7d234656b24e3138ec1b2a85 : SUCCESS in 2m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0fcb5c11a0b6416382c8780991874b42 : SUCCESS in 14m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/50f964bbe6144ea99b270ed9abd0bb88 : SUCCESS in 7m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a3015aa0204d4dbf9fd3764a6f400e51 : SUCCESS in 16m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c4c9e8f654eb48f4a05868a8b8e9ca1e : SUCCESS in 23m 06s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/9e860632ab4343e1b8a61c9b9a0316b3 : SUCCESS in 4m 30s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/5d27f0260a0346c3b5669ce413a2a28f : SUCCESS in 4m 58s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/479d3502f5fc4f04b133e47ca8615c90 : SUCCESS in 3m 52s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/d7efd79e311e423d8f773acdb804fe58 : SUCCESS in 10m 50s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e68cf7f406b74ee6ab724daa7589314a : SUCCESS in 4m 48s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/19a9ecd75c47444898a86ba873a740ae : SUCCESS in 5m 00s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/3ae61b11f777484cbfd0540088f160dc : SUCCESS in 6m 05s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/af0a7ab84f0f4421a8af34a014d8abb5 : SUCCESS in 13m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3e8c0b30687c46df8e0d50600ad252d0 : SUCCESS in 4m 33s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/8cd750a161e34e67b7852c8b2687ad46 : SUCCESS in 6m 53s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/5a1013dfa1c2469c89119aca3d532b91 : SUCCESS in 10m 25s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/cb4db62419d1405aaaed2e57ebb08087 : SUCCESS in 2m 20s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/c3c519e3bf1240b0a427998f51b6fa8b : SUCCESS in 8m 38s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/32e55b20a57d49ab82d1a13926f9140d : SUCCESS in 3m 35s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/1c71c6932566408db9df50e7a013b244 : SUCCESS in 4m 40s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/b2c5dd7ab0f546f1881cdfc82aa025dc : SUCCESS in 8m 52s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/a6f857ea85f14277bb9f0b396888db25 : SUCCESS in 22m 10s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1b14ac8e17044869a168baeeaef3962d : SUCCESS in 4m 54s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/1781c70b51354b0aa1ff8c29e4da299c : SUCCESS in 10m 18s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/dacb05c8a21c4e6bb44d54c3802ded16 : SUCCESS in 3m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2220c59fde9b487e9c59649cedeff3f0 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a2407aa740ea4f8e903e4ee8c11c0ab4 : SUCCESS in 20m 53s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/0dc4b281e92b41389f3ab93d4baee3af : SUCCESS in 4m 15s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/ce808c034ada48d199f1812ffcda6381 : SUCCESS in 4m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fe0bfda717434e15b88aba497b8dbc9f : SUCCESS in 5m 29s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/c6c265d9697549f9a8b71f2a231c85c0 : SUCCESS in 21m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6001aeba8224cfca6607905f96a7396 : FAILURE in 37m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/23164e7503e647559f1e9d178e5a020e : SUCCESS in 1h 37m 57s","accounts_in_message":[],"_revision_number":1},{"id":"5f9b0934aff868d6c7b1a16bcfa949a72ac6ab35","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-25 11:45:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b804361427ac859a113a8cea4dd845b69f15a95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 13:15:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de763dbf0625435d8b944dde170b8a63\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c9ce01c3a7449bbbc42c5c341a2f75f : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13b37b3a00c745dca9e2c50c64b62108 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ed82e07776a4f2daf4e01c03260f9aa : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb4951e6b8aa4c99861f50b3e92e9cfb : SUCCESS in 3m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a3e8e7a29494342a452acbb4b36f506 : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1615908649264387a71f9e7a2c3f1e5a : SUCCESS in 2m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e708ecce8a514c9495e729ad50ed1a01 : SUCCESS in 12m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/254b39a094034d1db30941d7598d7631 : SUCCESS in 7m 52s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0be77fefe08245f7bff69e141ad5bb68 : SUCCESS in 13m 34s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/cf5cba5810554130bb06d781859e4d90 : SUCCESS in 31m 08s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/fc266a65ff8b4944972d613f326bd991 : SUCCESS in 4m 03s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/f68fdd52884d4d9ea80ee1ba6ee9bf85 : SUCCESS in 3m 46s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/57818bc7f6354cbb95c5fd54f27d4c1c : SUCCESS in 5m 04s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/c090ca24a1e4427197afba730fe0ee3a : SUCCESS in 8m 55s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/aa89c9950b43499f9e54bcabb201b7bf : SUCCESS in 3m 31s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/c9bde21a4b834c5f9afa1281cf7da202 : SUCCESS in 4m 46s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/dab778be9ac5416c92ed8eb363db994d : SUCCESS in 8m 37s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/5d30d328fddd4ba191098c08a53d96e9 : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f962ed3c19d94c00b0f48a8a840befd2 : SUCCESS in 7m 23s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/ecee3b525e0b42bf82878f48c32f43d3 : SUCCESS in 6m 11s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/9536ba1d0adf4f33886791d60ac5143f : SUCCESS in 11m 46s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/105385af49ae4d8d8f1c9135aa80c571 : SUCCESS in 2m 05s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b5dac860d82e4f4a84492c09dc895808 : SUCCESS in 7m 54s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/ad43557badb14d6ca042cf6ae00c30b0 : SUCCESS in 3m 55s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/33a7972cc8a84b59ab0f256ec6d1df0e : SUCCESS in 4m 08s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/8ed6dd2731e547d1914a4e48145b5df2 : SUCCESS in 10m 40s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/6a64a848f32e4bb798c12c9af88e83f5 : SUCCESS in 18m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/722cda3b867d4e8daca95a0c252be501 : SUCCESS in 5m 48s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/cff3be56c4af4b4ba8aabf2e829f798e : SUCCESS in 10m 19s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/245f1fd8abfe4f8da5109b5f74e804fb : SUCCESS in 4m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c743bb7ff38143d8b2fc698c4e599884 : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dcd9cb31323744c082e21bfa2bf95a19 : SUCCESS in 22m 47s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/299688c6dc9b4f9db040fb797a7f1ea3 : SUCCESS in 4m 37s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/5ce4dece501b4bd8aa88c804507ab227 : SUCCESS in 4m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e863d3347204415b8054748a46cc08bf : SUCCESS in 5m 28s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/78e0f6f28fe94ecabebf3843d32c05c1 : SUCCESS in 12m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96fb125edb374899b893427a0a860a98 : SUCCESS in 46m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d42678d646c14fbe8646f864c5073722 : SUCCESS in 1h 27m 34s","accounts_in_message":[],"_revision_number":1},{"id":"87cc089cca24881a0bccc9918135043c02dbd69b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-28 10:00:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e675c997cbcc93c2d5e4ab45bb14f07d61731721","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 10:01:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dfbcb59ffdf31180eee38a6eb44ddbcaea1b22c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 11:36:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a63ba7d6531b465e8d26ec60290a8904\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff12bb854fbb45599f490fd57fc7f363 : SUCCESS in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a4b7050a9ef46e1a351cc624ad94e4d : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ae67dcc53174db1aa374e29da424599 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c01e4bdf2cc94d289382deb7bd0e34fb : SUCCESS in 3m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8148f07f4f4949b4b8bc135d282b454f : SUCCESS in 2m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0a83e4e4859342908a6a477b337c3136 : SUCCESS in 2m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e8364f6c8c634d4fa88537f20ebfbead : SUCCESS in 14m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cd3308e59e714c3a85cec02543b634ed : SUCCESS in 11m 04s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/47cc7ec76f0a40f6848fe8230040b5f5 : SUCCESS in 17m 11s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/414656a455814572b26b0d41356f4903 : SUCCESS in 10m 50s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/862c280a18af4cb9924a6f2827bc109a : SUCCESS in 4m 30s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/dae789ce59b142ac82ebfed00e5e53c4 : SUCCESS in 5m 03s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/0b6235f1dc2e4f87b2efc99b80c72a5d : SUCCESS in 7m 40s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/97977ff826874ca19d771753cc88564c : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1f49ecaa087b47a197281eea670e6be4 : SUCCESS in 6m 39s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/fadd41420b3a483297ee61c2208b9114 : SUCCESS in 8m 50s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/49f505db179b4b6e828c19d4bf84088a : SUCCESS in 11m 25s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/16088a48f67f4ad58b4cae9007aaa3b7 : SUCCESS in 3m 44s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/bffb4d22113a4bbc802ca941872764d2 : SUCCESS in 8m 01s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/0c94cd2f2e894f209e10446edc935bea : SUCCESS in 19m 17s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c7f2af01b1d84865904a973d0f478947 : SUCCESS in 6m 25s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/6bd23b1285b54ff394bf79981caf6037 : SUCCESS in 9m 52s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/fe1b77e322954a81af563e37054ec2d8 : SUCCESS in 4m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dcaa0a65c0ab4f668f26bb89ee61926c : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b5834405037b4a3599173d546a4b1f73 : SUCCESS in 22m 52s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/95aad84296b1486fbbb5c8bcbf67e2f7 : SUCCESS in 5m 00s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/f1335c77e00741b4a804df1393e8ca71 : SUCCESS in 4m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/56177b277d7e45a9a6002aafdd33c3bd : SUCCESS in 4m 44s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/cb5682cf12f241a09a8d87734c916571 : SUCCESS in 9m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9057e3b551f1494d846df8b209d4ac4c : FAILURE in 38m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e0a7098bc83437f892cb934bb74080a : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d0eaf67a745cfd4f0dfb0fc7a9964627c4c9178f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-31 04:38:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf5a2df7e5f0205f7bd39b04953f3c75bdf87003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 06:06:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/10299cbe51d046739662a58e714c4dac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24374f8bc2024c3090ebae13eb2a5e41 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a8d63c94aff490d8a10ff442d3b1e78 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0b4b5d9fb9449f58881c942ab48e99d : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c611de06cf5642d8b4007f98034b36a1 : SUCCESS in 2m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/711fbb939a184401a040cefd72c4d455 : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/61f62e9fcf0e49d6beb7c8deb19f2011 : SUCCESS in 2m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bee21d81af6945e19d861ea56550c19d : SUCCESS in 9m 55s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c42edd0e2d1c4cc4865f5c087b395a25 : SUCCESS in 8m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/484ea024a3f840939171103acff82727 : SUCCESS in 13m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/66fd14747a2a45b18ceb148b3fa2e49d : SUCCESS in 24m 45s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/2134a485583440829aa0438ae19d8eb4 : SUCCESS in 4m 55s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/2d19c4f0f8cc4450870fcfb0fe28a682 : SUCCESS in 3m 31s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/bf35c4792aca419c8fa83c9d518dc25f : SUCCESS in 3m 58s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/8103400cbb7d4584bda11e6ead0b9b1b : SUCCESS in 9m 13s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/8d779e1c08284302aab2d562a80e5936 : SUCCESS in 4m 03s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/637fda1dc2d6475d9d7058b90dee5fa1 : SUCCESS in 4m 11s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/6f2245ef2a3f445f8c1f95f7d9565e6e : SUCCESS in 5m 28s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/e820042e5cea4abb809a55492dea08a5 : SUCCESS in 9m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/809ceed1a89e48cda19bb70e0a471ed3 : SUCCESS in 6m 38s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/8617f22eded44c4cb91e7d12fc35d0f1 : SUCCESS in 6m 47s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/5422bcb4a88449e99cf74fd48a17d808 : SUCCESS in 9m 03s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/1e4011f38df64c14bfbf989bf5420a63 : SUCCESS in 2m 05s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/f0f3a3eeb49b49ab9d42332229f877e7 : SUCCESS in 7m 27s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/b1bfb6afb0434dbcbc50836300a42eff : SUCCESS in 4m 10s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/3077f64b315d453cb7008690c1b48067 : SUCCESS in 3m 53s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/c5781f917edb4a34b555ebfc7f864799 : SUCCESS in 11m 12s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/2e90bfb79fef4561b51237b84d6fdaf8 : SUCCESS in 14m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0d7c63616fe74202bf5ad77a22dd4ee6 : SUCCESS in 6m 05s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/d0dd5a27399946b6b8746129c955470b : SUCCESS in 11m 31s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/aad1e6b476824ee88d88d487b3500d87 : SUCCESS in 4m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/35fc152d2b8345d1a10adb0c360f82df : SUCCESS in 2m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a8bc5851f1af4f47bf4fc99cc1bda7e6 : SUCCESS in 23m 06s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/9cc86259725544e89480771d4de4b4c8 : SUCCESS in 4m 30s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/5f5f1a3ad3e14eb5b3e4765d81ae51d7 : SUCCESS in 4m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5e8b5d32e1df4d1aa2350e6d6b38402e : SUCCESS in 6m 06s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/0fc8e07d9aef4161abf880c3beb2a23c : SUCCESS in 21m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c06d2da683241598842fa7eae01eeb4 : FAILURE in 38m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55165665f040401288c86c0255547e5e : FAILURE in 1h 20m 05s","accounts_in_message":[],"_revision_number":1},{"id":"14a300759690c4f3f119980fade43ae6cbfb31a5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-31 11:27:54.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0516a6f2b014c4d9240977013f6ff2c594654555","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-31 11:29:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"084a17ee2ae9970b2d5bc3b7849ebd8b0b6db73d","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-10-31 12:56:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2189de9b80c5c72d5a4c063ed20f9a8389f99ed3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 13:18:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33cf14449c8241d28dd19e4aacf0a432\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c4f0566b0fe4dccad8b122e88187574 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bb8a17e7d8142659ec1c1b953577443 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7d41acea35e4eecb55f76c341ea4aa9 : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd9b05094ac440fd8d92b0608f8a13a5 : SUCCESS in 2m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/21810ca7c01846219a3e8538b124474e : SUCCESS in 2m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e35654d2cf824183b55f3c8f1350fb8d : SUCCESS in 2m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/17ff761fae1f4760ab5d811872d1497f : SUCCESS in 15m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a579eb571c6d47cebd2e3c96fe76a257 : SUCCESS in 6m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ccdcb6e5377142c0b70e0c5ea4f065d5 : SUCCESS in 16m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2efd8a3b951d42b796bbe345c5040d68 : SUCCESS in 22m 08s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/141b88179c7a4d7cbe14fd9ede134197 : SUCCESS in 4m 52s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/e3e39189a6364898b21bc75b240c8978 : SUCCESS in 4m 37s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/40cbbdac1320426f8dc0ee7adaf8fe18 : SUCCESS in 5m 17s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/e99826c239dc4606ba53ad8cbbf4ee4d : SUCCESS in 10m 17s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e54bd58dcc6840f79ad4221674530f52 : SUCCESS in 5m 59s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/52baefd255364f26a2df22300cb42795 : SUCCESS in 4m 15s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/90f2a013c28244fba0fc1c714af53477 : SUCCESS in 6m 58s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/92e5f01cbaf042b8add52cc45b268b70 : SUCCESS in 13m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fa98e2e8ec2d417d8c047ab7b9bd4d2e : SUCCESS in 7m 13s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/d231dda22cd2460ebd68b4424e5855b8 : SUCCESS in 7m 48s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/2c6bdf1467a34caeb36ca56f350330f8 : SUCCESS in 12m 11s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/d13df8bec651414bbdb4e2b40dc73e63 : SUCCESS in 1m 54s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/45eccc1d74be4ebda7dee0aaae11f276 : SUCCESS in 7m 08s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/dff3bcd5184e49bcb6857638270e4cc9 : SUCCESS in 3m 04s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/6bc18a4041d14fcc949639ce4bb072e4 : SUCCESS in 3m 28s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/06d2b7e97c2b4e61bef9f089a305046e : SUCCESS in 11m 36s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/310a5a7257e7432d9541ca46bf0dcaec : SUCCESS in 18m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/85acf8fa7c5a4d139d777a31bb201720 : SUCCESS in 6m 53s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/bfa7ffdb3f054af687daf01edd60eefe : SUCCESS in 10m 43s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/1d7600cfcd7a480c92e054c522481227 : SUCCESS in 4m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6081ee45032846c49ff3b86965f28bc1 : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/54aede0ec90e4365a22739f38b4dae14 : SUCCESS in 21m 56s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/d83ee956c9d24082bcea9e866fedeae8 : SUCCESS in 4m 09s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/834b0cefcec94f0db48d820a75e5ff24 : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/23e7899587544b65aa78ad7e47603b60 : SUCCESS in 5m 21s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/037f49193be94c0f9ce4527a8884126f : SUCCESS in 13m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8edec3a20344bbf91a384683a7ecf67 : SUCCESS in 1h 02m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2bca88e3f0144bcb75fb17f2e6807e5 : SUCCESS in 1h 31m 33s","accounts_in_message":[],"_revision_number":1},{"id":"494a00ced8412bd55aa90746ea1b8ef7e29d5738","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-01 08:49:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fdded025d08534de8b59810b3429c4d622cfc0d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 08:49:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cd6a3a651db0c8baa2b18dd79fd0e7ff8fd574fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 09:52:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3ca50cd2ea04269b2fdc5fbc9672341\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/577a37cf6eb346d18a9c2ce3d7a6e786 : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb3fe7833cf24d3db0ff02a07cc9bc37 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6001152415f64fd1970326e7fce7e171 : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9fb55b6f3f0948e8914566eb7b9100c3 : SUCCESS in 3m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bdc70ef474904d6cbb07860aad347804 : SUCCESS in 2m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7df302625ba641ccb84546b7619d269f : SUCCESS in 2m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d3af642451aa4a528d4393257c7fc8e1 : SUCCESS in 8m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ce5fe7e762374c4883b13fd10876faf8 : SUCCESS in 9m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0401d953adf6414bafec479e8b99ecb9 : SUCCESS in 9m 14s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/81aed13301cf44d6afa5da619acb2d77 : SUCCESS in 10m 15s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/05158672815c451fb2fc6f23585fb6fb : SUCCESS in 4m 09s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/905928a9b9c342b9992e45ac8cb4f18f : SUCCESS in 4m 26s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/dc4d7e8b0ea247699e4666ee1fb75c20 : SUCCESS in 6m 52s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/6cafcf432bbd4701a2703738f7dec85c : SUCCESS in 14m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/67268c7f177f4629a38a8d3c8e7f0876 : SUCCESS in 4m 35s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/dc63a914660540d38a6656ecb7f65920 : SUCCESS in 9m 42s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/373842b414b447a2b2d1539fccf135e9 : SUCCESS in 10m 45s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/985f0540164c4b2c89a7619f1c57d355 : SUCCESS in 2m 09s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/d1975503e4694bfe968cb2cc2b041455 : SUCCESS in 6m 25s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/d2a7bd1191464a75bf12eb872d632ef3 : SUCCESS in 14m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d720561705804238bfd3b85cf4d22a1a : SUCCESS in 4m 36s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/7bf0791948a6406aa2daaf368bb6fbe4 : SUCCESS in 8m 47s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/eae0e87b47cf49e89ea05a650be22b40 : SUCCESS in 3m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5ead9040d92b44afb59cac226758261c : SUCCESS in 2m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bdf54e0623084b46995aefcc27d918d8 : SUCCESS in 27m 15s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/6adcf1946eb24dbebb2b106d33a44f4f : SUCCESS in 4m 35s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/de882e9c85e844a89cc04f723e8a23b5 : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8c24c9b9e3644173aef100296099fd24 : SUCCESS in 5m 07s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/95d2629f306940daa1f79d3e459af2fa : SUCCESS in 12m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bfbb2b18796e4fffb1e3782490823932 : SUCCESS in 48m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8bcccd14d8b4413dbb6c51ab895d6d24 : SUCCESS in 1h 02m 21s","accounts_in_message":[],"_revision_number":1},{"id":"97f32a99905e39ebbda94f7e6d78af54554995b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 09:52:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7082406f95f9b9bc365a2a55a66b31d4375ccff4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 09:53:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e6fc8e390d347cc9349711dd8704e8a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b8f93c82d4d4635815a2cad90d6b8f0 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f62a144e76fb41cacb0166220fb29e2dfddca94d","revisions":{"f62a144e76fb41cacb0166220fb29e2dfddca94d":{"kind":"REWORK","_number":1,"created":"2024-10-18 10:30:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/932687/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/932687/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/932687/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/932687/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/932687/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/932687/1"}}},"commit":{"parents":[{"commit":"3b73ff32f098faef2d3c50175b37b928df25fc09","subject":"update constraint for os-traits to new release 3.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b73ff32f098faef2d3c50175b37b928df25fc09"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-10-18 10:30:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-10-18 10:30:58.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 4.1.0","message":"update constraint for openstacksdk to new release 4.1.0\n\nmeta: version: 4.1.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Commit: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:6ae5d3bef pre-commit: Bump versions\nmeta: release:034a68b5a pre-commit: Bump mypy\nmeta: release:187e88a55 pre-commit: Migrate from black to ruff format\nmeta: release:da0667e20 pre-commit: Migrate from flake8 to ruff\nmeta: release:904e361c1 pre-commit: Bump versions\nmeta: release:Change-Id: Ie698e091500949d7664e0005403fe54f35e99323\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ie408cd552101f68c6d29e0ed39bf2953eaa64615\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f62a144e76fb41cacb0166220fb29e2dfddca94d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f62a144e76fb41cacb0166220fb29e2dfddca94d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
