)]}'
{"id":"openstack%2Frequirements~932690","triplet_id":"openstack%2Frequirements~stable%2F2024.2~I01169b41a7f71249e64aae9202903764aec48f1b","project":"openstack/requirements","branch":"stable/2024.2","topic":"new-release","hashtags":[],"change_id":"I01169b41a7f71249e64aae9202903764aec48f1b","subject":"update constraint for oslo.messaging to new release 14.9.1","status":"MERGED","created":"2024-10-18 10:35:51.000000000","updated":"2024-10-25 09:14:51.000000000","submitted":"2024-10-25 09:14:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932690-new-release","meta_rev_id":"e37c2c000685042afad1e7c9db0321c77f2cc090","_number":932690,"virtual_id_number":932690,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-25 09:14:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-10-25 07:23:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-10-25 07:23:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-18 11:59:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-23 17:06:26.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-10-25 07:23:35.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"cf39e2083cd22b0f0d8acce794706f59b47df093","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-10-18 10:35:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3cb86f2314ed2ff6d5e8aff43f38dbd370109f8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 11:59:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60dce47ba48f4f83975acc8e313ef5e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d66d9497881843508d45de84e0ed0620 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb078fa2bee6481d893f82384fdc2385 : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/12d525f070414ed5b5dd966f4c16130e : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/199ef9aa2aac4c9d8554e68ae4d7caa3 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/001cecdd7f264bec95032c234f48782f : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/531de9c0a71f4e748c90f762d24f90aa : SUCCESS in 4m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e9f6e5a7a9fd47bca2cfe4534cc091bd : SUCCESS in 8m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ae5761b78889448b8202c3b21c47feb9 : SUCCESS in 14m 56s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a2f2413708f44ff7a3bbe4fb035578c3 : SUCCESS in 12m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/496572e65fff4378b5e766233f2319e6 : SUCCESS in 17m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/95a2528dc3314e94bf01daf3c9620f3b : SUCCESS in 21m 52s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/c779365a6f41428ea8719409006ac5b9 : SUCCESS in 4m 55s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/1b99610bca704cb689e79521690f231e : SUCCESS in 4m 37s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/8815efcd03394ac6a0a8ad4a1cbbf645 : SUCCESS in 4m 27s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/1809ea4ee0ee4c1096d3644db4070732 : SUCCESS in 9m 56s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/16ea3e21887c436bbf0fd4bc8c35e9cf : SUCCESS in 4m 05s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/4fe767eb58574ab8b3afcc12c44faab4 : SUCCESS in 5m 39s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/53e582c6483445dfb69a1a9be2ab1396 : SUCCESS in 5m 38s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/0fa183479d2e4618871b9f58f8b75375 : SUCCESS in 12m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/213982b57078400e9e142d99c5f663f4 : SUCCESS in 6m 12s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/657b5597516e4e56b086abff714b9427 : SUCCESS in 8m 47s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/a7bf293fc667481db2123e8a0130f5c1 : SUCCESS in 10m 04s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/a4072353754a4eaca8fadf15fe084f25 : SUCCESS in 2m 10s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/b0cd2a28f8e5499897ee2d713eac5b78 : SUCCESS in 7m 25s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/04619d6b1b1a4955a4494c120328a5b3 : SUCCESS in 3m 47s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/ea64a115a4894e74966b81c14938d9e6 : SUCCESS in 4m 08s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/66e298e8f2a94cd4a4413af41f610d3d : SUCCESS in 12m 23s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/1ac96004003242e88fa996b68b42d84d : SUCCESS in 20m 40s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0868623c6a9f4c32bee073487202d242 : SUCCESS in 6m 29s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/87ccdf812db14e13af77a8187052ec9c : SUCCESS in 10m 44s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/00324a4194f347bfb6c8b33df985c0e2 : SUCCESS in 3m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/abf328aa21994afaba3b3dd82fd71598 : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a41de876482441e5822bb545220814cc : SUCCESS in 22m 41s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/dc067abfebd148139ad52d2257daafe7 : SUCCESS in 4m 13s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/8550ef2c4acd48a4894eed6d1156c6e4 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ceda0eae2dd742609186b58b6b747382 : SUCCESS in 6m 25s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/29095a9fd1ae4967af67928af601e649 : SUCCESS in 12m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e15ceee9b714131bf98380e9c2a7d1c : SUCCESS in 1h 02m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/554d4b74363342ea89c7b3e2fddceb05 : SUCCESS in 1h 23m 06s","accounts_in_message":[],"_revision_number":1},{"id":"878d60f83ac427c82fa5fd3fd4db2a4edc0b982e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-23 17:06:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f68495c3386631eab00e497f530ef66401a53735","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 18:41:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85d69be2d83e429eab8bdb862453a7b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c4805d3564248ce9c48c07268d65268 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35c663326b464800ae91af76b95c2364 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/862438d976dc4a1e9d6c489308c951e7 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c7eb0554ec74f9bbf9e148278faa5a6 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94db519ff34c43daa63f088a17f4ecd4 : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/32220329b7cd42a79222a67175a43bc4 : SUCCESS in 2m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0d720beb1cdb473896afe088565df3e1 : SUCCESS in 6m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/93e09de675884ed69e283929a16c848e : SUCCESS in 14m 53s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/aeffad78d1ab42f59e06982bdd53cc9d : SUCCESS in 8m 07s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8b13a6c0249149cf95f485323ec6976e : SUCCESS in 17m 46s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bca76d67fc1e43dd91e2ff0bcf507d96 : SUCCESS in 17m 41s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/2261b00a0e0d48a7976c82110004f755 : SUCCESS in 4m 30s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/15b4c151eefd45ccb7e74a05222b3bbb : SUCCESS in 4m 55s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/66bca1dd495e4712b13b246db9be3eb3 : SUCCESS in 3m 19s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/32a966cd65e84219b7e6e48e04cdf0ac : SUCCESS in 12m 06s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/d751ff890c93413baf43ea43303ae259 : SUCCESS in 4m 55s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/bd7d22b97ff54532bb45d7c9abe02f9b : SUCCESS in 3m 49s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/2fe456f92aa94db59b41612e90ab09e8 : SUCCESS in 7m 00s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/e7e1b7c5953b4cc790411dc83f5405ef : SUCCESS in 9m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5858189562a94f4984106746b20332c2 : SUCCESS in 6m 47s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/0a252dcb803e42689d3f84991e368eaf : SUCCESS in 8m 54s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/c79df5793e1b4d3abf42034b9455fc76 : SUCCESS in 10m 54s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/8c2b916813724ec08c7222978d93dbb6 : SUCCESS in 2m 14s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/1e888783f6714c20be03e3c3ecbf4608 : SUCCESS in 7m 22s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/a734e2ec2b6443e6b15f28405770c3d1 : SUCCESS in 3m 49s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/4f6dbbc1f8214da78ea205873dee2108 : SUCCESS in 6m 16s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/d798db62311a426c90f4284c4d1ed6b5 : SUCCESS in 10m 45s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/c0bb5d262c01493691d4a7acb8c495ec : SUCCESS in 18m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a1d1e753fc3d4d4e90178b71f1f6fcc7 : SUCCESS in 5m 53s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/e1edcbd067394ef7bfefb57c32ca3985 : SUCCESS in 10m 39s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/1103eb8b78a1498caf5b328c5185c556 : SUCCESS in 3m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/18de1c43ba7441cc9905d9b969333fde : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4b62ed2c2459445889fd9365d2373b98 : SUCCESS in 20m 57s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/2d57d0383d9f4081a44609bfa5f03193 : SUCCESS in 3m 46s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/a1abcceba2854b759ef57e7538d484b7 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bb839d2fd89e478ea39b2ad4a6369b5e : SUCCESS in 5m 54s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/2fe2c73030fb4749919ae782a72146ab : SUCCESS in 9m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3309430bfe064a0681b09e5975a6400e : SUCCESS in 59m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0eb5276b276e4e5083bf462add13fb27 : SUCCESS in 1h 34m 14s","accounts_in_message":[],"_revision_number":1},{"id":"e6aa9d84b86b936efa55b57d3982511b03935ec7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-25 07:23:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ccb1a3f511eca18ff2e57fa7daf659c5e9d0a5b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 07:24:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed4df89f411fb56738ec7eddc70b22e5799a5c43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 09:14:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a74f3a5234140e28188c2bda3343f7e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f9ca33644a7459b85d1ce39e2739466 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b050577ce09e44a0abb57e822c33d5c1 : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2e3d975916a4373a189d00f0cb53803 : SUCCESS in 3m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/974c263d185d4fbb95f31f50bd65c31c : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8840a97cc4d845cda5279fca5e561f98 : SUCCESS in 2m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/67258c0a17aa43c9bb8a39418d6aa5ab : SUCCESS in 7m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d1d0e7bd3b0f4e0cb8d74ac8f3b3b814 : SUCCESS in 14m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3af96be4515c4be4b187745af0b4359a : SUCCESS in 7m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/092d2417eb87475d82d04a66b8858749 : SUCCESS in 14m 53s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/d2d29bb53fa148a7be03d21a59c4c02c : SUCCESS in 9m 56s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/f02bec6796804f45a894741df169b017 : SUCCESS in 4m 26s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/44362445c0b14ac8960b5de197c5292c : SUCCESS in 5m 25s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/7e1c4de4b3874a5d9249b3d96a73ece5 : SUCCESS in 7m 00s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/e7112cb9f5ad48e191c74501094c1799 : SUCCESS in 13m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/43727d035f354fef9f1e7b8e6db8fb2b : SUCCESS in 6m 21s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/dc8a6202ee2c4ba38de990b828a46b7a : SUCCESS in 8m 46s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/a1bc4782e844420eae5f0747a5922ef4 : SUCCESS in 10m 21s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/37f7ce317b6b4456a3fa449963f1aa52 : SUCCESS in 2m 12s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/043bf77667654960b3364d5d7c06fcb9 : SUCCESS in 8m 32s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/c22e4f65195f44d2b27e9e9fd18bd3ee : SUCCESS in 20m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/64c7784e15cf4882b769083d1acc27c1 : SUCCESS in 6m 09s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/f3fd5247f77c418484f9fe1a871ec60a : SUCCESS in 11m 13s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/b2a72b315a9e40fb9f4381a09249e681 : SUCCESS in 4m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/31fa47de51a549fbb50b36d867717db7 : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/761f61a568864950bb9c40659f4814d9 : SUCCESS in 20m 31s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/0b978479d9be46e2909a144fac55c322 : SUCCESS in 4m 14s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/3aa2bb8f0ddf4d4b926430ed5280b91c : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0c23d80b74e842bfab3649990b3761a1 : SUCCESS in 5m 27s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/390e48884b8c422aaf971bdd9f18dd31 : SUCCESS in 22m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9df14c2795df43a389c05deb68782e11 : SUCCESS in 55m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6da7cde29654b578faca30253154bb1 : SUCCESS in 1h 37m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e37c2c000685042afad1e7c9db0321c77f2cc090","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 09:14:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"74e57b113887918497159843b3e5836ec9029280","revisions":{"74e57b113887918497159843b3e5836ec9029280":{"kind":"REWORK","_number":1,"created":"2024-10-18 10:35:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/932690/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/932690/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/932690/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/932690/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/932690/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/932690/1"}}},"commit":{"parents":[{"commit":"5756844c731f2a2674916281470ff45b125b7ff3","subject":"update constraint for python-openstackclient to new release 7.1.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5756844c731f2a2674916281470ff45b125b7ff3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-10-18 10:35:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-10-18 10:35:48.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 14.9.1","message":"update constraint for oslo.messaging to new release 14.9.1\n\nmeta: version: 14.9.1\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: dalmatian\nmeta: branch: stable/2024.2\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Commit: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Change-Id: I5fba60f19aea6df18cae0487d82dc459c56415d2\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I01169b41a7f71249e64aae9202903764aec48f1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/74e57b113887918497159843b3e5836ec9029280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/74e57b113887918497159843b3e5836ec9029280"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
