)]}'
{"id":"openstack%2Frequirements~933484","triplet_id":"openstack%2Frequirements~master~Iec4403045e59030e13d5b42dc1a878af35899892","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints-noclob","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-11-20 10:46:23.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"Iec4403045e59030e13d5b42dc1a878af35899892","subject":"Updated from generate-constraints","status":"MERGED","created":"2024-10-26 08:21:21.000000000","updated":"2024-11-23 08:47:08.000000000","submitted":"2024-11-23 08:46:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933484-openstack/requirements/constraints-noclob","meta_rev_id":"bb1ac1241429dc06ad159ac76bb85c0d5dc8a4f6","_number":933484,"virtual_id_number":933484,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-23 08:46:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-11-23 07:22:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-11-23 07:22:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-26 09:09:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-23 07:22:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"07dc56c4fed429ca808001b8675fe5c785147f14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-10-26 08:21:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"473e5174d3226647f1d8be825c570c92de9f5055","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-26 09:09:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f8eb67868454f0f96d2c44506189950\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07bb4188f71844dfa917342a268d448f : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23f4c067c6624891b652ca00b34c1991 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce68eb698bda4f23b75908a0129ca2f9 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6009fe02558340ce8db08408807c05d1 : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/582722743c624e499846487648d207cc : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5ec5e3d1e56e41d89ba11e2aed4009f9 : SUCCESS in 3m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/21cd964b0fd24565ba59eecea4c4ba13 : FAILURE in 3m 13s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/2ccdd566c6ca4b83a77d366929d43e7a : SUCCESS in 12m 34s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fc2995cad5e64191a876b9df36a22246 : SUCCESS in 16m 25s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/90c3432b47814296924b433ac3a19b2a : SUCCESS in 24m 01s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/29f496758c74460b86060d2c75baa396 : SUCCESS in 4m 34s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/a4c1a4d830a94f70b5c096e6f69bf280 : SUCCESS in 4m 28s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/bf7150922ea045af9ee3b54fbd52bb01 : SUCCESS in 4m 07s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/15bef1a9a0b0468b83871aece740a9ae : FAILURE in 10m 09s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/55d9bd98bbef4dd8addacd01c7a4ed04 : SUCCESS in 4m 33s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/54c5b13209e847c287ee8dedaee2b816 : SUCCESS in 4m 38s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/15fc8971903849689a2912747c9230e7 : TIMED_OUT in 40m 54s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/4f5d09bffacb42e789e5a7c9995ae5b9 : FAILURE in 7m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3aab5cca1bf14411b10ce00d35bde8c5 : SUCCESS in 6m 59s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/048e0a52ed1e42ca9ef4827775b0229f : FAILURE in 2m 46s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/22d2dd9e8888452789ac3c28afce6171 : SUCCESS in 9m 57s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/7c693e9139b14f1a94a74b4266e211a3 : SUCCESS in 1m 51s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/cdd5ac8ee19f4b1abcba017223ebff7a : SUCCESS in 8m 26s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/e76910e819964e00817c687cd099886b : SUCCESS in 2m 50s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/c8bda1c72e1b4f1f90d74029bb73835e : FAILURE in 4m 14s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/988f8dfec3a7403eaf09e6d817b4dc49 : FAILURE in 12m 35s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/e6954edfab9240c885904bbd1bc0f8d6 : FAILURE in 18m 39s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c7391c135d394be8ad4e2f69c92e2049 : SUCCESS in 6m 34s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/db5505b02f094ae79568044aaa67d624 : FAILURE in 18m 57s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/a9a9f69c3de34d15848254f1f4c39537 : SUCCESS in 3m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6fa01d61607347eda4db5c6e213d8df9 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7811d0d987324d0d990ef8d0fd3165ed : FAILURE in 24m 48s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/07abd5a58d704256925220a9707942d4 : FAILURE in 4m 24s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/e8f60bd1bee541579011c7a2d356bfba : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3a60278e0ffa4831bb251ba5f7441adf : SUCCESS in 5m 45s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/14c84e50b73f4b489558fb4390bb9a6d : SUCCESS in 12m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56086ff166bc49f6b0f51c575285b338 : SUCCESS in 46m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc8d370519174bb98b02f7e4d5d8e5a5 : FAILURE in 14m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c53ce22650964aea9b9670220721041038ab98a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-31 06:19:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"060fbfac6cd21198d2fa50d51e7463039dca550f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 07:25:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/303dd3567f0a4248ab879ce8b7a69f8c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43733833af994cc0bf7c4e6788edd32b : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e22dda84d8974719aa797b0d93c71d2e : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/247a41ae858b480387ec9397e2e7053a : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c8cbb9c05c154b5bb28f66ebf58033d1 : SUCCESS in 3m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/843866a6f8954115a8bb72d911297da2 : SUCCESS in 2m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a0f511fb07e043acb6ffe597104f5662 : SUCCESS in 3m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4ff087b1c1654b209a70f4a6636ce7cf : FAILURE in 2m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7923e1b500e641cbb33d99ccd5ef0b09 : SUCCESS in 11m 01s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c7e764dec0784f778e9438786a84ac43 : SUCCESS in 12m 45s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c1e79c232a2d44aabb66bd187350b8e9 : SUCCESS in 18m 34s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/09914bd93cf54e8283f5178922c32ac0 : SUCCESS in 4m 25s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/42f12bc9acc04faabbe2d9b028c7ac60 : SUCCESS in 4m 29s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/cd24b43c4a564c30b6353917f8e51651 : SUCCESS in 3m 48s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/3642096ccf5a4e77bf6b0f3c6a4f1f29 : FAILURE in 9m 12s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/1cd218e12c1648fea1f46e954398c126 : SUCCESS in 4m 49s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/08861b73e1a940b0bd3292420773d38e : SUCCESS in 4m 18s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/8902a3def8924991a94441c64452d93d : TIMED_OUT in 40m 45s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/9e378a7d13d84fcdb6af8e9232e72ba3 : FAILURE in 7m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a88a5cd47d7d4e039381a60f1d9d0598 : SUCCESS in 5m 58s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/f676d564fd224474a53ad372f160e5b2 : FAILURE in 4m 16s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/24d41ce383f7470c82b9e3b985ff0e75 : SUCCESS in 11m 42s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/618bb5c9fc914934a7d394ebd4a4085b : SUCCESS in 2m 33s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/35f0314efbf842429bc0843a6bf22bb3 : SUCCESS in 8m 30s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/fb5f4b9c387048a7801f92400fee164e : SUCCESS in 3m 48s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/057e1186e6d24ea68df5a10ec72fdf8b : FAILURE in 4m 30s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/ef0612532895429e8e5e8ca94ae8e880 : FAILURE in 14m 50s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/05f4fc4f35284d1e8bbea29e99869567 : FAILURE in 19m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/d90195840ce047548550b3eb31881422 : SUCCESS in 5m 01s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/e55400e8b1064756b63e56dad9dc1333 : FAILURE in 30m 13s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/355e3a5cf52c40f394ae2ca15a053894 : SUCCESS in 3m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0a83fb58460e48fea39d695519e27811 : SUCCESS in 2m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/95297aacab444d2e8d192a6dba286653 : FAILURE in 27m 21s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/2cb55cead0c640caac53c415e3a23ab7 : FAILURE in 4m 40s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/896cbbbdbd6b4e8696048a65d7afa3d0 : SUCCESS in 4m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/496a048fef364f54861d2c5dd83fe0a2 : SUCCESS in 5m 27s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/af2ecb5b46aa449db819726fc25bfe28 : SUCCESS in 9m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b4f5d5b3b5a401abb98f0d13b74555f : SUCCESS in 59m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96bfd13af07a49029e8ebb6f6b3e43e2 : FAILURE in 16m 57s","accounts_in_message":[],"_revision_number":2},{"id":"e9be3881e74f40da52775284fbc602aa7dc5b9ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-11-02 08:12:26.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2d42f2e737b25eb3d7fe8fe9719b2096ce012594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-02 09:07:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d34c762b08f4fc3b45eb8ae07fdb3ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf27f63d69d46df883819a3ea645745 : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c11a1c1b84dd4c2a823795251b2c43f7 : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a74d740f39244225896803fe9ea9dc9e : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c9394b05c5bb479fa3cf0a54b1ad7e39 : SUCCESS in 2m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4874c24005884f9b82cae8f629f02ca5 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5746ecb045744699860705abde1571f8 : SUCCESS in 2m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a4abd53dd7514a20a5abef08b14bbed1 : FAILURE in 3m 06s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4cb06b44c1b8447ca49e034f89d56a07 : SUCCESS in 7m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/04d2dd3d8b274780a58ae0eaf5531ab6 : SUCCESS in 16m 31s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b0071ae692ac4867b0e90f608f5b5a8e : SUCCESS in 22m 50s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/6bbfadf62b594014ba0c5e57bd174426 : SUCCESS in 4m 45s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/d534ad51e78643c9a86e9226e0ca844f : SUCCESS in 3m 27s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/c58acf7dea014b17b8a3bc68e04b593b : SUCCESS in 3m 41s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/586add20b29f4ecf8d9b88a616237223 : FAILURE in 11m 38s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/35dd11d051c84cd9adc22815f6fefec5 : SUCCESS in 4m 33s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/c03443b07aea4a92afe9fee7b56ed1f6 : SUCCESS in 4m 41s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/b333ff84ce8e41499b13708498991509 : TIMED_OUT in 40m 54s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/c050060d716048efb111ce0ed3b91212 : FAILURE in 3m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1529aac1e553421bb4c1f8495161b423 : SUCCESS in 6m 10s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/523c8679f0b0425ebe0f018fd8009bd6 : FAILURE in 4m 36s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/388bf7bd13e74d418e409f42b8153997 : SUCCESS in 10m 20s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/b5daf4f0d45c449f9d2f66f13a1fa90f : SUCCESS in 2m 01s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/a43ef0c04da942c98a6fbf1811ebe76a : SUCCESS in 8m 06s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/186e03cc349446c3a54f5cc0d006d253 : SUCCESS in 4m 14s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/0a46ede7d6b842ff964b01867faab1ae : FAILURE in 4m 13s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/790310bc24a4422fa16af263b0b53bb7 : FAILURE in 14m 01s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/2531481b181d4819bf449a30d233da8a : FAILURE in 20m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ec4f4d89bba942b3beaa7a2c9938a174 : SUCCESS in 6m 04s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/1abe32b875bd402ebcbd2c52a3bdca5c : FAILURE in 13m 16s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/121cb70c3afa4a599359d92f3a0dd5fc : SUCCESS in 4m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7a2ad63324324aaab8d6aefe0897f4b8 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/520a851a2459404aae1c287b3d096053 : FAILURE in 23m 34s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/719475ab6c2c43e69d351194144b3bd3 : FAILURE in 4m 53s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/e0961b4e78bf47f797ed14ac377e8244 : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/53e7e1b3a1744b1aae69cbca15489b30 : SUCCESS in 6m 03s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/be16ab375ec54863aaac765fb8c7cc95 : SUCCESS in 14m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1c503a761854d84bbd76fb137bab9b9 : SUCCESS in 52m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9cb0741201e7478f9e22a2a478238c0d : FAILURE in 21m 02s","accounts_in_message":[],"_revision_number":3},{"id":"ad1783ce525591b59c7771989a1837bafa36f601","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-11-09 08:38:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"5af8a3395218098df50f54fe58b7f0340295d4ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-09 09:59:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6560a77e6f649b097a5d4280708ce1b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c80634258ebc4b7faeb4d5a0fc4b64c8 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26082be482f54874b556caaba283ebe5 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/480ef5e0a16347a5b5717bc007361be5 : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84ac330e410a411ca47bde3b5db4eadd : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/caec113eda584cdab258b512de38ee94 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/966da5756e9544efa38518b058f1be9b : SUCCESS in 2m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fd7119e6d6564d619458060a9f7d2c93 : FAILURE in 4m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5d3367f56a354857aa2dbc88280a1a35 : SUCCESS in 6m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d63c77b677b94fb2b1b74f98c28bfb50 : SUCCESS in 15m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a739e0b223924ff8a475be2611c797f6 : SUCCESS in 26m 00s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/14ffda1eb40d418a87c158789a3b6512 : SUCCESS in 3m 59s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/0f5a3489d2bd4cefa907ba4559476bfe : SUCCESS in 7m 01s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/7817c7d0756c470e9776123efa2e5448 : SUCCESS in 3m 15s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2e68d15e7b7748888387c932b78e3d75 : FAILURE in 12m 03s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/23813eb069794ccfb384948714f28ee0 : SUCCESS in 5m 17s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b0b02769dab64a748a8d9add06bcd45b : SUCCESS in 5m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0053063b429a495b98b100747cbced83 : TIMED_OUT in 40m 44s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1c3800499e76424ca17ddd6fe2fdfab7 : FAILURE in 6m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7747a9115fbf4f018ca0d4a2393334d0 : SUCCESS in 7m 02s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/79dae4a5618a4884ab90bd0da158d5ea : FAILURE in 5m 20s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/652a7b934ad04c2eaa43512bfd04557d : SUCCESS in 13m 16s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0f4731399cf248a29eb9a1f09b26cb7d : SUCCESS in 1m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ba52b530559e489e9a6ddc4e4abab003 : SUCCESS in 31m 00s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/a11c57c4dbd04b90938593088452d78f : SUCCESS in 4m 34s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/ab38c8c7062343d3a3bf125ab4f338a2 : FAILURE in 5m 57s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/b82c362e1ab84997abbaff8694a91e04 : FAILURE in 11m 40s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/bfe6f151fc2a4aa096cd34e7534b90c0 : FAILURE in 16m 56s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/89f28066dfa84ac59e0a64db15a1ca13 : SUCCESS in 7m 42s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/099cb70af3f7424ea399a6be27f819f9 : FAILURE in 16m 15s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/12226c389e77448d8586822acc70c15c : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/23849f47da0849be97a439018d1612c9 : FAILURE in 4m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/09b3f9498397431e9c7eb73b2870f673 : FAILURE in 23m 09s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/9214519a7cfc45b787bdc3b6ed2ddd14 : FAILURE in 6m 55s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f3f2938c8a004974949d0b0936e25b76 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2fce648412d24f34a47b461379c5a30f : SUCCESS in 6m 12s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/eb86527a4deb4f0ba7577a37d952789a : SUCCESS in 15m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad4ee2cb1237489fa0115102e2f5521f : SUCCESS in 1h 00m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2cb1f78019d3489ba5c7594ce5b90f42 : FAILURE in 21m 16s","accounts_in_message":[],"_revision_number":4},{"id":"71b61f9663cfcf5b2ac665a9eaa2998fcb889ec9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-11-16 08:18:49.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"123859b260ec3b33949469df9046a0b800ef2632","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 09:33:56.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9011af53dbaa43398c0642ca61c1f8e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29e5fb0d62f249feb2df3fbdaa4ebb54 : SUCCESS in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caaa85f839804dfcbcbf57c8e70bf90a : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9566eef4a33a4bb0b05bd6b86884729a : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c47d2d07b7fa4bac819f39d103c65be8 : SUCCESS in 2m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b439ae138ec14c8390871bcb94e2fd8f : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d1b88f3fe06d45cf9443462b8e598c82 : SUCCESS in 2m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ccce66e577f2476b9c16b12b2e816b9e : FAILURE in 2m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5c71fa94f6614d7e8f5617d767b5fd53 : SUCCESS in 9m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/90ff67660fea4b3993b3c3bb1a1629f4 : SUCCESS in 16m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/efd12b9dea074bd7bef84eb1becf4d40 : SUCCESS in 22m 04s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/d7524869c2484dbba691243b8c0c62c0 : SUCCESS in 5m 10s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/89114f06f8f749c39c5c84fd28f0803c : SUCCESS in 5m 43s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/bf044c436cc4479a88ff58684fb2988f : SUCCESS in 3m 12s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/61a5fbdc05cc4624a298a50c9a721c83 : SUCCESS in 13m 26s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b22dcd17ab2242f092434481b3aaea7e : SUCCESS in 4m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3a90a1e65caa424499d878360d371790 : SUCCESS in 5m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/c88450ea0095489385b372dde476b508 : TIMED_OUT in 40m 55s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/a8e8e3ed09f84abb9576af9c1645a133 : FAILURE in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8104dd9a7ef649fdb05e6f26f5eacacb : SUCCESS in 5m 34s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/3d699b4ea88c4f199a8851e6aa99ff6a : FAILURE in 4m 00s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/005d982b69d548dabc990c890abb438e : SUCCESS in 13m 36s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/05902f7172a44c2bbb20a850af5878da : SUCCESS in 2m 07s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ed01d375ec8d45249a08e4ce2e100fa4 : SUCCESS in 11m 35s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/415cb11ecf0343cd8dcd42325cf1d0fa : SUCCESS in 4m 23s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b86335116c3b4e87bbbd653534e0d276 : FAILURE in 4m 29s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/2d83761c6fbb4e85ad12a05125e20635 : FAILURE in 10m 12s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/cce828844215487e91e224e385f8135f : FAILURE in 16m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/c01f99d0c45143df8fdfd699e0fe0eab : SUCCESS in 6m 20s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b73a9001fb1849378fc9db46d8d42373 : FAILURE in 18m 03s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e15c22ae84e347beac499d6d428f259f : SUCCESS in 7m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/21fb04bad19a44fe916af1d71d09aecb : FAILURE in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0456d81ac2fe41f3918568c16ccb8705 : FAILURE in 21m 16s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a8428f881cc24a0e8aa3ca3efb9c4132 : FAILURE in 8m 17s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a92e12808b3f41919981467e6a338041 : SUCCESS in 8m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a37b340129c84b9ead008f0dcf85bb80 : SUCCESS in 5m 22s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3127b4b437354643a1a067a4c341ce2d : SUCCESS in 7m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c38aa03ee9dd490fb49863c030ff488c : SUCCESS in 1h 08m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/58c7fddae0644991becaee2314ff0352 : FAILURE in 18m 39s","accounts_in_message":[],"_revision_number":5},{"id":"96d7d0694776df391eb74df6cb705770d3112c36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-19 20:13:13.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"c5e4d6827141aca276e551b1cf894a1b93844acd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 21:21:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb4aad16eee44b94b31ddc1baa2bfc30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51283edbf5ba495a95d6ef90c3c72a11 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d157882ba450464a8f89f5f23ed60e26 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd408bb5f63a4c71b3358611c2efdd45 : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d1db5dffb0d443eb7b67222636b0b99 : SUCCESS in 5m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/acb34f90a0b44636b9bcec135314efa5 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8e4d510e1d17470aba347d3b2bd881a8 : SUCCESS in 3m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bd282986458b4f629efe8577c279b574 : FAILURE in 3m 10s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8db9bdd52b1141648b84b41465dc4aa9 : SUCCESS in 9m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/79b671c5df0c43a681d3963e0108cc8c : SUCCESS in 22m 06s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c08dd545ff174131b5ab81051fb5fdc3 : SUCCESS in 26m 42s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/71992d42cca14a68898436da22880b7c : SUCCESS in 5m 57s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/67a6515744584f909c492f99a1865871 : SUCCESS in 7m 07s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e4452ff1d8514c8fa502d903ee752fd3 : SUCCESS in 5m 06s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/444667bee08046a79a75ac0cb9161806 : SUCCESS in 13m 16s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9917c20df60642e8b9a829b002b6d755 : SUCCESS in 6m 56s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/594c6f1108bc416584bd4974ed4912c4 : SUCCESS in 7m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/038fff28c44a460aba0c065df604da55 : TIMED_OUT in 40m 55s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/fecad5f0dcca4bd199e5458bf3aec68c : FAILURE in 8m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ad41eaa6179415ab686829574ba5182 : SUCCESS in 6m 48s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c0c7e59e6be04dd0a1e85f3f538f1d74 : FAILURE in 5m 15s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/6c0861dccdd74658b26366106ddb2b87 : SUCCESS in 12m 50s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e66afa0bc7a346738c9f70e28bee9db3 : SUCCESS in 4m 10s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d6c061737b49460c8a1c96597a64eda8 : SUCCESS in 12m 42s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ed3eceef1a634b0d8938cd4dc4a841ef : SUCCESS in 4m 40s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/1490e4982bd24f5b87cc0f363823c7e2 : FAILURE in 6m 13s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/802cba5c225548588cc00ba3844ce26a : FAILURE in 13m 19s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d8119586ec49401fb7d23ad1c5e4d1b1 : FAILURE in 20m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/aa3e03889ec54c16baae3242f99cbfc3 : SUCCESS in 9m 19s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/054211f86942429fa9b38ef356b67075 : FAILURE in 11m 58s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0b029ef7357846e1b00550e0a1d297c8 : SUCCESS in 5m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9d8d05c3c5f74b80ac310a6afb485cda : FAILURE in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f52c4b2e3df44ff8b84401fb03469cf4 : FAILURE in 24m 47s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/42495cdc5c3843a78b06359a344b92c4 : FAILURE in 8m 02s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/aa542771493e4698bd5e7a142a5f5fe2 : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4f13055bc5214c9b9c94872681ff9747 : SUCCESS in 6m 09s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6fe58eb568fe42738983f79b8c55c77d : SUCCESS in 20m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bef6d21d72a47299d4c3ac63682f78f : SUCCESS in 59m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c5a32599b6540abaf7fb6ad0d332043 : FAILURE in 20m 43s","accounts_in_message":[],"_revision_number":6},{"id":"64029dd128ee3aaff5bb021106f09fb39390258f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-21 16:51:00.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"4135c332c10b39a1a04d8c1714a1c3faea0bd9cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 17:56:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eafbdd87d8c94fc084e20c5bd2637295\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecf1911c16dd4b77b27004dff8441cb8 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc3131241ad440c1811131557509890e : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/982827a9b98d432b84630fc414d7630d : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e2a12781f604f99a9eeaaa64cd60b2d : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ce3288c175d4075a46e55488545eda2 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/33b9e9767cb14b21a0de8178c9493411 : SUCCESS in 4m 26s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7c64d41af99843a7beb315ca5fb197ee : SUCCESS in 19m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/a7c8e040d2ec43349d2efa2534e1b8a2 : SUCCESS in 11m 18s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/03ce2273d9874a4faac6990bc186a254 : SUCCESS in 17m 38s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/92cee8b9c0a1451cbb06af33a07275b4 : SUCCESS in 20m 11s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/0c73d9add3144ff19e90dc1da7884034 : SUCCESS in 6m 42s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d3aa4932b37546e0a546be40da9f2f05 : SUCCESS in 7m 22s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/c6f846f2097541969273b2d02cebc00a : SUCCESS in 5m 02s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7dbbf1c761564bf48ebb2016ca926815 : SUCCESS in 12m 35s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ca19dbf300854a19946c71ee0c855e6a : SUCCESS in 5m 26s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/58de0963b72d436789c617540ae5c68c : SUCCESS in 5m 01s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/348af11abfba456192c24b3a401043dd : SUCCESS in 8m 34s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/26d5c9616ff849da9fc7131191dd1656 : SUCCESS in 13m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ad0585abfbdf432fba71ba87245cb171 : SUCCESS in 6m 25s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/e4afba07afe64c0ab18e71de07934067 : SUCCESS in 7m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/5f31a32b2cd54c7bb50d465a4ac1c2fd : SUCCESS in 13m 38s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5ceb890875da4879bb3748262501c5b9 : SUCCESS in 3m 47s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f065db7717ce47488db30b3fd800a9b0 : SUCCESS in 11m 29s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/5b9df41959714c26b9e9671a4b7c23e0 : SUCCESS in 4m 59s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/d30f42e062724e729670dd1e424c0902 : SUCCESS in 6m 35s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/8c423d96f2c64ee7a261a053247645e3 : SUCCESS in 11m 18s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/791c8a4c37934725ad20fac40654d594 : SUCCESS in 18m 50s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/03b7b059a35d4c31b32a7297a1079550 : SUCCESS in 8m 26s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/70f7656f0da44c45b290b51cc7598128 : SUCCESS in 12m 35s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/13d2a51a72f945e1a647f6eaa35cd002 : SUCCESS in 5m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0cbe20c8e34640f6b362af100e84009b : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9b5866da6680475db4f254a70c2357f0 : SUCCESS in 23m 55s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/898b3df032d640a7a4136979fce39004 : SUCCESS in 7m 27s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/c219660bec1b45cca3c6a519944a1b91 : SUCCESS in 6m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ff4ef5c095e94e359e73499d77e89e53 : SUCCESS in 6m 20s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d00c51d7103e4511a930cfaaac127045 : SUCCESS in 8m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4ea199d33534682832dd277635dcba8 : SUCCESS in 1h 00m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ce41ead2f4f405ba134fa5be9dcdf57 : SUCCESS in 56m 29s","accounts_in_message":[],"_revision_number":7},{"id":"79ce78b267480cd3b630340c375acacb54c29f9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-22 13:25:50.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"c2132a55c28e09d5d30539cc7e29944357b2588a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 14:54:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0601400d97c4478b9224fb2448f0d7e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a051f82386d74d2f9f1490dfb2c0a137 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/866e3e39aba8454a97583d6b150c023c : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2ec2b7c386f46188ba90caffda4860e : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fb70f7e995aa4f3c932b85508b66c2aa : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d71d0c48f324956ba9dae9f568222a2 : SUCCESS in 2m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/12896bc9a0b640bf8f8d3622b7dbaaad : SUCCESS in 2m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/38dd02298da840308082588aa33b0397 : SUCCESS in 17m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/45f9baec067242efa7d4159d5f1e4aa9 : SUCCESS in 7m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6aa24349cfc44818a926603755b27e38 : SUCCESS in 14m 07s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7180aceb819b495fba8beba2ae7d3ce2 : SUCCESS in 17m 50s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2f9326e255f84365a92ea362a7fb6d44 : SUCCESS in 6m 23s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e9f46b4e43ba4443ba73605ba25b9c88 : SUCCESS in 5m 05s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/40abbb36ef264060a9a2d20968d32d7e : SUCCESS in 3m 33s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/121497f4309540fbb0389e64dac73252 : SUCCESS in 14m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/181b730d5c654885ab2b3420228e49f2 : SUCCESS in 3m 44s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/39fbc57da5b44a5c95dfb5a3230c8648 : SUCCESS in 3m 55s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e6e4fab68a444dcc8fb5ef1c9b844ffa : SUCCESS in 5m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d449cf4d71fa4e4a927e16355de0a7bd : SUCCESS in 6m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d7e133bcc5c84850b2c40428c7e0de0c : SUCCESS in 6m 00s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/efbde59804c04ece8edc9b9280ced5a6 : SUCCESS in 8m 25s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/420e0c76457947828aae5b3964d85389 : SUCCESS in 10m 23s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a0e82c1469e142e39a0bf60c94151be7 : SUCCESS in 2m 06s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f93bca1f92564af8b735c641617ec8a0 : SUCCESS in 7m 48s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0f6c98b444354b7fa1978f7e69d86d87 : SUCCESS in 5m 01s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b19dda8a36244b10b8bc9dca78166019 : SUCCESS in 5m 35s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/04b16d45ba9b4d8db987b504dc1c0eb3 : SUCCESS in 8m 50s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/66c0e539ffd1434497a03e511ede0765 : SUCCESS in 20m 32s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1f99701c9547453f916b87813eb686fb : SUCCESS in 6m 43s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/d0901e2993844031a25e878e20eb95f1 : SUCCESS in 10m 09s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7009946541d74fd5967b4aa6164813a4 : SUCCESS in 7m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/81e28596f52c468494c6477e64fc0d80 : SUCCESS in 2m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1d8f38cfc8a441fb9e283e50fc4ddfad : SUCCESS in 22m 42s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/beabd64d4ce043899197ef25c96ef23a : SUCCESS in 5m 48s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8db610a65adc4e28b89ff99b1d7d933e : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/23ddeceae4cb4264beea58742be47e1b : SUCCESS in 7m 20s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/0cbcb68e224d40e2ada6a13476e90b39 : SUCCESS in 8m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9b3184cfb074fe08e9a9cee9a5c90fc : SUCCESS in 48m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a14f8f43562e4ee894ac9078655abc15 : SUCCESS in 1h 26m 04s","accounts_in_message":[],"_revision_number":8},{"id":"37c43752229a0cc9bfdf2d7ed1704c297a1287ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-22 20:27:44.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"05b3ce29fc29dd7eddd7dfa6915ee9c9cd16f419","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 22:04:11.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6520dfd32744502bcba7003dcf7fc86\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60130350bddb4ce6af0a792d01de96f1 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a08ec8f8f2bc49e98c111fcb884232c3 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e9b84c486b0452286c964baa1c0f553 : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/35012af5260046b6b31ae085e53e938f : SUCCESS in 3m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c48d9d0e027149329a06fce7dcf5dc84 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/21b209b356b6440aaba6a36818cb0e66 : SUCCESS in 4m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d31b346b805e4084891b8e19241ba8ff : SUCCESS in 16m 37s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9a95d12a208c4f1e918c79568e854b92 : SUCCESS in 8m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/4e485900f20e46b39860ff16c5e69aa3 : SUCCESS in 16m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7a5b7e84b1df4f04a116a4df8f2da773 : SUCCESS in 20m 31s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/33f9b9c8ddff4c138b0091c226302731 : SUCCESS in 5m 53s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d9c1e61c8cb949d2bc74dd9917baa3fc : SUCCESS in 4m 52s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e66bf880d25248fcb9aa2bea3c9d5bf8 : SUCCESS in 5m 45s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/356355226d6d4b3393ec6343f083f855 : SUCCESS in 11m 52s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/df42e9bd61c94be88508c6689923d242 : SUCCESS in 5m 23s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b5c0e7d2b7a340e59ac6a278bf4ed6f9 : SUCCESS in 7m 23s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e1f0423938e54b718f1fb6a745765c1c : SUCCESS in 7m 24s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1ef38b5bfe30402d8bb71a986829d9f3 : SUCCESS in 14m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5b523d41971b4fb7acde7413342141de : SUCCESS in 6m 51s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/bc7c99abc2f24dacb9490b127d6c5f37 : SUCCESS in 6m 59s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/6bd278e63e5644dfb9b32d7df10e5c29 : SUCCESS in 13m 51s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e6514a30c058481c8c2f29a389402ee6 : SUCCESS in 2m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/85297f3328c349a4b3c6b24c9d83410b : SUCCESS in 7m 44s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ef455bb82d8241b4997439cd176359c6 : SUCCESS in 4m 44s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/58ec1deffeb84ed886c648f226c91962 : SUCCESS in 7m 05s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/abd3f7ac797642759a1af3724860110c : SUCCESS in 10m 22s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/84e7a5d353b345cc9cbc8a07646ce4a3 : SUCCESS in 17m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3b4e60e2b5dd4c76b37cf6ce8ebad18b : SUCCESS in 8m 28s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f163f9e8380848dd982a553e5371e661 : SUCCESS in 10m 32s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5c643f5421f14610a8b3a6524d97fc92 : SUCCESS in 6m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d352bd6c32f04600bd62506333d76b1a : SUCCESS in 2m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a08d6a663ab3438ba144127f5da8c1d5 : SUCCESS in 24m 56s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/44ee608b1ed6416d9d08776db256e21e : SUCCESS in 4m 04s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/fe3ff0d2ff464dd6a5b6f50c910cbca2 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b341611aed564e7ea0d23e7d8d3b5575 : SUCCESS in 5m 31s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/64e8006bf9844da785bb19e1c5583219 : SUCCESS in 17m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80fdd414d9514a7d8c17296e9a2466ad : SUCCESS in 1h 00m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc2de1c3b56c4784b72a96f7a3fd3616 : SUCCESS in 1h 31m 30s","accounts_in_message":[],"_revision_number":9},{"id":"e392a1012b10efe82d7716e63dc8268ed030f9b2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-23 07:22:23.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"35659a4be04a9e642d3406851d4b59158f0b2d7e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-23 07:22:30.000000000","message":"Topic openstack/requirements/constraints removed","accounts_in_message":[],"_revision_number":9},{"id":"2a409339fc43f32f59e30c550ddebae0b35286c1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-23 07:22:55.000000000","message":"Topic set to openstack/requirements/constraints-noclob","accounts_in_message":[],"_revision_number":9},{"id":"6c934d71edcf80288ee9edd86ff710ca7428e9b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 07:23:05.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"462043be02506d8bacebd3f2eb57701157badfc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 08:46:09.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62a39a0755834aebb37bf2cd9d938a76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ef72893799e4e05ab95ecbd4469cf9d : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7132637dc4db40cda9828d6750a4e4d3 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad7e678541984071bdbae9f7609bbd8f : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f9e472db3160473f9f0b93063ab793b3 : SUCCESS in 3m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae51c699c4e546b4a5ad286aaaaacee5 : SUCCESS in 2m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c2ab0c04f222486682012f518b938e67 : SUCCESS in 2m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d62818fcbed945daa7ed8b5b01d664b2 : SUCCESS in 15m 01s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8e4c7213b988473aaf2ed085f3773c84 : SUCCESS in 7m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/77b898fb524148789a18d7281beed6c1 : SUCCESS in 8m 59s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f7d9bfe6def747f6ad43f4deeee4a5b4 : SUCCESS in 14m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e16fadf600444f7282356108558fd4e3 : SUCCESS in 11m 57s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0a4d5e07547749e8ae5a475d8e0cd6ec : SUCCESS in 4m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0849d6441a3146629ae9d8f6d69ffb7b : SUCCESS in 4m 24s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/02ff32d5a07341d2b31aa3b615dac225 : SUCCESS in 7m 42s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/29a3b48ec29d4677b308e1d6d1e07ed3 : SUCCESS in 6m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2ba386a4de2249be96385a69f3a42711 : SUCCESS in 7m 45s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/27bafc8b5f6744b1bd383b907a686348 : SUCCESS in 8m 05s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/46be0eab99844c6eac9097557c375a8a : SUCCESS in 8m 01s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/916d6b18ede04017986c3996f9dfe5a1 : SUCCESS in 2m 09s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1040dbdfffdc4e4ab0cfe7068cffff0d : SUCCESS in 10m 05s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/611c206302e54a019fd91693b7b9f1a1 : SUCCESS in 16m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b92def65e797483cae8d2a561d8d27f2 : SUCCESS in 5m 58s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9cb2f54cb8b64f7c928b6a35fcfc5093 : SUCCESS in 9m 24s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/3f47c40275ad4f3cbcdddef797e86698 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5a981ad2d53f4bbaaff4a161d0751b2d : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8f9d3e159abb4c468580c24bad6aacd0 : SUCCESS in 24m 10s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/de3d9bea435a4501b92d7d3299f80048 : SUCCESS in 5m 46s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/dab5abfd21e8463a83fb095791be4029 : SUCCESS in 4m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8a51052a784943a090c7398f9e2c627b : SUCCESS in 4m 30s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/07392e57078d4b6cb5d0e6c23728ca23 : SUCCESS in 12m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a853d4eee704418796ee087252f58ab2 : SUCCESS in 45m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f20c155ccd55446aa24341c1aac42ebc : SUCCESS in 1h 22m 33s","accounts_in_message":[],"_revision_number":9},{"id":"7cb2d3119fb598039226436baadff81d9dfa57c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 08:46:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"bb1ac1241429dc06ad159ac76bb85c0d5dc8a4f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 08:47:08.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6acaad24d3904151904181265e180dea\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a119755b25284b1ead9568261a04e0a5 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"af4b18e4f1519948e18a1d94cec31b75b13c37ba","revisions":{"0483c9a758d1ffe90354967bd76249d8263cfae7":{"kind":"REWORK","_number":1,"created":"2024-10-26 08:21:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/933484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/933484/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/933484/1"}}},"commit":{"parents":[{"commit":"5b521399c0deedba22865b93ea45ba99ba146888","subject":"Merge \"update constraint for python-openstackclient to new release 7.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b521399c0deedba22865b93ea45ba99ba146888"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-10-26 08:21:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-10-26 08:21:19.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iec4403045e59030e13d5b42dc1a878af35899892\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0483c9a758d1ffe90354967bd76249d8263cfae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0483c9a758d1ffe90354967bd76249d8263cfae7"}]},"branch":"refs/heads/master"},"01e6de6a9a9bbd8601b4249bec37ce449096ed00":{"kind":"REWORK","_number":2,"created":"2024-10-31 06:19:43.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/84/933484/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/933484/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/933484/2"}}},"commit":{"parents":[{"commit":"f2d82273f2bad2dd670b7d1f6047353f61a098f4","subject":"Merge \"add rudimentary asyncio database clients\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2d82273f2bad2dd670b7d1f6047353f61a098f4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-10-26 08:21:19.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-10-31 05:48:53.000000000","tz":60},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iec4403045e59030e13d5b42dc1a878af35899892\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01e6de6a9a9bbd8601b4249bec37ce449096ed00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01e6de6a9a9bbd8601b4249bec37ce449096ed00"}]},"branch":"refs/heads/master"},"ad0ea335a5c9c1d4d701c8894d9ce42030b4fadb":{"kind":"REWORK","_number":3,"created":"2024-11-02 08:12:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/933484/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/933484/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/933484/3"}}},"commit":{"parents":[{"commit":"e2bc5f849c196f110d63762bfac400490f6df673","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2bc5f849c196f110d63762bfac400490f6df673"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-02 08:12:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-02 08:12:25.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iec4403045e59030e13d5b42dc1a878af35899892\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad0ea335a5c9c1d4d701c8894d9ce42030b4fadb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad0ea335a5c9c1d4d701c8894d9ce42030b4fadb"}]},"branch":"refs/heads/master"},"f2b62990bc3bdefdacf40e90ea5b6ba1abe7de0e":{"kind":"REWORK","_number":4,"created":"2024-11-09 08:38:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/933484/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/933484/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/933484/4"}}},"commit":{"parents":[{"commit":"1daffb38e8d91ac33de84c850ba406bcec7a8aaa","subject":"Merge \"Bump importlib-metadata to version 8.0.0 (\u003e\u003dpy3.10)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1daffb38e8d91ac33de84c850ba406bcec7a8aaa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-09 08:38:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-09 08:38:43.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iec4403045e59030e13d5b42dc1a878af35899892\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2b62990bc3bdefdacf40e90ea5b6ba1abe7de0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2b62990bc3bdefdacf40e90ea5b6ba1abe7de0e"}]},"branch":"refs/heads/master"},"fb62b7c39bff32f47e273579900bfa935a40c87e":{"kind":"REWORK","_number":5,"created":"2024-11-16 08:18:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/933484/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/933484/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/933484/5"}}},"commit":{"parents":[{"commit":"1372e0d4d66c0640f63be5176a1bd7a25eda39eb","subject":"Merge \"update constraint for python-manilaclient to new release 5.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1372e0d4d66c0640f63be5176a1bd7a25eda39eb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-16 08:18:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-16 08:18:48.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iec4403045e59030e13d5b42dc1a878af35899892\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fb62b7c39bff32f47e273579900bfa935a40c87e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fb62b7c39bff32f47e273579900bfa935a40c87e"}]},"branch":"refs/heads/master"},"822be7f4f872298a68823f0e88fc7d98051ce034":{"kind":"REWORK","_number":6,"created":"2024-11-19 20:13:13.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/84/933484/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/933484/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/933484/6"}}},"commit":{"parents":[{"commit":"376cf882ece239abee0e2c2ffb83f3bbefc1819c","subject":"Merge \"update constraint for tosca-parser to new release 2.12.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/376cf882ece239abee0e2c2ffb83f3bbefc1819c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-16 08:18:48.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-19 20:09:43.000000000","tz":60},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Iec4403045e59030e13d5b42dc1a878af35899892\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/822be7f4f872298a68823f0e88fc7d98051ce034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/822be7f4f872298a68823f0e88fc7d98051ce034"}]},"branch":"refs/heads/master"},"a48992d08a6cdd6697360d644a6e2074af36647f":{"kind":"REWORK","_number":7,"created":"2024-11-21 16:51:00.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/84/933484/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/933484/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/933484/7"}}},"commit":{"parents":[{"commit":"f133a446a4b69d08f65be2870736f6774288a938","subject":"Merge \"update constraint for stevedore to new release 5.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f133a446a4b69d08f65be2870736f6774288a938"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-16 08:18:48.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-21 16:50:46.000000000","tz":60},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nAmended with the following fixes:\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nAdd version specific pins for newly added requirements:\n\n- flexcache\n- flexparser\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- waiting\n- croniter\n- netaddr\n- eventlet\n- cryptography\n- pysnmp-lextudio\n- tenacity\n- jsonschema\n- oslo.utils\n\nChange-Id: Iec4403045e59030e13d5b42dc1a878af35899892\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a48992d08a6cdd6697360d644a6e2074af36647f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a48992d08a6cdd6697360d644a6e2074af36647f"}]},"branch":"refs/heads/master"},"4bb425de0f0a80e10bc55c9ca92ebb76c7bda789":{"kind":"REWORK","_number":8,"created":"2024-11-22 13:25:50.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/84/933484/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/933484/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/933484/8"}}},"commit":{"parents":[{"commit":"f133a446a4b69d08f65be2870736f6774288a938","subject":"Merge \"update constraint for stevedore to new release 5.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f133a446a4b69d08f65be2870736f6774288a938"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-16 08:18:48.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-22 13:23:21.000000000","tz":60},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nAmended with the following fixes:\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- waiting\n- croniter\n- netaddr\n- eventlet\n- cryptography\n- pysnmp-lextudio\n- jsonschema\n\nChange-Id: Iec4403045e59030e13d5b42dc1a878af35899892\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bb425de0f0a80e10bc55c9ca92ebb76c7bda789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bb425de0f0a80e10bc55c9ca92ebb76c7bda789"}]},"branch":"refs/heads/master"},"af4b18e4f1519948e18a1d94cec31b75b13c37ba":{"kind":"REWORK","_number":9,"created":"2024-11-22 20:27:44.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/84/933484/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/933484/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/933484/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/933484/9"}}},"commit":{"parents":[{"commit":"76de0f944002b8f206e9fda055fa196922c9bc15","subject":"Merge \"update constraint for osc-lib to new release 3.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/76de0f944002b8f206e9fda055fa196922c9bc15"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-16 08:18:48.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-22 20:27:07.000000000","tz":60},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nAmended with the following fixes:\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- waiting\n- croniter\n- netaddr\n- eventlet\n- cryptography\n- pysnmp-lextudio\n- jsonschema\n\nChange-Id: Iec4403045e59030e13d5b42dc1a878af35899892\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af4b18e4f1519948e18a1d94cec31b75b13c37ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af4b18e4f1519948e18a1d94cec31b75b13c37ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
