)]}'
{"id":"openstack%2Frequirements~933643","triplet_id":"openstack%2Frequirements~master~Iec518dacd19029c7bf03491541e11ebb7d29fa62","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iec518dacd19029c7bf03491541e11ebb7d29fa62","subject":"update constraint for python-openstackclient to new release 7.2.1","status":"MERGED","created":"2024-10-29 13:00:54.000000000","updated":"2024-10-31 05:57:30.000000000","submitted":"2024-10-31 05:56:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933643-new-release","meta_rev_id":"a75ba20f7ba7cf4e8112c10b6f6de57daff9ee8d","_number":933643,"virtual_id_number":933643,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-31 05:56:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-10-30 10:56:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-10-31 04:39:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-29 14:40:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-29 18:08:19.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"69d09f6866c700c30d0aa424c3019276cb0a9245","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-10-29 13:00:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e6e27d1094d9eeb0facfb89f448163e4c02edb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 14:40:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b52cad702c4a428db0f27de315ee64df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9a0d9d697164dc299b89cd3a2ff9302 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5653f5ab7a9a4b9c97580a7020589ac9 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c074a486229844879b2dd217badd32f0 : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8708bc6e794a4a3db68539db019b9f59 : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f1f419d637548639f11f15f57d92e7e : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/93804b7d9d214e80aa4b2b6f95a0b8c8 : SUCCESS in 3m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/08727e38cf9e42f4915986b6270bf278 : SUCCESS in 17m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b7745eba3c4f42919e6a306a3b8709df : SUCCESS in 8m 10s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/e0237cbe9bdd40c1b54ef1b3349e2604 : SUCCESS in 18m 18s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/834ba989489449b3849ef71dc60cad27 : SUCCESS in 24m 04s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/995676c8ee684800b41efd77ba3b1951 : SUCCESS in 4m 12s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/fb3f5c6727c44b66957e1118da98d18c : SUCCESS in 5m 53s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/a42124f8060e4d318b10c766822b2b0c : SUCCESS in 3m 54s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/3c77cadf71304d5eb120b6a2bc2d74ff : FAILURE in 10m 39s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/47a327b7f9694396943c877d5177b403 : SUCCESS in 5m 38s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/fc897497db6b47b38067f5f6d708ab92 : SUCCESS in 4m 48s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/17c6be77ac114334b15315807f6d9f56 : SUCCESS in 7m 02s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/df302ea18aec4444972d59dbdc97b47c : SUCCESS in 7m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/46c24a2a85a24e219b8a4cff35240bff : SUCCESS in 6m 37s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/760d1b2e70ed4791a065ccb1d6691471 : SUCCESS in 8m 07s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/c565202d09de4949966a8a6002d74415 : SUCCESS in 10m 11s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/d744dfa43f83494799b50f1ca156b309 : SUCCESS in 1m 56s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/a38b799dd12c4af0abd970405ad78f59 : SUCCESS in 7m 46s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/e86682c6730f4da284baacbab10c25cb : SUCCESS in 3m 30s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/e554d3d2c4a541f69d584899cbd5e285 : SUCCESS in 4m 06s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/e4025e95312f402c84fa86ffc86ba460 : SUCCESS in 10m 35s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/01f9ecc6c80649b6879c2a478448d0d9 : SUCCESS in 22m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2b80105cc3d1492686827ec729ac7211 : SUCCESS in 6m 21s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/60cb9783d3fa4c09af84037be9ca16c0 : SUCCESS in 11m 11s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/faa41fb29d40480eb04cccb46af963d1 : SUCCESS in 3m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/58bb443bccea465897bb6798171a8e7a : SUCCESS in 3m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/193f9afdafd34c8a87a7cb9f5672ed48 : SUCCESS in 21m 15s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/cc2793fb6792447fafdcc860c891320a : SUCCESS in 4m 42s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/c4483b5c7e0d4923a10321c9c7cae01e : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6426087d1a924fa8974ba53016f369d3 : SUCCESS in 5m 33s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/3a81639783ff44f69275328d3ea364c0 : SUCCESS in 18m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2eb4bc03935d4dbeab58ccb90975ff80 : SUCCESS in 59m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e5407e6ce6a473da10e443ec05de1dc : SUCCESS in 1h 38m 37s","accounts_in_message":[],"_revision_number":1},{"id":"9d6f82cd8dbd72d1037bb9094c95303c929a5ff2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-29 18:08:19.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab5700f2b7a2a32fd66c5767055d6228363fd5e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 20:07:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a11078fb39e4c09b7671a0acc03a8e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de9fb6bc9b8d40c7be05ab343d438179 : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/216412f493cf4379acfd74aa0e2e415b : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3730dd07d82e48498ad4c4248abf4c65 : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1e4bd6fa85394f2f978c3889461e58ef : SUCCESS in 2m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d706e25ba371476bb6e611ca944fd047 : SUCCESS in 2m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8bfc7a2774694fd495fb9ff1d3397955 : SUCCESS in 2m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/559df560471c462c82c57f632f318198 : SUCCESS in 8m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/af0a6617198141739046c8e3723a057f : SUCCESS in 36m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/84aeac3acb724e1ca90758a9f43997d2 : SUCCESS in 13m 15s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/bdd888d146d541308e71b44cb693f4d6 : SUCCESS in 20m 45s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/e965834bf84b43f090dab2a721ef0f45 : SUCCESS in 3m 37s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/c2f6ea0f61464ff69a5d2c8cb6d7fb17 : SUCCESS in 4m 49s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/35de5d76e1e3489cb661ef5f662e89f7 : SUCCESS in 4m 17s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/9511110478ce49cc99985cf28bf08b88 : FAILURE in 11m 21s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/8f1c5abdd0b84305a2c5ce6f12f55c67 : SUCCESS in 4m 33s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/5d96551865fd46e6b1af0794cbd52d31 : SUCCESS in 4m 57s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/900a9617787848c68b4b4ee4331ba52e : SUCCESS in 5m 03s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/4b17c728c4bd4ecc9d078c3fc1ccbe29 : SUCCESS in 13m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e0877d61b0fc49e38dd615dbf9005bbc : SUCCESS in 4m 48s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/ed98bae699924bc7813d6e83490a6cba : SUCCESS in 9m 33s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/baae32c337994643b96202098123b946 : SUCCESS in 11m 06s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/a2f28a3e165545c4ba4fa5c0471776e1 : SUCCESS in 3m 11s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/5f6985fd0065466bad3d31795e8f9eaf : SUCCESS in 8m 28s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/9471f8d2758a454fb9140cd73be4e699 : SUCCESS in 3m 57s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/4f6c6c542e3a411493a0350302164f37 : SUCCESS in 3m 51s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/a71e8e5b0ccc4c80b96467ae1f4f7660 : SUCCESS in 12m 14s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/989600876980458b87493fd005229cd6 : SUCCESS in 16m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1972c0bedce64a36b29ea375822f0ba6 : SUCCESS in 4m 45s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/9348ddf4821641d097f3a656a0b8cca2 : SUCCESS in 13m 19s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/f21f775ae149435c8ecf8424788980d1 : SUCCESS in 5m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/937605c0eb0d4e00a6bde12b1bbac598 : SUCCESS in 2m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/60ec58d0200a496eabbccfeab8ab46e0 : SUCCESS in 23m 00s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/c8976b4b60b543629f4cb29dd48bb276 : SUCCESS in 3m 50s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/128148d3098947f29404bad969c80c7b : SUCCESS in 3m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b96fac9172714df795a19e7f8e816536 : SUCCESS in 4m 15s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/8f9d1b587a1742c493ea58401c5c000c : SUCCESS in 14m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dcdc6260aee49d592b5bdca58056d48 : SUCCESS in 45m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c76cbf41f3e456e8acd99aac768163b : SUCCESS in 1h 46m 53s","accounts_in_message":[],"_revision_number":1},{"id":"0d239ae67d98cbf2f59af6e52f3847189263e197","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-30 10:56:15.000000000","message":"Patch Set 1: -Workflow Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cddb1f5e9c1af6b6278c0d27c27de6487ccddfa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 12:14:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce23245b82724628a49c5085f107093c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07dc61239cfb4974872090fc4d91186f : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c79225b12fb4950abc8b79eeac64e39 : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd179dbfcd5a4f82b018027988f10586 : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e0b73fb81a94471ba3e3cc2bac955355 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a363f099bf24211ab458521840681a3 : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d8aa4929639a443a9e238ffe7cde869c : SUCCESS in 3m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ada954ddbb7d4b8e84d0309d3b89e33e : SUCCESS in 15m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/1bb33c733f6c4a4cb582fd6d07e7d6b7 : SUCCESS in 9m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/baf1801dc3fc427d99af8429ef71c633 : SUCCESS in 16m 12s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e06fa08aaa6d444d9b1235bd7eea6c52 : SUCCESS in 24m 21s (non-voting)\n- cross-aodh-py311 https://zuul.opendev.org/t/openstack/build/1fb99529a16c44a08a5940fcb1b2eea2 : SUCCESS in 4m 32s\n- cross-barbican-py311 https://zuul.opendev.org/t/openstack/build/f5d5235f9d2243819045b417bca7639d : SUCCESS in 4m 35s\n- cross-ceilometer-py311 https://zuul.opendev.org/t/openstack/build/b14b5bc5c8224e9f8caac383818e0b0b : SUCCESS in 3m 31s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/a1b2a373ab784b36bc7eed0f66ed2e08 : SUCCESS in 9m 33s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/109f1ddf745c43499cb9ac8d5e81c078 : SUCCESS in 3m 37s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/25d1d15e394745a4909fa86b65f91a60 : SUCCESS in 4m 57s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/ece0013b246846d0844bb11d70e76547 : SUCCESS in 8m 46s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/92a584e9ce9e47c4af858feef5c0d802 : SUCCESS in 13m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/88a35209c5ed42eaa9367038c0e84b75 : SUCCESS in 14m 02s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/60459e2bad834512930a327de118aca4 : SUCCESS in 8m 05s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/fe6aa00497ae4ceeb0522f90405678d2 : SUCCESS in 10m 17s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/e6a8c0d1f1884cea92f73fbbc22ea088 : SUCCESS in 1m 52s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/4a50eb82846a48ddbaed62cb15518ab1 : SUCCESS in 9m 55s\n- cross-magnum-py311 https://zuul.opendev.org/t/openstack/build/53e7a03043d24377a599c5ff7704953c : SUCCESS in 5m 01s\n- cross-masakari-py311 https://zuul.opendev.org/t/openstack/build/2f9823eae3654f7f8e2010136a7eaf11 : SUCCESS in 3m 08s (non-voting)\n- cross-mistral-py311 https://zuul.opendev.org/t/openstack/build/53aff4e54d8f44b281f8a4e5e452881c : SUCCESS in 11m 03s (non-voting)\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/f065c8ffa8994087ac3a2d2cd20907a5 : SUCCESS in 23m 35s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/de1c73d3ee9c450a9efde031a0eb632b : SUCCESS in 6m 48s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/64422cc346cb4f2380324c64a86b6405 : SUCCESS in 12m 48s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/ab42f09d6e354e60bcec56f281c4e237 : SUCCESS in 8m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/09709f29fa584953894a365919830520 : SUCCESS in 2m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2ba761afe2e948848f698c8ab5af67e3 : SUCCESS in 23m 47s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/2103c60d24de4ac695a1cbb393d3b210 : SUCCESS in 3m 19s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/64a0d2d579f6418da165939a921f5e82 : SUCCESS in 3m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/971b06bf0b1941cfbc15e0d5268a6aaa : SUCCESS in 5m 41s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/e82ea7e1f7744cada11077a3c935a69a : SUCCESS in 20m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f967bdec208841c783e99b8da71824a8 : SUCCESS in 1h 02m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa07edc3c3d544c49c1f254e80477c68 : SUCCESS in 1h 03m 28s","accounts_in_message":[],"_revision_number":1},{"id":"075f1fd7b9fa0b20f1afacd779b1289333ea63e8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-31 04:39:55.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fd94b7b746f634fce770be01f8f12f53d43f225b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 04:40:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23dc3e1883274149643cce81430dbbcbccf199ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 05:56:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e88e641c3255436aaa406b4033ec9d9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3419257b01954450b4cbe73b1e86a809 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95e67d0ec5864e55b3e9ab98abe0cefb : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15853c903e664be29613ab91778e01ab : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65ec498d21d1468a8bdf26061be84cd2 : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bbd51ddcf2f24a07a0e13b60667e621b : SUCCESS in 2m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fcdd2729937744838395afeb97efd479 : SUCCESS in 2m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/98ab3823593c440fb0f6437051b96536 : SUCCESS in 11m 12s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b6c019017ace47959ad19fce23bf2bfe : SUCCESS in 7m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/994d3d1bba184489b6d597f5017514aa : SUCCESS in 11m 57s\n- cross-cinder-py311 https://zuul.opendev.org/t/openstack/build/7d16182600044e71ad1f517263358d50 : SUCCESS in 9m 15s\n- cross-designate-py311 https://zuul.opendev.org/t/openstack/build/e65361e31fcd4b8e9130edfb1018cb1e : SUCCESS in 3m 50s\n- cross-glance-py311 https://zuul.opendev.org/t/openstack/build/7dad5c568f204d54bab605084edc08e8 : SUCCESS in 4m 00s\n- cross-heat-py311 https://zuul.opendev.org/t/openstack/build/c4ada5b104e743b09d364f0eb7690f5c : SUCCESS in 5m 30s\n- cross-horizon-py311 https://zuul.opendev.org/t/openstack/build/2bca8b37268648e490f481ca673e880c : SUCCESS in 11m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6219453f00d94269a0b8ff11acf02839 : SUCCESS in 9m 47s\n- cross-ironic-py311 https://zuul.opendev.org/t/openstack/build/efd8cab7cf934d7eaae626ec88856384 : SUCCESS in 7m 57s\n- cross-keystone-py311 https://zuul.opendev.org/t/openstack/build/542dce48bb2d401799ebb56bdeabfd16 : SUCCESS in 10m 04s\n- cross-kuryr-py311 https://zuul.opendev.org/t/openstack/build/be9df684fed14998b6116d27c69050fc : SUCCESS in 3m 10s\n- cross-manila-py311 https://zuul.opendev.org/t/openstack/build/afd1423b749844779f8eef37ea98a9d6 : SUCCESS in 7m 02s\n- cross-neutron-py311 https://zuul.opendev.org/t/openstack/build/be6f6caa4d144e3bbdd58177e975b95c : SUCCESS in 17m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/7f52cc0134c34a15b932162856b6c1f0 : SUCCESS in 4m 45s\n- cross-nova-py311 https://zuul.opendev.org/t/openstack/build/99fe91755e2641b4b5996fb9aaf55dc2 : SUCCESS in 10m 52s\n- cross-placement-py311 https://zuul.opendev.org/t/openstack/build/c4775d57bdc64d779446e643b11deba3 : SUCCESS in 3m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/40c74d54f6b84f68be60618bf7f1dce8 : SUCCESS in 2m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/12f9a29d2e9241b89d7d76ff89e139c8 : SUCCESS in 24m 30s\n- cross-placement-functional-py311 https://zuul.opendev.org/t/openstack/build/4de6d0f586a143f58cb5d070d897b732 : SUCCESS in 4m 13s\n- cross-octavia-py311 https://zuul.opendev.org/t/openstack/build/f3fc06749bc14828ad6a31adf48968f3 : SUCCESS in 4m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d1bdc2e85c4a40cbb0aa39048dd57e85 : SUCCESS in 5m 01s\n- cross-swift-py311 https://zuul.opendev.org/t/openstack/build/9fcdf291040041fe939873f2a285363f : SUCCESS in 17m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05a1d15bd639497ba9a7cf741d996ca0 : SUCCESS in 58m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e11d707771ef4291bd65d0d531cb46ad : SUCCESS in 1h 14m 42s","accounts_in_message":[],"_revision_number":1},{"id":"4974e1abb68060777daad011507aae48d1bab9bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 05:56:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a75ba20f7ba7cf4e8112c10b6f6de57daff9ee8d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 05:57:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/368e25072791447c9192ea9c599fb551\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bd268ca79d74bb3b798468c12828a40 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af5c54811a9f6fba363da929de2e106fe474324c","revisions":{"af5c54811a9f6fba363da929de2e106fe474324c":{"kind":"REWORK","_number":1,"created":"2024-10-29 13:00:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/933643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/933643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/933643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/933643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/933643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/933643/1"}}},"commit":{"parents":[{"commit":"f2d82273f2bad2dd670b7d1f6047353f61a098f4","subject":"Merge \"add rudimentary asyncio database clients\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f2d82273f2bad2dd670b7d1f6047353f61a098f4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-10-29 13:00:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-10-29 13:00:52.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 7.2.1","message":"update constraint for python-openstackclient to new release 7.2.1\n\nmeta: version: 7.2.1\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Commit: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Change-Id: I6c504329aaf3be63141913ea07047e4f70f43ded\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Iec518dacd19029c7bf03491541e11ebb7d29fa62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af5c54811a9f6fba363da929de2e106fe474324c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af5c54811a9f6fba363da929de2e106fe474324c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
