)]}'
{"id":"openstack%2Frequirements~934253","triplet_id":"openstack%2Frequirements~master~I67b818cdff01d29c62497826d4b4e3387c74e9e9","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"1":{"account":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"last_update":"2024-11-08 02:10:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I67b818cdff01d29c62497826d4b4e3387c74e9e9","subject":"Update requirements-check job to run on arbitrary targets","status":"MERGED","created":"2024-11-06 16:09:43.000000000","updated":"2024-11-08 02:21:39.000000000","submitted":"2024-11-08 02:10:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934253","meta_rev_id":"196ef43a897d2b24cd71b09f257c52b65f0dd68e","_number":934253,"virtual_id_number":934253,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-08 02:10:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-11-07 11:47:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-11-08 00:26:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-11-08 00:26:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-06 16:14:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-07 11:47:09.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-11-08 00:26:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"70871453158848f7f41c1eceae82e0297cce7fc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-11-06 16:09:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c3947c91cffdde0b2de0b9a58453820c54c80ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 16:14:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91d9b413a1de4b6fb7199cc1fb682109\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b7a1205909e465bac2e47b8813c386f : SUCCESS in 3m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d592f514242d4fcc9b44829958699c84 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"a8f91ee3765b0cc21682e0d54ff7829d644ce3bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2024-11-06 17:15:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0b1743f54fbb233755e3dc02fe3d1430f4bc9288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 18:50:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d26894b3879497c89a4f1e14b4e3230\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec89ffd51a4d4df1b4d1487703c7f1ce : SUCCESS in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec5b5161e1d34a3e8d7036dbc52b292a : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d3ff4939c2644f38f38a9c34f1c3e92 : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f0d8f9802eba45ddb04173e5e4c215fb : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/606478854c6a43a0bdd3bde2f91a4a02 : SUCCESS in 2m 51s\n- requirements-check-self https://zuul.opendev.org/t/openstack/build/b7373a74d39749339c145763486b1e54 : SUCCESS in 3m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d51e6d2bb27416191e625c142210900 : SUCCESS in 57m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/81b9539cbabc49aa9cfe640ae1daabb8 : SUCCESS in 1h 32m 29s","accounts_in_message":[],"_revision_number":2},{"id":"badd8780cda6e462a3795861c1f094721e77d90b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-11-07 11:47:09.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1ca26ca849b6ec97c995cae5f51ae909b0c3a834","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-11-08 00:26:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a4619d7d161f4336430a433421ce991cba64db98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 00:27:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9ee436e10da9eb489d8fa569cd8b9dda6f0a0b09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 02:10:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45ed33d252ff47f1b7c982f12bc630d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa9b737984234e16b11dbab62cf08b30 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/573f518669f44f31b80814b09aecf07c : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/364176bea45b4ecd8fc362e9ea3e29a5 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1bcbbe0de66b4ad7aa04af01f5964dda : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/82debd72bc294cf3a956f0d96fe16174 : SUCCESS in 2m 55s\n- requirements-check-self https://zuul.opendev.org/t/openstack/build/d9cc6ffe5ea44d2792931274cda09922 : SUCCESS in 3m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/455ffa181cdd4eb9903f9590c9c77fcf : SUCCESS in 45m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/668935531e1a46aea1c1743fc059d54e : SUCCESS in 1h 37m 06s","accounts_in_message":[],"_revision_number":2},{"id":"0a81d671b93921ecf621f681d655295406746cd1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 02:10:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"196ef43a897d2b24cd71b09f257c52b65f0dd68e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 02:21:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54ed97b2f92a4869b64d59c388f15df2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5615a90feb544f95ae297d7d1450e89f : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4fe91e20fafd897e966e3dd65feecf48c4fb9a81","revisions":{"2c47212d76b71e736ab255f563585f7ce811204a":{"kind":"REWORK","_number":1,"created":"2024-11-06 16:09:43.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/53/934253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/934253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/934253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/934253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/934253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/934253/1"}}},"commit":{"parents":[{"commit":"da4047aef168b13a5304638a778578cfbfc5cd9f","subject":"Merge \"Switch cross tests to python3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da4047aef168b13a5304638a778578cfbfc5cd9f"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2024-11-06 16:08:35.000000000","tz":-480},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2024-11-06 16:08:35.000000000","tz":-480},"subject":"WIP: Update requirements-check job to run on arbitrary targets","message":"WIP: Update requirements-check job to run on arbitrary targets\n\nUpdate the requirements-check job to run on repos other than the\nzuul.project, but keep that as the default.  Add the job as a check\nand gate job to this repo, but targeting nova.  This allows us to\nrun the job on changes to its definition.\n\nChange-Id: I67b818cdff01d29c62497826d4b4e3387c74e9e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c47212d76b71e736ab255f563585f7ce811204a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c47212d76b71e736ab255f563585f7ce811204a"}]},"branch":"refs/heads/master"},"4fe91e20fafd897e966e3dd65feecf48c4fb9a81":{"kind":"REWORK","_number":2,"created":"2024-11-06 17:15:13.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/53/934253/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/934253/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/934253/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/934253/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/934253/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/934253/2"}}},"commit":{"parents":[{"commit":"da4047aef168b13a5304638a778578cfbfc5cd9f","subject":"Merge \"Switch cross tests to python3.12\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da4047aef168b13a5304638a778578cfbfc5cd9f"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2024-11-06 16:08:35.000000000","tz":-480},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2024-11-06 17:14:56.000000000","tz":-480},"subject":"Update requirements-check job to run on arbitrary targets","message":"Update requirements-check job to run on arbitrary targets\n\nUpdate the requirements-check job to run on repos other than the\nzuul.project, but keep that as the default.  Add the job as a check\nand gate job to this repo, but targeting nova.  This allows us to\nrun the job on changes to its definition.\n\nChange-Id: I67b818cdff01d29c62497826d4b4e3387c74e9e9\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/934249\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4fe91e20fafd897e966e3dd65feecf48c4fb9a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4fe91e20fafd897e966e3dd65feecf48c4fb9a81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
