)]}'
{"id":"openstack%2Frequirements~935048","triplet_id":"openstack%2Frequirements~master~I9a2aa6ee4148fd4fb4bde957b1457dcffd28cca4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9a2aa6ee4148fd4fb4bde957b1457dcffd28cca4","subject":"update constraint for oslo.concurrency to new release 6.2.0","status":"MERGED","created":"2024-11-14 09:32:39.000000000","updated":"2024-11-17 00:50:20.000000000","submitted":"2024-11-17 00:49:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935048-new-release","meta_rev_id":"e5d0c3399e922dc26c1d0efe99c7f106a88e6b01","_number":935048,"virtual_id_number":935048,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-17 00:49:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-11-15 19:48:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-11-15 19:48:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-14 11:01:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-15 16:42:32.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-11-15 19:48:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"3198ce1bdcad43ce7fdf39a49a19c27d0ebe8e25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-11-14 09:32:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f6fb7ca19067eaaa59dbd79cfad9359d5df5420","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 11:01:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8f23f46c6d7445d817e4f6b2005fc07\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9cd2e9ee64e4f72b36d4e77b368d02a : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f8bf8a10f5d45a8b4d59cbd246f2548 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93709baf1289416aaf8c81b71de9e482 : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d47dc69588834b4ea66d86306f5d3a8e : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/796427363edc4eadaecbbf4e9f580340 : SUCCESS in 3m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71f755b5a1c0480585327ab5652ef1a5 : SUCCESS in 4m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/37e327febfd74f5687958aa0b4860f1f : SUCCESS in 14m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ea87ceb9f6ef43319b9e5694372315dd : SUCCESS in 10m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/1df5984c6b8647718e5770d9492b7287 : SUCCESS in 8m 28s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0827c989d3894eec86fa2be6d97bfed5 : SUCCESS in 23m 52s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/70bf93c283084b5eb6932e58426c610c : SUCCESS in 5m 45s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5d8dcf15c4a143a1a873b6af762570d6 : SUCCESS in 4m 47s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/0cf9ee8a42b04bb587cf2ce9aa29d93b : SUCCESS in 5m 37s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/69913747895348a686c1df5830aec3a6 : SUCCESS in 12m 03s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/420f68e4a27143f1b708dba9a6b194f5 : SUCCESS in 3m 58s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/33869660c0464b678350659f05cc2bb2 : SUCCESS in 6m 01s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4bbe6f3293824c40945fdef2b4e0c697 : SUCCESS in 9m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9ca4cac381e044ac8f638ac34c3b8a73 : SUCCESS in 14m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8a81b175b1cb43fd81066372255d86b3 : SUCCESS in 4m 36s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/682763b7376642eb93f10cb08cb769cd : SUCCESS in 7m 19s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ac7406045c1940a4aafc9ea103bdb30c : SUCCESS in 11m 36s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6cfcdd1f2ed2444e9aa344c91da680d4 : SUCCESS in 3m 15s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/1e22b4bc6e274881957717eb5e4183f7 : SUCCESS in 8m 50s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/35c0ca8028f14b5c9ccfbf809bfc9059 : SUCCESS in 3m 38s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/c05c596ea7454b9e8b8f169fdec44127 : SUCCESS in 3m 38s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e4be25fe14f14a5ba0a73b9ad71b0003 : SUCCESS in 11m 53s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/211388533f04481db90a86767d15096e : SUCCESS in 25m 09s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/529c64e255e34486b580b106bf90c261 : SUCCESS in 10m 17s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/65983feebb644ef097299242d7fb95c1 : SUCCESS in 15m 23s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/48e1e34a81f849378775a1dbafe8f515 : SUCCESS in 6m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/425bca0f65eb47808e061cb617734d6e : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/119d75b93d884ed1810dd2fe61010a34 : SUCCESS in 24m 14s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/38b68a560d10433f83e9616801a505e1 : SUCCESS in 6m 15s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e66c2cb771344600a5bdb625c8da0437 : SUCCESS in 6m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d690883576914c1da7411d7249854442 : SUCCESS in 5m 46s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/957a1f2b7db2492eb8d891da285e35cb : SUCCESS in 16m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d8653f2967e40c887a8a761c329d284 : SUCCESS in 1h 04m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6705c4f056084464aa392b605db0e7f5 : FAILURE in 1h 16m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a748638bbd95e7e20ca70f9425668e2e33460066","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-15 16:42:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"43a2106e02a06b27597a3639afa3bac9c9bd8a3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 18:16:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5785624954bd413a92f6845c821d1686\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/753fda44ac41404d9f2832fe0bfee496 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4db4d8d333c4efaa257b5c64d61b86d : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d3a2b1365694dfc806d670eb72fe57d : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cbb13830fa54495dbae005e8559564b1 : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/508928e321504287950632e3f502a4d8 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f7ad906a27d4f95bd350a24e72e2b89 : SUCCESS in 5m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7a3278a06e8349aba80049b7f60e373e : SUCCESS in 8m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8b851a777f2e4d3a8e5004ae060bda52 : SUCCESS in 11m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cb149a31292f45e28577ebe2c5022c7d : SUCCESS in 18m 34s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/f3947e4643f9400b90dcfd3c5e53b476 : SUCCESS in 22m 47s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/7f1de145f4af480590df44191b41de75 : SUCCESS in 5m 41s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/1175a00ed591409eb18c91dfcc8be8fb : SUCCESS in 6m 18s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/dbedf6dce1854b1b80767b722cf577ab : SUCCESS in 4m 05s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/8c7e12fb7f4a45ae91b9f5dfb9e579b5 : SUCCESS in 12m 05s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/47d1d30b93934dab84ab968ebc9617a0 : SUCCESS in 5m 33s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0382b7f34b2640798b831394ff8606c8 : SUCCESS in 7m 41s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/2023a3d12f414b459e2ce415597b3ed2 : SUCCESS in 8m 40s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/874fca93206c4ce280d30278f2636f66 : SUCCESS in 15m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2123368bb1664abeb3399cea0d623d41 : SUCCESS in 6m 33s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/f94812dac4994e488b5b00ceddc0356a : SUCCESS in 10m 09s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/16d44f8905a74f8491102fb6e5561b32 : SUCCESS in 11m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/59d9194f43c34675afe10851cfe9dd5e : SUCCESS in 2m 21s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/bdce6748c8b74d629a5a1b6217421d9e : SUCCESS in 10m 28s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/1727190d1d354ec39a4b1297a85d5c4e : SUCCESS in 5m 52s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/719f233fb5b947f38c00012039de9061 : SUCCESS in 5m 36s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e6fdc8f62c174b568b6bc751caaa630d : SUCCESS in 12m 11s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4aedb0202ca84376bed041e2c7ccab1b : SUCCESS in 21m 29s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/64237ccebb2a479994714cd6f54aaa07 : SUCCESS in 9m 33s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e6c0c0568c1545ae845eb83ec8a04977 : SUCCESS in 14m 07s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e2a2514959ab48829c06e84adc2cca36 : SUCCESS in 7m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aed26c6cd86b4eb8bcc06a6635a074f3 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/155123a4c8c841ab9e5f9cbd9d0a3b92 : SUCCESS in 28m 26s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/7733a7ba1a1c4c29bf3598d79f6b479d : SUCCESS in 5m 42s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/49ab5a2f18d54b05b7bb8ba3252fd232 : SUCCESS in 6m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/72f3600fc61246d2996e6e54323327ff : SUCCESS in 6m 39s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/ccba22bb505241f6985dcf886be648e7 : SUCCESS in 18m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/798987a748da49e6a4fee6267d528703 : SUCCESS in 1h 06m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83cc404defd5405db6ad984d5e733813 : SUCCESS in 1h 27m 29s","accounts_in_message":[],"_revision_number":1},{"id":"49061bb936707172a727bc5b98b98d728ef35bb2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-11-15 19:48:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3dd72d69ed29956cf3dadf0fb3815a37d0702430","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 19:49:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"96a9e997459d81bf01fb0e1d1c9592cfabcd4a61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 21:31:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4c100ec6f8a4754a2b173d23c1045d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46fb3a56a06a4af5beadd583c3585f6a : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a849a5c37a8841239f8cb6d65cf5b649 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b6278fdd8be4cf7985f03e0744b56b8 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac3aeca1368f49c3ae52f2623203d214 : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/da4cd0964717489599301a012692ad28 : SUCCESS in 2m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/698f142adf18420292a5f0827cabd9c1 : SUCCESS in 3m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c4b0a2a0c73e478c8e011d1db12c7815 : SUCCESS in 16m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b856f389bb9b409884407bde65d2b0c1 : SUCCESS in 8m 47s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/f3d87c32b0644b7d8d5fde60e58ec39d : SUCCESS in 17m 56s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b7ebd898df7d45aba915636071e2b926 : SUCCESS in 22m 43s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7713673696794780924fa38541f818ed : SUCCESS in 7m 55s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7e088440ec4043d9a73580705db82bd6 : SUCCESS in 5m 30s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c34b7a58c1fd438c8f81b0a8b3cdadab : SUCCESS in 5m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/fd1c68f445bd4a3c99474d663b34f1da : SUCCESS in 9m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/649b72a8350442dc939b894070368ac9 : SUCCESS in 11m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6b43b1ee9c4f4857bad642a95c6a26ac : SUCCESS in 9m 15s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/41c9bf95edff486a8925251e548efbc5 : SUCCESS in 5m 29s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/6e1b98746a83431ab6b2d4b1fc80c253 : SUCCESS in 12m 25s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/eb02f4b8219244cda76b215fd1abb3ab : SUCCESS in 2m 07s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3f61cef90372432d8cb6ba80c099ff97 : POST_FAILURE in 11m 57s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/12401a97f650454e9f759486dd33b909 : SUCCESS in 17m 55s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/db42add82cbd45258834380053d61d29 : SUCCESS in 10m 13s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/002c3be1e2524804a4f991d04412ce81 : SUCCESS in 8m 40s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e392aba5b1924c70bae6516080e84cc0 : SUCCESS in 7m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8154243638b74876bbe5fcc4ae5357c4 : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc4e3b5ca2454b28999f0c622b899f80 : SUCCESS in 21m 38s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2cd6398267e845648c45dae3875ab633 : SUCCESS in 8m 23s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4af82b51e51f49f38d64a5ea1d38651c : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2213020f4c26442f8f010a7d3c8d22d8 : SUCCESS in 7m 16s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9540913b554a4848a17d639cbf96c0cd : SUCCESS in 20m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebf8aacbcc19492e941524c427dff6c0 : SUCCESS in 1h 05m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5b2dbca7fdd4e2691808d4f497ca8f9 : SUCCESS in 1h 37m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ed76687fb3bfc735a4e7e91c7402ccf0338b5705","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-16 20:52:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a153998aa53b8b5f656365acfa59ae4e06d763d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 22:53:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29aac4fcec6a42569a38681cba668251\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cbd3235801d4ee08cb6728880e1e717 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e73b0f665274bbe87c2fb188b75f2b4 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08f612064bdb42c4b5ac9c2c637cc585 : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc3059ce176b4ff0b73832616b24c556 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0cbff84a24e742438d03c3b8eba51360 : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f50490800f6849fdb6de00359411fd63 : SUCCESS in 2m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2ae59511427c4a0498af52896d90aac0 : SUCCESS in 15m 16s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c90e370281684c8da60eaa5cea691f44 : SUCCESS in 6m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7261579aa50741278bbbbc397a5e64ab : SUCCESS in 16m 00s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0326284c83c44bc280ec8f33fa852ded : SUCCESS in 10m 23s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/95b5dbfca9784ba19f6af2a715d979c6 : SUCCESS in 5m 30s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/0f0040edcda7447fb743f8971891abda : SUCCESS in 4m 23s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b04ae1f048c8464fa268f9e88c8e4af8 : SUCCESS in 3m 59s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/87068226c6754a7bb6e040922210961f : SUCCESS in 9m 04s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/156b90527cc847d6bbe00f60dc543bba : SUCCESS in 5m 17s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/708a744f5822488d803941d7101adcdf : SUCCESS in 4m 33s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f52852148ead4a2488bc105f0a276e3b : SUCCESS in 7m 30s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/b2b03361e18a4208b77d9d4f117742c1 : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8bb23e304c7d4fd69522dfad235d8b5d : SUCCESS in 6m 09s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/470700b46d544839825eb47160e95b46 : SUCCESS in 6m 38s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/03834f8c4e6f48fa95a65aab2c90f04f : SUCCESS in 10m 17s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/6c50353446a84f4b8870e10f1aa8ff27 : SUCCESS in 3m 02s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f2f7d26576184866a6a096c332b761ed : SUCCESS in 8m 09s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/90409bf11334428891bb67c55e28c595 : SUCCESS in 6m 20s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/2dc52f62a6844c2198302eb213e8f81b : SUCCESS in 4m 57s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/5fd9a19780de45d984830daf375d1689 : SUCCESS in 11m 46s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c2406fe212d447128c6deac237f94fbd : SUCCESS in 16m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8d4f7228d99d432fbf8121cc1fa1414a : SUCCESS in 8m 46s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7caabae1f6214e1a9e90b654f9aa3e7f : SUCCESS in 10m 49s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e625332480764721bc42630e2f08ac9b : SUCCESS in 7m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e8100dba831e4a2e954ba4daa0bf767b : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b3e9f82825324ee486f22e7671dad19f : SUCCESS in 24m 37s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/0bb5ac6eb1444fa488cc15419fee84e6 : SUCCESS in 5m 06s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/16a9a1ad16a54f7e8654c10a3831b371 : SUCCESS in 4m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/48fe8c9671604c3b89e832a32c0627e6 : SUCCESS in 5m 51s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d300d98231a243ca9197800fa36a1b41 : SUCCESS in 12m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6afb043f39194034aa041e63c9ccc3a0 : SUCCESS in 42m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90e777d2e75547fe8143cec0878f1e5b : SUCCESS in 1h 54m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6c46be31a172b9a8694a23b01e6f6246cacab896","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 22:54:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c68e403171945b4a7ae3a8db0349f369da9dfa65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-17 00:49:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01155118cdad4b4dbbc972f64d03713d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92d36c4ad08143fbbe6bdf3bfb32cdec : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6996f6c4bd144e98b5367148cf72a66c : SUCCESS in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41dd21d6d09a4aefb8cfc59fdc41a305 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cacfabf0a62f48ac9c0541be0f1ed098 : SUCCESS in 3m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dd5bf9c5ac394fe2b3959a512ce5a663 : SUCCESS in 2m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/040bb3d40cfe4acf97357a6d27e7957a : SUCCESS in 2m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/639d45c5ac64436e99259c6ab677d3e6 : SUCCESS in 10m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/603a03f228894368af192f735dc0fe3a : SUCCESS in 6m 13s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/5f1da82af7954beeb04dc75fcafcff96 : SUCCESS in 9m 51s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/0b866b8ab2674d6e9112a73a0741b0a9 : SUCCESS in 15m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/892cfd1ccc904c21b2c8395363290b12 : SUCCESS in 9m 15s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/76b730fca16c492997a496424d6df4a1 : SUCCESS in 4m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8ace6f2d8d1f4f01a5ad09282ff0f67f : SUCCESS in 4m 20s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/459f2c9ba6f4450f875a30b0048302b6 : SUCCESS in 6m 16s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9a725fb89f3f4e358c90ff50e12b7a6f : SUCCESS in 14m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cc5392b6dc994834938b0f6442805891 : SUCCESS in 5m 51s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/945e0fcd750e4e65b02cb22099df1b0d : SUCCESS in 8m 16s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c12298360f904df890e6472270a0ee3e : SUCCESS in 9m 08s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/dcf4e387d17d43558ec156cabce9cb59 : SUCCESS in 2m 01s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3b311cb2875c4e57bb0057d8efb917c9 : SUCCESS in 10m 49s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/b1f26576e140459d851914f4873b7e3d : SUCCESS in 14m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0dbca17d42d54b7fb7027f255aa635c8 : SUCCESS in 8m 24s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/3066e564dd074c6197658e8445dd1bdb : SUCCESS in 8m 57s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/51d1d8f04d2644e9821535530deac48b : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f0ef1249cf1143c49a6de160a72f2253 : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/803e1f2f009c41f58a8844a8f3dbb71e : SUCCESS in 18m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2c648d118de7403fac7b1a29507eded2 : SUCCESS in 5m 44s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/aa6957138ca740d3a9af33da9f375220 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d4ac2738871148eeaa37ee45f8dd8d71 : SUCCESS in 4m 00s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/8178f57ce00e47ea9fad6cd93cf3b695 : SUCCESS in 11m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/065aee9fced345ffb2e02c860a953aef : SUCCESS in 1h 00m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5f334c2bd4d44a2afede98e0e22ed9d : SUCCESS in 1h 43m 58s","accounts_in_message":[],"_revision_number":1},{"id":"1ae4e5018fa3c8b36c0259d8bc57d2a0caa911bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-17 00:49:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e5d0c3399e922dc26c1d0efe99c7f106a88e6b01","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-17 00:50:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4076e2b72fe4c9197bb80dec18e5c85\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82c446cd245a46c3823e816a956aca5f : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f3b2c324ac7b4a3dfe920165a2eb91d60843763","revisions":{"7f3b2c324ac7b4a3dfe920165a2eb91d60843763":{"kind":"REWORK","_number":1,"created":"2024-11-14 09:32:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/935048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/935048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/935048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/935048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/935048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/935048/1"}}},"commit":{"parents":[{"commit":"1daffb38e8d91ac33de84c850ba406bcec7a8aaa","subject":"Merge \"Bump importlib-metadata to version 8.0.0 (\u003e\u003dpy3.10)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1daffb38e8d91ac33de84c850ba406bcec7a8aaa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-14 09:32:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-14 09:32:33.000000000","tz":0},"subject":"update constraint for oslo.concurrency to new release 6.2.0","message":"update constraint for oslo.concurrency to new release 6.2.0\n\nmeta: version: 6.2.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I721a032eb3d4dff2a86ba9b0802ad4eca356712e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I9a2aa6ee4148fd4fb4bde957b1457dcffd28cca4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7f3b2c324ac7b4a3dfe920165a2eb91d60843763"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7f3b2c324ac7b4a3dfe920165a2eb91d60843763"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
