)]}'
{"id":"openstack%2Frequirements~935050","triplet_id":"openstack%2Frequirements~master~I71b84e55f4174396c6377e19cc2c15bb348927ef","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I71b84e55f4174396c6377e19cc2c15bb348927ef","subject":"update constraint for tosca-parser to new release 2.12.0","status":"MERGED","created":"2024-11-14 09:33:01.000000000","updated":"2024-11-19 17:14:21.000000000","submitted":"2024-11-19 17:12:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935050-new-release","meta_rev_id":"79e5bcebb0bca6c67d32fb4a37f9c70f0b677e9c","_number":935050,"virtual_id_number":935050,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-19 17:12:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-11-15 19:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-11-15 19:48:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-14 11:17:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-15 16:41:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-11-15 19:48:34.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"010656692d5f92fea91d4b64a5f9735b50bb6104","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-11-14 09:33:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"023ae0027ad037b883c68e5c6cb2fff52bf46e12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 11:17:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/163c69a7ff0d473bb95651be5f9a48a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9f4aee8abf945dcb79c9eb8c3be3d9a : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44025ffcd3074023b3c5015d046de4ff : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c76e11bfa9794b12a7da67cdbf6e7d20 : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d0d122c13ce44cf5b22447c22648798e : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/786d8c29ede5450097562a0c1480371d : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f0ccfedc8f8840fd82c0540bdf0a7c1a : SUCCESS in 2m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5fe1aeb770f742e2b2dd7b9f4df59a3e : SUCCESS in 15m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/92d7e7a8dedd44aa9a48bbb2e019be0c : SUCCESS in 8m 46s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ca1c0249463c4146b5333d2f9224df09 : SUCCESS in 9m 23s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3e32868151ae43e3a3ead47f06952a46 : SUCCESS in 12m 05s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/a2bc435b9296441695d951fdb44a69ac : SUCCESS in 3m 45s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5cd703217ae9441298b004e5e025a10b : SUCCESS in 7m 23s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/27116cc284d246aaa8b78022e3b03fae : SUCCESS in 5m 39s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c6a4fdc7f99f41bb84cb0246fc198965 : SUCCESS in 13m 42s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b148b7e3321749dc964b2d187722f808 : SUCCESS in 6m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/4d81a825a3bc402ca00f15776b058664 : SUCCESS in 7m 29s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7b39714b4afc48309c915534d6c48144 : SUCCESS in 8m 28s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9276b2944a624f93b624bd5bfb306fb8 : SUCCESS in 17m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/243bcce8744f4b7fb6e0712841f01bef : SUCCESS in 7m 37s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/cb463b6b98584d7b80d229b6b15c495d : SUCCESS in 10m 22s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/9310e88612934e6b87e18721b3191747 : SUCCESS in 13m 47s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2a05373e3af849699e6c4a189bfe2eb2 : SUCCESS in 2m 17s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6486f1e0635043a48ecf27dd1a3176c3 : SUCCESS in 8m 44s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/637777929ded46b0870549b286d6a631 : SUCCESS in 5m 47s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e79ad1be3f424d6bb3fd5b3d17b6ae7c : SUCCESS in 5m 30s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/776dc9ff367644438d9c091e10e1ed43 : TIMED_OUT in 40m 51s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/69783cd51c8d424aaa790971a26c7a29 : SUCCESS in 22m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8d9eb346d01e49bfae80fceeead86860 : SUCCESS in 7m 41s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/f3071bf3b1f94a99965616d8cae8bf5d : SUCCESS in 14m 04s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/2372ef21e89f4adbb4a04e242d3cfaa0 : SUCCESS in 5m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/17426abaa11541a0af105c01c2495057 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5a8125ff05de4b40b55a2e47ae5e7906 : SUCCESS in 23m 32s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b4ddeb83ad3b44fe92c9aa26d11f90eb : SUCCESS in 5m 32s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6156bade073542ec9e01910cba6bf60c : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5a86bdad00a44684955e6313012fc0b6 : SUCCESS in 6m 15s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/90dc9f052e8f42d68b9f651524f33070 : SUCCESS in 13m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/441437ce9ade4d679ce3aab216a7c83c : FAILURE in 32m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82473ccd305b4d69aadde6893e84fd78 : SUCCESS in 1h 31m 02s","accounts_in_message":[],"_revision_number":1},{"id":"06447c8b7880cadb0008c5582b259e7b06219e45","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-15 16:41:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d6123126b30245a9046b1398ad2b3f3c3fb8bfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 18:06:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7553219fc1c84499820e5e3d1ded801a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3eaa985f5741aca0cccffdbf1e3294 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/329f1f0a7b454b6bbaeffb47f553a185 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24859037325e4ff18bc1a940f6d8c0f7 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5bfb2db51dc344e193ba7c1c96c46594 : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c344ef8cef4a4a9496f27af9339cab48 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/04e19dd7fe594848bef50102d57d414c : SUCCESS in 3m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fbcea940c8d442d8b2bdb56f0f645d0e : SUCCESS in 13m 27s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/4865b472c8904af58158cfdd7d140058 : SUCCESS in 11m 55s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9bef452f10ab460b9438502eb0111bf8 : SUCCESS in 17m 35s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/81dcf5425655400ea5642e17f18a11ba : SUCCESS in 25m 14s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/376aa0976de74dbaa10520e1609fea5b : SUCCESS in 7m 34s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a00be9ebd7494caeaf44246e40c970f2 : SUCCESS in 5m 49s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/9885aa4d635b4b70b185ede3d620b3ba : SUCCESS in 3m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/336e68fbd196453dbb33701a21e3a898 : SUCCESS in 13m 25s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4951c6d5ced44e5ca72c6e71898715c1 : SUCCESS in 7m 46s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/de99fe2ceb82475595b9bf38af467fa3 : SUCCESS in 5m 21s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/77e981406ed74ffe881b52b5158b2ca8 : SUCCESS in 5m 42s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8a97e2a79a1f49ae824d1377345b40a6 : SUCCESS in 14m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/78667609b8704157b1d422f5be3c37a9 : SUCCESS in 6m 28s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c76d500b60a5491d9a244f95bc56c2ba : SUCCESS in 8m 08s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/227e639610a84a1d91fd7f93814c4385 : SUCCESS in 11m 12s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/93dedd9bed294d10b81d817ab4ce6861 : SUCCESS in 2m 13s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/46479ba0b399471dad67298e764c756f : SUCCESS in 10m 25s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/0b2e9c91001b4c56a2ec39ee6925885b : SUCCESS in 4m 28s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/5431f30ba22a44ed94e03b6b3e95dc01 : SUCCESS in 6m 32s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/1230a6445411422a93139358cd0e9308 : SUCCESS in 11m 46s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/691b21b5c353476d85aada5d10a3259b : SUCCESS in 19m 06s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f28176dbbbba4ad098722fb7b4c2ba68 : SUCCESS in 8m 03s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c918569c15924663aa8cfbf05beb77f7 : SUCCESS in 12m 20s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/72666be5b02c444691bec2aee02b6839 : SUCCESS in 6m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7588f721726748599853297334040bd5 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8bbe294454684acdbe9235f70df0da4f : SUCCESS in 25m 15s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/26c172d66cfd4940ad8b3174e6b0db12 : SUCCESS in 5m 24s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a97664446aee479f9ffe7e83a994c2da : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3f5411c6093a48c9af2224c986dfffe2 : SUCCESS in 6m 02s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/53e07d6043b0423ab2565cb19b2e6c02 : SUCCESS in 15m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/701204c3b738456b9be9510e232c239d : SUCCESS in 1h 00m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dbefb025cd4c40f287f09f861299ce61 : SUCCESS in 1h 20m 53s","accounts_in_message":[],"_revision_number":1},{"id":"bd7c81a5c5e10459056f2363690be5671b8b8035","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-11-15 19:48:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4795d10b92e2fe85e44e34134eba900854359f61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 19:49:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e9b446bdf3082ad7dcdafdf0fb0b10c4acabb13e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 21:44:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8fa1824a5a643768de341b296606fba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/716e3cbae42f4ebd9cb4318d60240365 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecab2669146a4c78a25504be2dfcaa68 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6379158d7ea4474da8f642f309636059 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d539d4b71134b369a259482e07e66e4 : SUCCESS in 2m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/58ed0f35073845aa968d56f94478013d : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c7511dd5d114467ab038ebcf26743ae : SUCCESS in 4m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c97282cfcb444ce3bd97feee4ba7bc78 : SUCCESS in 10m 52s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/32d9d6bbeaab4b608a1b4f938702b09b : SUCCESS in 8m 58s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/869d2dbca53349e8a0924f9c9258f03b : SUCCESS in 16m 44s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ff55e9b648de464c9887a4d3fcf86762 : SUCCESS in 22m 18s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a96dfe720d8a4645ab155f805370f7cd : SUCCESS in 11m 29s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/4e1026c47a6041d28fc7f510abf23f67 : SUCCESS in 5m 19s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/cb355a04782a4c5781b6a5be23fa4978 : SUCCESS in 5m 04s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/777e74b795c2495199a9f3dfbac70bec : SUCCESS in 7m 26s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0434dc92b5384fa09b593ec93c69c5e7 : SUCCESS in 13m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9cafb02f7a3a4905b8c9bd2818fd7be2 : SUCCESS in 6m 18s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/bb7e939170f54f578b925072f331bcc2 : SUCCESS in 8m 57s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e2df72115d2a40cfa7ffd257d982fbd8 : SUCCESS in 12m 29s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/87221e71335948aabad61f92c6901c63 : SUCCESS in 1m 58s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/7e919e176cdb47f3b76f1bd894af8601 : SUCCESS in 10m 05s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/13e7a4a3ee3f42659447cc1cfa980dd0 : SUCCESS in 21m 34s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/388d18095af6482e8a361546b6ffa17f : SUCCESS in 6m 20s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/8080c9ebe5ec4608899c4aaac7792372 : SUCCESS in 10m 19s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/915518eadc9d4ff7875539af04948eb6 : SUCCESS in 5m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ad46b13e3eeb49c39f49e95cf58af01a : SUCCESS in 2m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6cabe42bc4d5472887bc1a3f936171ad : SUCCESS in 24m 50s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a513763da27046838c753b8488138298 : SUCCESS in 5m 46s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/918e7a72b70447e2a8298078eff2698b : SUCCESS in 5m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/24790204716e4b85939cb731b2dfa777 : SUCCESS in 5m 13s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/90ffae6c54164c93b95b98b76219a6cc : SUCCESS in 12m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7f8161b8fdd42f09aa8b0572da645e2 : FAILURE in 37m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31cbe994605e4918bcfc6c9be7e49c48 : FAILURE in 1h 27m 47s","accounts_in_message":[],"_revision_number":1},{"id":"3e4553020e82bb3d5c4e4cd5e6332eb7146310e1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-16 20:51:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c286f8b97770d3a31c81baea50471e4fc89c0966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 22:22:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33041d51492c4dc9ab91a11026d6cf1b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7e173a0781e4e8c844363dd5f98e6d2 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dab1fef67baf4b15ad4e239260b08cad : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb1ec86badac4ea9bc73dd516d0c7b8a : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/996f7ac330bf43e68df87a37c6db3f1a : SUCCESS in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae174af77f614ab791223128d893242a : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/911ac0d9ccd448e5aa0f5c09c49b9ab8 : SUCCESS in 2m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/930f909895594597b0e769b50aa46f17 : SUCCESS in 15m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/b154eaa230a74a46b436964f552ac957 : SUCCESS in 8m 25s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c9712da07d7e44938a32a33d9ae854df : SUCCESS in 16m 39s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a45002062e5c4201bca1216093b12141 : SUCCESS in 23m 10s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/93045693b9f04b05a68ff2cfd68d983f : SUCCESS in 5m 42s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b5c7ff853782463d9c5247c396d8a33e : SUCCESS in 5m 36s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ddb23e43a2b24231bc15ca8f33f202b6 : SUCCESS in 3m 06s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e2c124a67e304d59a7b823dddc5afa0b : SUCCESS in 9m 43s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b5a849f93db44057b59e29d9aa4a524f : SUCCESS in 3m 42s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/1f23e8fef6974d57aacee98230f49590 : SUCCESS in 3m 46s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f474381c4ca94447932be8a0fec620d0 : SUCCESS in 6m 11s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4ceb68d5fe234812a633289297ea5bb8 : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/22f0fa45f5ff42348a3856a0a87d03a0 : SUCCESS in 5m 42s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4daf02060ccf46788a8788fd11a6df97 : SUCCESS in 5m 57s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ff6c7e22abc1457494e4119fd7fee9a4 : SUCCESS in 12m 10s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/4cee7f331ff644d79f8aeab576eec101 : SUCCESS in 1m 55s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/36e8cfc99ea641a3a2a6ff429bb23c32 : SUCCESS in 9m 22s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/7b52641e7b9e4d259e866d7fc1ccdd02 : SUCCESS in 4m 31s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/29e969e16d514dd2bef6fa17dff0ac73 : SUCCESS in 4m 51s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/301acc3a01cf4485bf6251d07a40b29a : SUCCESS in 11m 23s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/76f1a896c16b4475ae252cc39dbde61e : TIMED_OUT in 1h 00m 45s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/ae4ed71201a64eacbc42609b819ab370 : SUCCESS in 7m 57s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/97058633a8d6462189c7b5d6a0e30720 : SUCCESS in 13m 59s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/8d70fe800b5a4520945986f3b110b63c : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/897893473614473180a60857b787b686 : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/be6d6e3f26234c7193c6d9528bd60c43 : SUCCESS in 19m 12s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c9de7b6b0f2f4bf1a7858369ffc961a4 : SUCCESS in 5m 40s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/595462cf212049599c38cd5e40782b2c : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d17554e5045446e2afe2551e8057d0e8 : SUCCESS in 4m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/09e4be01886047e5bd14035eb7e7f4e4 : SUCCESS in 13m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/466494979ee34ed4b890186152891bdd : SUCCESS in 1h 00m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa31673dc29c44be8ab16375a7aca86f : SUCCESS in 1h 29m 50s","accounts_in_message":[],"_revision_number":1},{"id":"48354e26e2549e7b444a587bb8d697eec7e3c1d2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-19 13:26:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"338c4cfcfaf848f2c3432eef26619e49cf59be93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 15:14:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c179d4afa26448cdb4ed1f50789277c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4ebc6551343425f86cbc63ba69250bb : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39f4d52a5f824a4c8c83ecc90311b5ca : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/581e0bea8f00437694ab6130de45a436 : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f2cc0f407824d1788e1150bea77d819 : SUCCESS in 3m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/60607da4f01d4e988a4923e38d4ff901 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1643835694164e9186d3b9771474d846 : SUCCESS in 2m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/143de1b0ddfd44d58f015b61a38ff7bc : SUCCESS in 14m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5fa8e1f34708477f858e3f64417a6287 : SUCCESS in 9m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8bfb7cd7c6fb4096a8c042e13bd16430 : SUCCESS in 21m 20s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3b81913f6c5d443ba36c95dd3664b2f5 : SUCCESS in 22m 57s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/2a2a9fd9d867454c9c22213ca9b34db0 : SUCCESS in 8m 14s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/e1810fbcac664b9ebad139783efc7179 : SUCCESS in 5m 22s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/7e42a7c3122e4407a7d09a6d18f0e68a : SUCCESS in 4m 24s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/35b4dfe0cc3c4e518ff61e2a67f2d60c : SUCCESS in 13m 37s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/516b36a2edcf4603a9325bb8b7d3b035 : SUCCESS in 5m 22s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/df4afd1268a1411ab216bc8d075b2b47 : SUCCESS in 5m 53s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a1853123d4b54dabb53e4c092202d9a5 : SUCCESS in 8m 06s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0a80e4dab2d64a588b6fa79456b9b2f1 : SUCCESS in 14m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c2884576ca714f9d8f2b1ae6aab2a953 : SUCCESS in 8m 33s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/526ae8a507644ca384badd2058015783 : SUCCESS in 7m 35s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f6dd3730a7884379a6e756d3544af1b9 : SUCCESS in 12m 33s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e73995f4a819408f8412e1f222a57925 : SUCCESS in 2m 33s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/9f0edc4a3e3e485d8f3dbaa6635323b8 : SUCCESS in 9m 38s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f90bcf30e87647b09f7c63ad10d424e3 : SUCCESS in 5m 37s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e62a9f139d4d438ab5d7b7099a5b8441 : SUCCESS in 5m 07s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7e6a3ed184894aa0a53492fb4b119817 : SUCCESS in 14m 00s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6af945124da64eff931318390f67075a : SUCCESS in 24m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a19ec070c83445bf94f1349e09471f18 : SUCCESS in 7m 52s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ee6c23f52d124b4f9332e5e92b35af64 : SUCCESS in 13m 21s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e453090ed98847d88f706e676a0f6870 : SUCCESS in 6m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3555ac7d928046fea7347d862a50372b : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/376b094e5fc74f20aa27a3364635924b : SUCCESS in 29m 29s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/637e4e7c3a3845f287fda59f9affb322 : SUCCESS in 6m 20s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0bec2776b7854286bbfaa610f39c41c7 : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4433f8b4f6434fa99a4166d6098df29f : SUCCESS in 5m 28s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/79eb90a7f1ad4f828d958a4fc282bacc : SUCCESS in 19m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a81b3f3f08b415eb2f0734c304e937d : SUCCESS in 1h 02m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14aa82e16c9c4134a3eef10b0b31e38b : SUCCESS in 1h 36m 12s","accounts_in_message":[],"_revision_number":1},{"id":"004dd1868f80fb9941866d403e4dd3c432c5ba9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 15:15:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4cc1732047a87fea946e075da38f67e5378f350c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 17:12:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d77c12a75454f87a2e633508fcf71c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85e295ccc2144ca2b03263ba0021424a : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e174c4b9875a4ba5b5ec270210fc60cc : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/726c58da10e4423698e43aa0196fec3c : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fcfa5fd8bcb04526b340e41407053f2e : SUCCESS in 5m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c7b820183ff646879689c7cffcf1b1d8 : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c181a897ef784a279b55b05219955362 : SUCCESS in 4m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/18837547dc8d44a2be4574887836ad39 : SUCCESS in 14m 38s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f276250df493474a88d7d788ccbfdc24 : SUCCESS in 9m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7688c75f563544f4a3dba25991442c2c : SUCCESS in 22m 33s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6697eafb1950407589e2d55cc4cd1642 : SUCCESS in 23m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/57235f2d795543858564fd2b47c7ce99 : SUCCESS in 13m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/98ba0100101d4bbeac0bcff2a25707fc : SUCCESS in 6m 52s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/9dd00824730c4161a4b9ac76ef6c1e5c : SUCCESS in 7m 11s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0e2aede9fafb4c3cb8388a750c9ef764 : SUCCESS in 8m 52s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9d9deac798a54ed1b009ae5b5394b77c : SUCCESS in 14m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dfdfa1146e8e4637b2858b790d5af718 : SUCCESS in 6m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b4a66ff0af2d4805bb73e82a1feb8671 : SUCCESS in 8m 58s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c6bb9355ebfd4b93a230026f69fa4c10 : SUCCESS in 13m 45s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8361139aa5374a44ae521b1e1f4592be : SUCCESS in 2m 34s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4efa4596e46b42eab328b39b0e57ab62 : SUCCESS in 12m 37s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/e0e0d0c18ef1457195a747813c291d45 : SUCCESS in 20m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4a1549bef3764183b28bbbe25ecda3ab : SUCCESS in 8m 38s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/2f9f3a08f7dd4d1092a40e58e7cc13dc : SUCCESS in 15m 51s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7236abb4e6b144eab0ed107d908af17d : SUCCESS in 6m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9bfc5ad8b6a8487c9e793eebc9dc6b5d : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a15394312d514dbeb92177288cd5b625 : SUCCESS in 23m 41s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/574585be05f04194baf7aaadad0860ab : SUCCESS in 7m 22s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/79b4dc2a927444b886088f86bcdef9ef : SUCCESS in 7m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dd31b3bc945d446f9f1e332fb33cecd4 : SUCCESS in 7m 20s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/65d41201cf8a4bcbbb05ae4e1963f4ff : SUCCESS in 13m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7aac38b7d344c47b5dcfeda5d91b552 : SUCCESS in 1h 04m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f320f04c59f344ba9db055766c139ebf : SUCCESS in 1h 22m 54s","accounts_in_message":[],"_revision_number":1},{"id":"44d8229621eaa81b360accfb61808efbab7a92d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 17:12:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"79e5bcebb0bca6c67d32fb4a37f9c70f0b677e9c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 17:14:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a297f38b7e2342efbc8d0a80920b67cf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99042a5315e04c4db26af99d815c7f24 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b4ed999323b0c179974c850797dc3a0473deb7c2","revisions":{"b4ed999323b0c179974c850797dc3a0473deb7c2":{"kind":"REWORK","_number":1,"created":"2024-11-14 09:33:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/935050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/935050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/935050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/935050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/935050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/935050/1"}}},"commit":{"parents":[{"commit":"1daffb38e8d91ac33de84c850ba406bcec7a8aaa","subject":"Merge \"Bump importlib-metadata to version 8.0.0 (\u003e\u003dpy3.10)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1daffb38e8d91ac33de84c850ba406bcec7a8aaa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-14 09:32:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-14 09:32:58.000000000","tz":0},"subject":"update constraint for tosca-parser to new release 2.12.0","message":"update constraint for tosca-parser to new release 2.12.0\n\nmeta: version: 2.12.0\nmeta: team: tacker\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I466624167ab4bd9b3c6c49881cc8958a7535f775\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Yasufumi Ogawa \u003cyasufum.o@gmail.com\u003e\nChange-Id: I71b84e55f4174396c6377e19cc2c15bb348927ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b4ed999323b0c179974c850797dc3a0473deb7c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b4ed999323b0c179974c850797dc3a0473deb7c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
