)]}'
{"id":"openstack%2Frequirements~935051","triplet_id":"openstack%2Frequirements~master~I31b0a208b842af65599b07428880ed7f21f3872d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I31b0a208b842af65599b07428880ed7f21f3872d","subject":"update constraint for python-watcherclient to new release 4.6.0","status":"MERGED","created":"2024-11-14 09:34:45.000000000","updated":"2024-11-14 20:36:25.000000000","submitted":"2024-11-14 20:35:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935051-new-release","meta_rev_id":"e86070330cc1128d7d2b05094ce5b9f12d0916da","_number":935051,"virtual_id_number":935051,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-14 20:35:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-11-14 17:38:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-11-14 17:38:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-14 11:11:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-14 17:38:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"59721a49491966233d08b536c9742bfca23ca41c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-11-14 09:34:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eeca57c2bf3c6dc3e592466546bc57693209d31a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 11:11:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fe55ec31438440b9028c4dfac67d192\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f940d084b464d03a8a84c68558006ba : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/581542a5bbe34a7188d19b8f5afdc6bc : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c8f7e2c35004920aad4f55c00ecce2e : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe3e767fea3049cb8f2c072bf802a032 : SUCCESS in 3m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3abded7c471441f6b8a8357963e19564 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ffa1ae8eaf9949d5932f396ff42e9770 : SUCCESS in 2m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/385940cdb76446ee8ead1de92b74f4c1 : SUCCESS in 16m 47s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/79b432d3836a454d9792b8379669d1ae : SUCCESS in 10m 32s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/8dd5f92bc290498da1a1d14c6261a10e : SUCCESS in 16m 26s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/70b653a2cecd4b27a881546540a45eff : SUCCESS in 30m 11s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/4a62e1a4b7444b35bf29154385303beb : SUCCESS in 7m 19s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/f8d656147f6740bf8fae1146ecbb342c : SUCCESS in 7m 07s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/82645897558445acaf43eee21db14cb5 : SUCCESS in 5m 50s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/5c921612b0924a46849a9f3f79b9f33d : SUCCESS in 11m 43s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/5638a200cf3e4355b198d4dc0ae44a4e : SUCCESS in 6m 18s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/26002c97ca494a9bbca052212b62c9aa : SUCCESS in 7m 25s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/d45d7073333046f1af1214c0b2fa6689 : SUCCESS in 8m 48s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6161fdfc4c28447282594060d44750d1 : SUCCESS in 15m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/97de0fb6da48405a8aa8bddee212ef07 : SUCCESS in 5m 40s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0084b5d21c734dbea530de2eb268e06c : SUCCESS in 10m 50s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e1f7bee8f4fd469eabeffc828438cd4c : SUCCESS in 11m 11s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/fe8fbdcce5d54db695bcf508bc68d782 : SUCCESS in 1m 50s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/17fdff0b05a045109887fdbe71c80429 : SUCCESS in 9m 15s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/1ecb2a9823fa4ddeb2983ced32e65cb0 : SUCCESS in 5m 52s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/090b95e264b1404b93ade1e36db59ac0 : SUCCESS in 4m 54s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/5ca38cdb6b8a430098aec2a6471103da : SUCCESS in 12m 57s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/cefcd273d63949318fffb8f62f80ba63 : SUCCESS in 18m 27s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/5bf32f00c58a4ed48f7fa2f8536966ea : SUCCESS in 7m 53s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/35fdc1eea1e84f1482d29d4016208335 : SUCCESS in 15m 08s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9a89684583f04dfdb65e7b1a9478e9b6 : SUCCESS in 5m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/120964728be14908b19ded96d7b87ed8 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a8f73e03982d4c20829c206170a273ab : SUCCESS in 27m 54s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/00a5603e7ad348e88ff827cc75b25626 : SUCCESS in 5m 48s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5de7a8f032f143c7881637c99fce2625 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/93831d792ba048d5828d1901284fbbd4 : SUCCESS in 5m 16s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/8ab3d7cbd0e44473a0243a29c151da4f : SUCCESS in 15m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/edff32067a4b4abda33454e7e8bc7c9e : SUCCESS in 1h 02m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c67a8a7271f049a9af403f332a474ccc : SUCCESS in 1h 30m 44s","accounts_in_message":[],"_revision_number":1},{"id":"0090c373f33edaf71a173ce64d9cfa3c48bbdb77","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-11-14 17:38:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dd7dbb965b2c81da612ac2db94d1d398ead74aa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 17:40:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f202a3726a831bac91b60ae30452c474cf2f123","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 20:35:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/850aae8258184869bc2a8c407957b05f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c0ff1a42c724c669434f9ff0e27fff1 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5d55fe99abf48dcb3abd3ce7c02e0c3 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e7b6aa843434173971825fff2e4887e : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f3313d14823a48e2aba0e5676778d4c8 : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffcfc1f5c01f44ada2171df821d201d2 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1ae045849bac43a285868d756520a89f : SUCCESS in 4m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a59ab0d65bbc4f2c88fa8bae595732c8 : SUCCESS in 17m 26s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f418030010ac465f8bc3143b12f52c07 : SUCCESS in 9m 38s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/01657e74430548378ff2bb3322f3ef3d : SUCCESS in 9m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/d5d720df34fc4312888e0e43f8ea03b5 : SUCCESS in 27m 52s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2d88f63a280c4333bf471bd9976e9c94 : SUCCESS in 9m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ea6ab94ada0d4ee1ad86f22e70f1e404 : SUCCESS in 6m 02s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b446f2c3bf7d4e2fb62eea6725bbb65c : SUCCESS in 6m 00s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/27edcd78c23040bcb8ff56d8fad3fb00 : SUCCESS in 6m 39s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2ff46e9c852c436b83cdd5d076027c04 : SUCCESS in 11m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b0c8ff5bc25742cbb701bfed05b6c59f : SUCCESS in 7m 36s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/060437c128c5476fb3db8e06a783445c : SUCCESS in 10m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/dc0b76781dd6407cac43b952228c7979 : SUCCESS in 14m 44s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/60f864a63fb940de89910ea107722f92 : SUCCESS in 2m 42s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/20da90f8913a408a940a5da65f7fee92 : SUCCESS in 11m 40s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/fdbdb459f64a47aaa802602095c2d437 : SUCCESS in 22m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/dcdf53d1152e4bd9a7605adedbc6a6c5 : SUCCESS in 11m 04s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/da685d6a15d44a29b65778216108972b : SUCCESS in 14m 11s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a57edede5ccb44a2922c5a746b1cd3be : SUCCESS in 8m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cde6d6c7c0a844dd87d01a98bbe4540f : SUCCESS in 5m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c8e8924de7124f77969423b50e51ab42 : SUCCESS in 24m 42s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/dc4279bf5f914d7aa65560f4a1a4158b : SUCCESS in 6m 59s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/8007b5e3bef2495091979889e203e1db : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b9af5d7a9864496e95599615b3563fa6 : SUCCESS in 5m 43s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/12f4ba36412a4f858773c3a1afc866a3 : SUCCESS in 19m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9dcfe1cded284befb2e33aeb3188566c : SUCCESS in 49m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10d910081bb64025ab84d7360d93a46d : SUCCESS in 1h 27m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f93fd833dced39c1e209dc44892dae99bd7037d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 20:35:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e86070330cc1128d7d2b05094ce5b9f12d0916da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 20:36:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02f5fdb581744ce8bdf9bfdda32f1bf2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82777ff28c2c4c1db636d84b96178d0e : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e30d05d25827e41b549e853fa3602b418828d3e","revisions":{"2e30d05d25827e41b549e853fa3602b418828d3e":{"kind":"REWORK","_number":1,"created":"2024-11-14 09:34:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/51/935051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/935051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/935051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/935051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/935051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/935051/1"}}},"commit":{"parents":[{"commit":"1daffb38e8d91ac33de84c850ba406bcec7a8aaa","subject":"Merge \"Bump importlib-metadata to version 8.0.0 (\u003e\u003dpy3.10)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1daffb38e8d91ac33de84c850ba406bcec7a8aaa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-14 09:34:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-14 09:34:44.000000000","tz":0},"subject":"update constraint for python-watcherclient to new release 4.6.0","message":"update constraint for python-watcherclient to new release 4.6.0\n\nmeta: version: 4.6.0\nmeta: team: watcher\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I3ba31dd5c53c5ad65d6b40c546c33b0d44642d33\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: sean mooney \u003csmooney@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I31b0a208b842af65599b07428880ed7f21f3872d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e30d05d25827e41b549e853fa3602b418828d3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e30d05d25827e41b549e853fa3602b418828d3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
