)]}'
{"id":"openstack%2Frequirements~935056","triplet_id":"openstack%2Frequirements~master~I6e3f3b764500d2f7a57bc4be59f3aa86469540cb","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6e3f3b764500d2f7a57bc4be59f3aa86469540cb","subject":"update constraint for taskflow to new release 5.10.0","status":"MERGED","created":"2024-11-14 09:59:32.000000000","updated":"2024-11-14 21:25:42.000000000","submitted":"2024-11-14 21:23:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935056-new-release","meta_rev_id":"fa595e64aeb2dd4831bfec585d09f341e5a18451","_number":935056,"virtual_id_number":935056,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-14 21:23:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-11-14 17:38:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-11-14 17:38:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-14 11:27:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-14 17:38:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"fe264182baa5e4c49c8e849510aa5440a3d29a42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-11-14 09:59:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99daf49e3511c29187ea36623e84ad7a30cd2fd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 11:27:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c5ceb15a06549c8a13087bbe8d89b5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/936120e501254d5693c71df10e3c3787 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fab8563a12aa420186f830e22199b75e : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e7ec21a6a374bffa75c50f1ee0c2cce : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64989fb8f7cc422cadbef1e4b7dc4d87 : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fcde6245e0e14db6a66ac9ef565ec5e6 : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/00202cdcb3ed4c8382bbe6df530fae61 : SUCCESS in 4m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8a470157a7a147c1837c1a9c2fbb8d28 : SUCCESS in 14m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/5c5bf7e57e274a46897e5cb82908731a : SUCCESS in 9m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/12bdbdb14da9480abe7c1ead4368d61e : SUCCESS in 16m 05s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/570f69767e004d50b8e71441502e1cf7 : SUCCESS in 21m 23s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ed72884927964dab8e063202bf0de902 : SUCCESS in 5m 13s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/ba14f26a263142729e9c440a6b84b4a5 : SUCCESS in 7m 12s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b536b61746364eec814bc11765dbe3e0 : SUCCESS in 3m 57s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c2b42af58734436798677df0774c6ffa : SUCCESS in 12m 47s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/21750d8e4c6e41bbb1a26c49f1d0ef20 : SUCCESS in 5m 23s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/d837362a0a2c40c2a35ebc7f0c4c0bec : SUCCESS in 5m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/43028e4a8f9c48d0a8f6ce477a087d3e : SUCCESS in 8m 28s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/9e2e4267e0e14bdd9a38e74a52640cab : SUCCESS in 15m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d4516da6572c4d48bfcff8883a27f2a7 : SUCCESS in 7m 33s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c3b68c4495b64a88b897427e68978fdf : SUCCESS in 9m 53s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f9c62de21f6c423b90991725f9c0fc05 : SUCCESS in 14m 39s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/cb6702df78274f0d92e862f91cff375d : SUCCESS in 2m 41s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/dff1107310f74e2595ccf4ab19452160 : SUCCESS in 10m 50s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6f9453e933e24120b4b0bd4eae5f1efc : SUCCESS in 5m 25s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/0c7431543dcb4f9587d7365744ca0b54 : SUCCESS in 5m 16s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/16848a2c6572419cb986786efa22525f : SUCCESS in 9m 02s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/30067a1edd2d4461a5995b7fd86f127b : SUCCESS in 22m 30s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/8368b8a77c98475f9654ce0bf632b73e : SUCCESS in 8m 34s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/eae39faab2504cbc95656c591d8b8727 : SUCCESS in 13m 33s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5921a9f8522e4c889b73935ab4d146d5 : SUCCESS in 4m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e9d48ec2fd694fb2b6625c27869a1d3f : SUCCESS in 4m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5323baec7b554c4d94464186074f0ba1 : SUCCESS in 25m 57s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d3ff9393514041a58a5ea3e4e2fd499c : SUCCESS in 5m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/ec7e7fa1fd344b4d9d503f065dc36262 : SUCCESS in 6m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fb4a178230ec411d86f3b1b425aedc56 : SUCCESS in 6m 28s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c1f7718473b743e980e44691ed520819 : SUCCESS in 18m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/709a4eefa15f48e9a0f7cfde6d93fbfe : SUCCESS in 1h 04m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d0e3a5dc11c84c8bb6cb18765549fff6 : SUCCESS in 1h 23m 21s","accounts_in_message":[],"_revision_number":1},{"id":"535038ee77f75195bbcbdd4300711ef185934406","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-11-14 17:38:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"30c2473fef0867d583c4245bb22abbbd9e8a2689","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 17:40:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c6e096d9d172a788f7699d6f06f90e540087e38c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 21:23:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afac0d20e064485bb5894c2e31e20719\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be3215ce924f4913b4e7fbda9f3333de : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f2fc354d1624212bfb725efb43f71c0 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b51b09d2cd444e48aeb152b55d397847 : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e47f0aef509448c84e60ed76dfb24c7 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/00697d3471304984b441a57f0bb71d55 : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/968d05c44815403e99e18883a7f0c930 : SUCCESS in 2m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1f5d2c6b613c497190737848141bff70 : SUCCESS in 16m 30s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e764c84e5e76423eba32d3cbfa65e920 : SUCCESS in 9m 53s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/adf0f73cf3b74c52b7b575d858cfa652 : SUCCESS in 19m 25s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/e7ae7c89c57c47f4a45840389d602523 : SUCCESS in 22m 16s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/96a8676bc5ae4a6cbfa62c466091da56 : SUCCESS in 11m 57s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/ba6efe5399424460a9fed5f0aaa5c442 : SUCCESS in 6m 30s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b2e019be8722481b99be2d0f1f0c55f6 : SUCCESS in 6m 04s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/bc4bdc5cc18a4477952c262faaf443f8 : SUCCESS in 9m 42s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8046f4d93a664b748d21cd2089a7163f : SUCCESS in 12m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0ab0ae86d0624902b4037a8a3fed1253 : SUCCESS in 7m 56s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/7f23419643e34db5a1b8005cb855774a : SUCCESS in 8m 46s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/aff64412ce784ee7aab0fb47d01cd57c : SUCCESS in 12m 25s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5663e81ef0924202a8e8782efade20d5 : SUCCESS in 2m 32s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3528715d6c6b4344ab329858f88c31af : SUCCESS in 12m 38s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/d0ba898ce80c430fab4d97b7b727feeb : SUCCESS in 19m 04s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e0bbb599b0ef4f118a6c2be65cb78bf5 : SUCCESS in 9m 24s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e06b57d5a31d4685b40871f9663fc153 : SUCCESS in 13m 52s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/38a4de29476744e3a9f30de335011d9e : SUCCESS in 5m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c1214152be5341e485e25f9c0ce3ceb4 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dfeeb29f09094364b424c8a5d78681d6 : SUCCESS in 23m 06s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e67e7c36250c4a81a628f3ccb5133c08 : SUCCESS in 6m 02s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/95a14c03b40e451f8cdc17ad05221e1d : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/99fdf409f0ce4386bb2fe6b3abc56e53 : SUCCESS in 5m 27s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/800bb0e2dac44298bbd8c37e51828de8 : SUCCESS in 15m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a432123dee6047a9bdd485ffbe98157a : SUCCESS in 50m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04503d33dbdc47cc961d188260380e45 : SUCCESS in 1h 33m 08s","accounts_in_message":[],"_revision_number":1},{"id":"85b32357e0c61383988d683547ccbb5bd6a56540","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 21:23:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fa595e64aeb2dd4831bfec585d09f341e5a18451","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 21:25:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92a11de100454e12b49a2cb2a209bd94\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06c61e8992554708b6f727719852cd09 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"913580089ea2a58410dbf40d18ae7b26a9e3f82f","revisions":{"913580089ea2a58410dbf40d18ae7b26a9e3f82f":{"kind":"REWORK","_number":1,"created":"2024-11-14 09:59:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/56/935056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/56/935056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/935056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/935056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/935056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/56/935056/1"}}},"commit":{"parents":[{"commit":"1daffb38e8d91ac33de84c850ba406bcec7a8aaa","subject":"Merge \"Bump importlib-metadata to version 8.0.0 (\u003e\u003dpy3.10)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1daffb38e8d91ac33de84c850ba406bcec7a8aaa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-14 09:59:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-14 09:59:32.000000000","tz":0},"subject":"update constraint for taskflow to new release 5.10.0","message":"update constraint for taskflow to new release 5.10.0\n\nmeta: version: 5.10.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I04c2c31b7f0167351ee3983a6c5709b9f34c1c64\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I6e3f3b764500d2f7a57bc4be59f3aa86469540cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/913580089ea2a58410dbf40d18ae7b26a9e3f82f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/913580089ea2a58410dbf40d18ae7b26a9e3f82f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
