)]}'
{"id":"openstack%2Frequirements~935063","triplet_id":"openstack%2Frequirements~master~I00ad22615a5f5891ed254ee160a82d023c0bb2d5","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-11-18 20:09:23.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-11-18 22:10:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I00ad22615a5f5891ed254ee160a82d023c0bb2d5","subject":"update constraint for octavia-lib to new release 3.7.0","status":"MERGED","created":"2024-11-14 10:03:21.000000000","updated":"2024-11-18 22:11:32.000000000","submitted":"2024-11-18 22:10:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935063-new-release","meta_rev_id":"b3b70d732616fe77e92c6f3f6df0306cd127e480","_number":935063,"virtual_id_number":935063,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-18 22:10:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2024-11-18 15:41:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2024-11-18 20:09:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-11-18 20:09:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-14 11:55:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-14 14:54:10.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-11-15 19:48:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2024-11-18 15:41:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2024-11-18 20:09:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"435bf6700e660221b1cb582862813b4f05ad6bf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-11-14 10:03:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26bbb1a53dc42923a8735c74d435590eb38f9573","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 11:55:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ccff7c648e6461fbb109d20f19a9d3c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b4df314138542b1ab65446f4544e265 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/955cd0ce2e6842378f3354b8745dd939 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73a77ed793ee42a3aac7988b430eed97 : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4b160fd6c0de4047a884a6c4c3410207 : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7257638754ea4e1c8bc9592d07e0ca51 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a1e9cff00797482e8c837723cd1d775e : SUCCESS in 4m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5fc74b3147df49d38c09b352564da2db : SUCCESS in 16m 41s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9d05be6e198548de81fa263fb94c3fbf : SUCCESS in 9m 21s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/0be7b9ff415749a78725dc2baf3e1fd5 : SUCCESS in 18m 40s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2cf0ca8e313b4a12a5061c870e78e03a : SUCCESS in 23m 31s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/1a112484b9944f81bc9f498120a8ca62 : SUCCESS in 6m 10s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/24e1f0ee333a4344a9668a1b01852ee3 : SUCCESS in 6m 39s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/8503b53330b34493ae7829ac7486f10d : SUCCESS in 4m 28s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/de171e9466434dfcba2848918112a96f : SUCCESS in 12m 32s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/db40ae305de7433c9e53cc36f2b25e5d : SUCCESS in 5m 55s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/805fb35de74f4d1392256d261f8f40d7 : SUCCESS in 6m 30s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4ed4217cf5884a8abf61fd5d0bf407ea : SUCCESS in 7m 45s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/dca3e1ff56714dd88a651eb3b917aff2 : SUCCESS in 13m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/984c484656d94f48930560912f2dbc79 : SUCCESS in 7m 30s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/bf9ca06fbfd64ff89ad76c399d8b2429 : SUCCESS in 9m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/495579a233ec4194bf1cf87931ce61ab : SUCCESS in 13m 35s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/5de7de6b9da1443792fca02fe746a69b : SUCCESS in 2m 12s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d6a7dc501e4e4f51867603f0de3b0e4c : SUCCESS in 10m 11s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d487b91c3c4442a7aa58c553be5b1649 : SUCCESS in 5m 06s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/21daf303baf84417a98cc4af1ddd5d25 : SUCCESS in 6m 40s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/eac40ca6500d42259e887a56d398124d : SUCCESS in 11m 51s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1820230bbfc14cb8866337346a84df6b : SUCCESS in 21m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/e06fbab6e36646929728be30369a6f2c : SUCCESS in 11m 11s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/692f975970b04b3780133d36106a2ec6 : TIMED_OUT in 40m 57s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/268e269741844511af27ab07f84a24ce : SUCCESS in 6m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3dabd28bd5bd4b32b1593445ff654606 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f7c9a01a75064a028fad75e9417b7500 : SUCCESS in 24m 46s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/696435304da447b0920d5f7a371d2185 : SUCCESS in 6m 12s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/90e221bdd7c0454793280710dc2fc365 : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ee607a9b56de44f5ba98632c1ef84f0f : SUCCESS in 6m 33s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/991cb82d3cde4fdf8e4ec335ee81e1c2 : SUCCESS in 17m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41cea86abb6d4904a75900c938083d22 : SUCCESS in 59m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc2e5ccdccac4b6ba0d10da9f3fe967d : SUCCESS in 1h 45m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d854bdb30ee959fddfcabe631cde35de8172f0fc","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-14 14:54:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1b661fc9ce98f2f74c9a3a82490b80d3106b7373","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-11-15 19:48:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5ad4e2d0430cf7772286d38c5f9d6f7602b0e321","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 19:49:04.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 935063,1","accounts_in_message":[],"_revision_number":1},{"id":"3853ac5c3e2c044df46a040c78664b1ff2bc05de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 19:49:12.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 935063,1","accounts_in_message":[],"_revision_number":1},{"id":"843602dcb575a8206454cca7872f73b686b54972","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-16 20:55:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"aaec1e038c40aaa94a7d2ae468a56f9d11fb5579","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 22:45:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5df5aeab18fc4c2781485c9f3a21bc6c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86796f02aa3249fcbe55064ee5d93727 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8a4bbfe92ea420fa0ac6ee637e92e94 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/217bc90dfee9464598481c241ab7c826 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e6fd34c559d54eb3abd97d8d8c5b8f9c : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/33b3aa7186b446538b314553e1997077 : SUCCESS in 2m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c25426bab6f548cba434d107bbc1902c : SUCCESS in 3m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9e39290ed0f64f7589bdf24f09e61b23 : SUCCESS in 15m 07s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9f08a9a2e4774994a226961124e55a81 : SUCCESS in 8m 22s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/529cc03fa6304c37afed086b578cc214 : SUCCESS in 17m 27s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fe59b81304fd4a21829ae5771cda9233 : SUCCESS in 22m 47s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/0597452075414b259e591289b589e02e : SUCCESS in 5m 17s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/111e6f80f51a4df39b18517ba3c6e374 : SUCCESS in 3m 48s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2fc29e3831e846588af97c3a4fc52f1b : SUCCESS in 4m 11s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/62b95b4fd9984eceb540ca4ad44acaba : SUCCESS in 9m 08s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/1b5d51db89544f8abe9d82f14d0ce505 : SUCCESS in 3m 42s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/24b91274167f4e98b273781aa28e80d5 : SUCCESS in 3m 40s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/e66bf2b33f374eec8c1f9a7be3d00b93 : SUCCESS in 7m 16s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8146503fbbf94307a78d19d439288b28 : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2207532ec0c44ffcbe416a37ad7dab01 : SUCCESS in 6m 21s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/611b9c2b643f44e581e8c00ebacc1d40 : SUCCESS in 7m 41s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/dd71153782eb40ad820360fbb58dad71 : SUCCESS in 11m 25s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0ee0e1c83f4344c49beb8500dee0e5b4 : SUCCESS in 2m 04s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/66bb0575783141d9afa482a6266cbee9 : SUCCESS in 9m 29s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/3e813f6a514a4aa98f5014941cd7b640 : SUCCESS in 3m 59s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/dc5274d9c35948d09eb3e514f703365d : SUCCESS in 4m 36s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/fc2f3810728a419ea47a4fff16f7f39f : SUCCESS in 9m 58s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ea9ff7cbfc81435ba508f95082437551 : SUCCESS in 15m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/584fe9ca1c804d679d67352fa26567c1 : SUCCESS in 6m 38s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/97c6b8d768e04b20b1d236878602cfad : SUCCESS in 10m 22s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/e974fd4e93884b0abd193d5dc96ece52 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b08f289caa424eb0ba90883c06210516 : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e09451ddc81c42f8823041cd295636c1 : SUCCESS in 22m 15s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/fcbb9b38beb9414fbbc2128136aa2234 : SUCCESS in 5m 24s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e0eda1678b484e4c82bb9738dc76a427 : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8128ec3fa1864abfb2b99b1358c02395 : SUCCESS in 4m 40s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/2081a0e1c9be434f9cb96b4056c1b0ae : SUCCESS in 14m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01b0c891ab32477bbfe4efc902ee0591 : SUCCESS in 1h 02m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6dc4ae1123ce4541946fca81280ebfd6 : SUCCESS in 1h 43m 49s","accounts_in_message":[],"_revision_number":2},{"id":"39218440605d4ced919bbf53645fe633ce2c9ce2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-11-18 15:41:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2964c32bbdaa7cd0545bd3188a6b7c74f94a5b7e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-18 20:09:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4e078f6b751dfd000a92fa03d7b6bdd2807b8ac6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 20:10:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6ab5c0a34a5e7c96cb6ed4ea16b5cbd5ff0d8cc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 22:10:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5d862afe7bf412d8f9546c4a0d49408\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a3092fafc6046b2aebfbaa81b181745 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/405a235ef4214f22abf0c307ef7dd9a7 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72a9a4da2f924715b7f344788bd4431d : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/60b9e6a1e7de43aeb37db7ac8806bdde : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/505f6815d9524af0a025a57f009643f2 : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8bfbfee7a6584c7994d221db87c095d5 : SUCCESS in 2m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/61627d27b77c43da8bdedf9e3c9b320e : SUCCESS in 13m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/ed88372799004707b5a72d198f3d0afe : SUCCESS in 7m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/ce290318bb6e4d578616cce704ea3005 : SUCCESS in 17m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ce94c2e384884e898f29985562ec9b57 : SUCCESS in 23m 57s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9166e7b4d37b43ee8575f426c6ca145f : SUCCESS in 10m 43s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/f7a64253b06743bea6f91be348bc82cf : SUCCESS in 5m 10s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0c5d70ef46854b608d9fecbd2a079518 : SUCCESS in 5m 28s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0740995b09984fee839c01cb14a11d79 : SUCCESS in 6m 36s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/e4209f62b4ca4fdebc866063e064274e : SUCCESS in 14m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/165abd01bdde4bef925107c90e29853b : SUCCESS in 6m 53s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/67ff30aedf1442fa86293ae6833872fc : SUCCESS in 5m 53s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/215f40690e3a431b8951a65e9189bd23 : SUCCESS in 10m 47s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/2b5833ce93eb4205b3965f42c74504bc : SUCCESS in 3m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/85a1e9bba9de4bae91c7f20c87332940 : SUCCESS in 7m 36s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/21b255b49def4aed821738289488d69c : SUCCESS in 21m 38s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/fe22de40a65e4086af85d2f148b71727 : SUCCESS in 4m 57s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/292b7e5697e84f788fb438f25289201b : SUCCESS in 12m 27s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ce8b94f99efd4b55a690a47b6e3e6da4 : SUCCESS in 6m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9442bdde60244590a82c35cab3f6e0ab : SUCCESS in 4m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2125d16e37b449a3b6127c173979f2cc : SUCCESS in 24m 43s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/976023c1ed3647ecb0cb040515ec5504 : SUCCESS in 5m 54s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a80c83f2f5fd48ea83ce9f01c05bd369 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c019d9c1b1a94e698303015ca99c2f5e : SUCCESS in 6m 16s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/846301f6fece4041aa33b4238e0d35b4 : SUCCESS in 18m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34c00081787b45e7a09bc7305a74ded5 : SUCCESS in 1h 00m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ce8b762292a44e8a77787e73c6c9393 : SUCCESS in 1h 55m 11s","accounts_in_message":[],"_revision_number":2},{"id":"c4aa11d36d4dcb78fb1b0c296a574cf6668c627a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 22:10:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b3b70d732616fe77e92c6f3f6df0306cd127e480","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 22:11:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa236db4af3b471893fa51e9ec2ddb16\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b65f2e6f3b64060adc8a66d7caab9ab : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6d69db2495b98f2115137a1c49a5c4e44a550233","revisions":{"f6abe7ab25f680087b29b75d89a6cf8f0f3ed60a":{"kind":"REWORK","_number":1,"created":"2024-11-14 10:03:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/63/935063/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/935063/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/935063/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/935063/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/935063/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/935063/1"}}},"commit":{"parents":[{"commit":"1daffb38e8d91ac33de84c850ba406bcec7a8aaa","subject":"Merge \"Bump importlib-metadata to version 8.0.0 (\u003e\u003dpy3.10)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1daffb38e8d91ac33de84c850ba406bcec7a8aaa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-14 10:03:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-14 10:03:20.000000000","tz":0},"subject":"update constraint for octavia-lib to new release 3.7.0","message":"update constraint for octavia-lib to new release 3.7.0\n\nmeta: version: 3.7.0\nmeta: team: octavia\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ie56f1b29143328efeded746d7e28a799772d3081\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I00ad22615a5f5891ed254ee160a82d023c0bb2d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f6abe7ab25f680087b29b75d89a6cf8f0f3ed60a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f6abe7ab25f680087b29b75d89a6cf8f0f3ed60a"}]},"branch":"refs/heads/master"},"6d69db2495b98f2115137a1c49a5c4e44a550233":{"kind":"REWORK","_number":2,"created":"2024-11-16 20:55:40.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/63/935063/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/935063/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/935063/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/935063/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/935063/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/935063/2"}}},"commit":{"parents":[{"commit":"1372e0d4d66c0640f63be5176a1bd7a25eda39eb","subject":"Merge \"update constraint for python-manilaclient to new release 5.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1372e0d4d66c0640f63be5176a1bd7a25eda39eb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-14 10:03:20.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-16 20:55:26.000000000","tz":60},"subject":"update constraint for octavia-lib to new release 3.7.0","message":"update constraint for octavia-lib to new release 3.7.0\n\nmeta: version: 3.7.0\nmeta: team: octavia\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ie56f1b29143328efeded746d7e28a799772d3081\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I00ad22615a5f5891ed254ee160a82d023c0bb2d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d69db2495b98f2115137a1c49a5c4e44a550233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d69db2495b98f2115137a1c49a5c4e44a550233"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
