)]}'
{"id":"openstack%2Frequirements~935080","triplet_id":"openstack%2Frequirements~master~Ic1a071c2fa1f3726a237c7864f009aec786deaed","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic1a071c2fa1f3726a237c7864f009aec786deaed","subject":"update constraint for python-manilaclient to new release 5.1.0","status":"MERGED","created":"2024-11-14 10:22:58.000000000","updated":"2024-11-15 22:43:01.000000000","submitted":"2024-11-15 22:42:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935080-new-release","meta_rev_id":"e8677c05f97d150d1648f951ce7bedeec944e2b6","_number":935080,"virtual_id_number":935080,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-15 22:42:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-11-15 19:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-11-15 19:48:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-14 12:04:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-15 16:40:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-11-15 19:48:34.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f2f2036a46c4c797387ff1a263bd8dd9a5c5e91d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-11-14 10:22:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1e468f6ef8eca53a058de3207e5d2be651df4a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 12:04:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4c019b94a8e408e8a645f79b7334d99\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9197d48d7ce04d06b5458ed6610273a5 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd21cd575b9e476ea6f3bf82555b625b : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2282c6d94ed74f16942909e9281c30a5 : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2dd8646814994708bce95747330de01b : SUCCESS in 5m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cca2697009ae4ffd80875057f88f86e7 : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0fd9393ced3b4800b87f7f019f975b9e : SUCCESS in 3m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cee8bd1de0f74135ae532799e4d9854a : SUCCESS in 14m 54s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/634e5787272441e395c52e7dc2362efa : SUCCESS in 8m 35s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/580e7f9ea4004a469d2accb99d6e3e72 : SUCCESS in 15m 57s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/1cd8234622c04d3ebaeef5b3471c30cd : SUCCESS in 25m 51s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/0876e35f9f244d5ca08dd0e88ba4a06f : SUCCESS in 8m 24s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/d145190bb1a34ede8e68406eb788de03 : SUCCESS in 5m 44s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/949a5b11bbea492c8f57fcdcd0c2eea9 : SUCCESS in 4m 12s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e979d1dfdbc24d908c3dcea9916903a9 : SUCCESS in 9m 24s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/b4d6c716f1a848fe9f83470c4cc8989b : SUCCESS in 5m 15s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/821e6d1ab09646e4b01478c8cf4e7e84 : SUCCESS in 5m 08s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/db441f31c11c4597841942651bea9b93 : SUCCESS in 9m 22s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/91325996c25347a1a221b759c4ddf6ff : SUCCESS in 14m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/40ec80c336674412aad17f5f241f1bdc : SUCCESS in 6m 32s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/d18f9acb07f049c4b0c27eadf857ccdc : SUCCESS in 10m 43s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/58508543cc414a999fe7e68c7b93b253 : SUCCESS in 12m 04s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/93c391d34d364c088dc1c80f3124882f : SUCCESS in 3m 20s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/63a1e03c82e74b6abc75359caff3c6b4 : SUCCESS in 8m 28s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/209602da9d324d0484bf0570729c933d : SUCCESS in 6m 12s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/31329259fa2447f1a46d5c83ad1ce0a9 : SUCCESS in 4m 57s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/f9c57d20e3b241d0beaf80d8fd227c3c : SUCCESS in 12m 46s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/5ca04065a8c34f1d92641fcbe035392e : SUCCESS in 17m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3f11a3e634014ab1852227ee52eb91ef : SUCCESS in 10m 40s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/c8a16370e0bc4336817047dcce0dfe84 : SUCCESS in 11m 40s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/eb9b6996dfc9428e9df28bc4428149f1 : SUCCESS in 5m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/39fb58b9621d4330962353602e23c9b8 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/219daa1f04624dd1b45e366e47a3c398 : SUCCESS in 26m 35s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8570d7d22939446682c7e2845ee840ac : SUCCESS in 6m 03s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/d08e45a72f2548d1a71faeca69fe8150 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a18056a6c5ff4856984c45d5aa358ffb : SUCCESS in 5m 24s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c88193828736460abd49e99c005e2886 : FAILURE in 18m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a764002b0fc54bbf9fd55dd6bfcd25ae : SUCCESS in 1h 02m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f15d6cd3a5fb4b348c4c924166518bd2 : SUCCESS in 1h 35m 08s","accounts_in_message":[],"_revision_number":1},{"id":"3583f7c74c72efe8ec41f8f83f5f0ae7dbae7bb6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-15 16:40:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c33caf563926f01791f37d11e61469808a6eeb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 18:07:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb4af4df59e8453c8f28ea22e22e0e14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc07a7288eea4df983d3c245344997b0 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57d55172440642a091863985337f22d3 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57fd4bec1be246598146900712d541d0 : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85c767700e47454da9e961657804aca4 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bed0ff2e2c474ab79099dc8501bf5d54 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/04a2762927794389acbac6f8f758df60 : SUCCESS in 2m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/039af9d59db24f8ab3ad80178ef5e056 : SUCCESS in 14m 51s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/176b2e7b63a6466f8e1ec495e14dc2b5 : SUCCESS in 9m 59s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/fbe6a45e679241449305eef2030c8962 : SUCCESS in 17m 41s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ed1c7bad1d444aab965b51093472bc4e : SUCCESS in 22m 42s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/fb55d9ba45cd48e68bd2803a5d9ed21a : SUCCESS in 6m 45s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b1cd84cdcfd74e808b8060d504e527d0 : SUCCESS in 7m 05s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/118865cdd1ba4f00be0b65b554794d73 : SUCCESS in 4m 54s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/b78b38dafe9e4eadb9844509da64d664 : SUCCESS in 14m 58s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/73a3ea984db84f4083c8ef8afb22c8da : SUCCESS in 6m 13s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/22cc5e6a805b4dd08c0cb72290ffb257 : SUCCESS in 6m 58s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/cacbfe16b34d4b22b2e8e614d3597782 : SUCCESS in 9m 04s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/678408bad4b24e7792b59aa99b8988ec : SUCCESS in 14m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5eaa51e5ab944906b7f1242d8f733339 : SUCCESS in 9m 13s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/677a92f963c94387ac955caa448ea3a7 : SUCCESS in 9m 16s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b959e9414105452397d8d68b4950fca8 : SUCCESS in 12m 58s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3589ee6c19594a6a9f38e364a3f185f3 : SUCCESS in 2m 37s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/893ca9a2203142beaf85d3a720a4a37c : SUCCESS in 10m 56s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/3c1e742b454d495991bbcb680bb17fda : SUCCESS in 6m 33s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/e0466978433c4889a4c0a86619141396 : SUCCESS in 5m 34s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/e0cdbdd673af4e81a256fba3c93d3d1f : SUCCESS in 10m 17s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a8691f84cbce48eda305dbefa1898cdb : SUCCESS in 20m 26s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/023db31fa4834cd08a4ca9c267b0e2f5 : SUCCESS in 8m 35s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e82d8014b5b14a5fa86ed9468bdc0f4f : SUCCESS in 15m 39s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4110099659d4417b8f13cf030f7d533d : SUCCESS in 6m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1b6187bce5d1405e8f65a77133588cd9 : SUCCESS in 3m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/233e729e10ab4ccabc7ece6fa9f9493b : SUCCESS in 21m 51s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e058b93f40c74d96a159a4863b5445ef : SUCCESS in 6m 13s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/9fbf86f1e28742d683462e4abdf9500f : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/69b65df8fb494a5ea8aa30444d1dda07 : SUCCESS in 7m 21s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/80b6cc05a5434df69f4b80496d1ae1d4 : SUCCESS in 16m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d073f546e2ea4f17802796220b8781b0 : SUCCESS in 1h 02m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f9faa22b3ed432ba7e2af1e4aad4977 : SUCCESS in 1h 26m 11s","accounts_in_message":[],"_revision_number":1},{"id":"048071a8f030fae8f3b4ab5857bb9bccc364107a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-11-15 19:48:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fbc1b675e3798edd3f30383914384e4421ba3e4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 19:49:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41549c27d9f7327eb6286359dd6103e1eb434963","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 22:42:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/086e9beae2ea486da0b56fa8c0768bc4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e636de1fa966488b9a18fde6c59f1dd2 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5440abae2f574f5384e0217513952497 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/295d9876201442fa87e47ce8fb4c3d6e : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/616fbe0c35f5461db6968f3baeab9bd0 : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ee991e5564704f97add6365d2719a1e2 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c095bd663ec646c283356972bdc97c70 : SUCCESS in 4m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2c6c8e18190b4113bcccf2bd901da1e7 : SUCCESS in 17m 36s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c0415715ab684458b570d77a617f4429 : SUCCESS in 9m 50s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/183d6ce312b040d9a008fadf44e3f059 : SUCCESS in 17m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5b6512ca058a46c98f9c69ae6a987cef : SUCCESS in 11m 55s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a585efebfcbd47c8a94f4069a35c1911 : SUCCESS in 11m 24s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9f18a8d0df8f42bbac0abfb05100581d : SUCCESS in 5m 24s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/f8d2ae8a234b443b87335a1ea73b45f2 : SUCCESS in 6m 48s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f2f0832d9e3548d99ecce95bcec907ec : SUCCESS in 7m 17s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8f84236dcc784550b82fbcfc47c0e584 : SUCCESS in 14m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2d8c5ea8e1034e74b864a65a67d87dff : SUCCESS in 7m 24s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/4559bae64b5e4456ad0eca50a809a240 : SUCCESS in 7m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/7506b92b907c491aa1ee82e316e547c2 : SUCCESS in 9m 44s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/53e123ad2cbd4c9a93cdda0a08e8c567 : SUCCESS in 2m 25s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/a9ee8695c0c64ec5a5972cb3898e32ea : SUCCESS in 9m 34s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/44330b2775b24dbeba9a7d99c665149f : SUCCESS in 22m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3f85ecb1cf6b44ac9fe3d417dd63eb21 : SUCCESS in 6m 49s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/9b1985a3db2e4a95bfb377ad5eeca79d : SUCCESS in 10m 40s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/acef41834d2a4dc3ac061df764b325f1 : SUCCESS in 5m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4e40e0dafe614aa5ac27672d1921fe3e : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4dd94e4c5b714c198c2ded7a9858a8ca : SUCCESS in 27m 04s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/2d7ca73f8adf4bbfacdc3c297408318e : SUCCESS in 5m 26s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/38e1d8be71a34ef0a7787cc190c72c0e : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5a01b82c508548e7a68cd434b53f1899 : SUCCESS in 6m 55s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/b385a78159914426ad2033bf73983146 : SUCCESS in 15m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/088a75ea6c884cffb023da678ffa790c : SUCCESS in 1h 04m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f0a8e616f4240a48b9b785b2cc71df9 : SUCCESS in 1h 50m 03s","accounts_in_message":[],"_revision_number":1},{"id":"0fa30f4b55fea78e77a145379ad6863136cd39fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 22:42:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e8677c05f97d150d1648f951ce7bedeec944e2b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 22:43:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6afa30fecea484da2fc9e8d5ceac9be\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce01efb44bef4b96a39e7c1f2c764b6c : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e9e5b5125cf0eee2b56b8106dc34251fff3d2b09","revisions":{"e9e5b5125cf0eee2b56b8106dc34251fff3d2b09":{"kind":"REWORK","_number":1,"created":"2024-11-14 10:22:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/935080/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/935080/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/935080/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/935080/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/935080/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/935080/1"}}},"commit":{"parents":[{"commit":"1daffb38e8d91ac33de84c850ba406bcec7a8aaa","subject":"Merge \"Bump importlib-metadata to version 8.0.0 (\u003e\u003dpy3.10)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1daffb38e8d91ac33de84c850ba406bcec7a8aaa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-14 10:22:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-14 10:22:55.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 5.1.0","message":"update constraint for python-manilaclient to new release 5.1.0\n\nmeta: version: 5.1.0\nmeta: team: manila\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ib247f4ebb2787568d375ccef05f2c99f5909aeb8\nmeta: release:Code-Review+1: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ic1a071c2fa1f3726a237c7864f009aec786deaed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9e5b5125cf0eee2b56b8106dc34251fff3d2b09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9e5b5125cf0eee2b56b8106dc34251fff3d2b09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
