)]}'
{"id":"openstack%2Frequirements~935232","triplet_id":"openstack%2Frequirements~master~Iade27d584791d6a4ea1d986e52412f266a99d14b","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2026-06-03 15:44:05.000000000","reason":"Change was abandoned"},"2":{"account":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"last_update":"2026-06-03 15:44:05.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iade27d584791d6a4ea1d986e52412f266a99d14b","subject":"Add netifaces-2","status":"ABANDONED","created":"2024-11-15 00:23:49.000000000","updated":"2026-06-03 15:44:05.000000000","total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"1260edb143a1d63c4a35381dc21ae15369c68973","_number":935232,"virtual_id_number":935232,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-15 01:44:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"date":"2025-03-28 16:15:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":-1,"date":"2024-11-15 00:49:51.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}],"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-15 00:27:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2024-11-15 01:44:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-15 09:54:22.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-02-28 18:33:38.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"1178fcfae36a564ba7cd269330737bdb38c470eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2024-11-15 00:23:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57cb74a36ded0dffc01f38524f35bf37a83cdb65","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2024-11-15 00:27:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"988db85dd25e5a091e6535dbefaf7bf93f2f3af6","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2024-11-15 00:49:51.000000000","message":"Patch Set 1: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ba43e4f4c90f23c1775dd50169781416fb633ab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 01:44:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6fc014701ad48ab80464ef6bd6de64d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e7e2e97b3ca44febd6364cd106ec3de : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5143a605ea024f0ea730818d241c0ed1 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fcd570fc42a4dc5b9d487227fa730c5 : SUCCESS in 6m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84ff4a200772423db07d25a04546d1b1 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/18ae3ccb64a442c0aca6e6571b8ef7b4 : FAILURE in 4m 00s\n- requirements-check-self https://zuul.opendev.org/t/openstack/build/dc2f054226c04811917a179cb1397db9 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9ca39cca87a247fc9de9b70e2ceb05d7 : SUCCESS in 3m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/85d24ceff6b043498d9761b5744b40db : FAILURE in 5m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/21fb0a6d1e4645b29b22839ddf48c2ed : FAILURE in 6m 39s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/94d6104a2d9e422183862c19de8b4d1e : FAILURE in 6m 03s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/300b8b881ba9411992cca5ae784a7547 : FAILURE in 7m 42s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/9f6efa2584f2454b8f79e50504788779 : SUCCESS in 6m 15s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a11ecc3b1e56438c94f68c581ff8369a : SUCCESS in 6m 14s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/10e2e1f755d74a2399b6cf0784427172 : SUCCESS in 4m 02s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/45b9eedec96c4be8b612dcbf9e823146 : SUCCESS in 11m 38s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/22826b90bc254aa898e405ad9dbbc961 : SUCCESS in 5m 04s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/b855d0be55334b90992d457aa049a293 : SUCCESS in 3m 35s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/a4c5af755ff5493486097b32f7193de4 : SUCCESS in 8m 04s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/3906a048bae94ac2a687fe18caa5c565 : SUCCESS in 14m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5e02671aa7224d68a11116093148c7d6 : SUCCESS in 5m 03s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/cfa511eb6c374b03bf3acc369c60ff0f : SUCCESS in 9m 17s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/ebf243f55f1e44298254ec1773a93cee : SUCCESS in 11m 09s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b105e74fe1fa4cad8ddd36fed1994667 : SUCCESS in 3m 16s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/aaf439107f0c46d6b14a7f2d4ad72fa7 : SUCCESS in 8m 51s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d5eb17476df94a79840d12308e54a185 : SUCCESS in 4m 32s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/a700688eaf6b4fd6874fe16f8d55f1f2 : SUCCESS in 7m 39s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/07f71b6003c04f489a1524852f7f71ff : SUCCESS in 10m 31s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4ab19c94f8b946c88f7d898a2d4f3f30 : SUCCESS in 17m 15s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/72a838c4e1fa4a3f89d39bb47d804aff : SUCCESS in 9m 32s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/89421c7049614a749a88c90f5e96be6a : SUCCESS in 11m 12s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/4e17b19ce33c4d41895a89d57d957b36 : SUCCESS in 5m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c114758b86904e678df48ed7ffedd6e3 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cb2bac0ad3d84f798c1a552c9bded800 : SUCCESS in 27m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/1673ca84f11548dabf7c90a00a731536 : SUCCESS in 5m 39s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/08c6f53ac7df44b69495e310879a3c7f : SUCCESS in 3m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fa3c81c6d00348879dcba7561ae0ab24 : SUCCESS in 6m 10s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6aecd3b3fa4543c6800e6806f273c504 : SUCCESS in 17m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8160294c41f64c78a1df27ba256554fc : SUCCESS in 1h 00m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f443dc5a54d94dcd8ede7e4ed0ec05cb : SUCCESS in 1h 18m 23s","accounts_in_message":[],"_revision_number":1},{"id":"35a37d9d80fe074a2cd038b842a61c424b1aa9bd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-15 09:54:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"160deffcef226b93ac26c290db5ce59b73200d1e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-28 18:33:38.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3bd5a0cab2aa229b6c2f3c002a1d1075fc5c28b2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-28 16:15:09.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"1260edb143a1d63c4a35381dc21ae15369c68973","tag":"autogenerated:gerrit:abandon","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-03 15:44:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e06577f93b09579e7f4e1c72ed7e4dd84a87db2a","revisions":{"e06577f93b09579e7f4e1c72ed7e4dd84a87db2a":{"kind":"REWORK","_number":1,"created":"2024-11-15 00:23:49.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/32/935232/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/935232/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/935232/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/935232/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/935232/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/935232/1"}}},"commit":{"parents":[{"commit":"5de06faac68d34c7e6810cfb9fb6f9d828ffb19e","subject":"Merge \"update constraint for python-octaviaclient to new release 3.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5de06faac68d34c7e6810cfb9fb6f9d828ffb19e"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2024-11-15 00:22:11.000000000","tz":-480},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2024-11-15 00:23:44.000000000","tz":-480},"subject":"Add netifaces-2","message":"Add netifaces-2\n\nnetifaces is dead, but openstacksdk depends on it. netifaces-2 is\na replacement for as much of it as openstacksdk uses. Add it here\nso that we can unbreak sdk issues.\n\nThere is also a suggestion to replace with psutils, but the\ncorresponding psutils functionality doesn\u0027t actually do the same\nthing.\n\nChange-Id: Iade27d584791d6a4ea1d986e52412f266a99d14b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e06577f93b09579e7f4e1c72ed7e4dd84a87db2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e06577f93b09579e7f4e1c72ed7e4dd84a87db2a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
