)]}'
{"id":"openstack%2Frequirements~935358","triplet_id":"openstack%2Frequirements~master~I4b04da1ea95c2ec52727a9bee4921c2a097c9f1e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4b04da1ea95c2ec52727a9bee4921c2a097c9f1e","subject":"update constraint for python-zaqarclient to new release 2.9.0","status":"MERGED","created":"2024-11-15 14:21:14.000000000","updated":"2024-11-15 19:02:51.000000000","submitted":"2024-11-15 19:01:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935358-new-release","meta_rev_id":"0f211a51c098e729d76770de389c863d53d8b9b6","_number":935358,"virtual_id_number":935358,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-15 19:01:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-11-15 16:43:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-11-15 16:43:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-15 16:14:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-15 16:43:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"7c33352468b71343bbbaa7a93cd6b02699ed547d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-11-15 14:21:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8d2f4b756487310576571f6de3f2a1a4e95d526","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 16:14:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53ef105fe11844bf8acd602d3cd89d6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2452da3c88f14273883228ebd20adde6 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08b2cc8bf40e489b938b4e69be94ec48 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65c8b3cd9c7a47128f52ce6ef1415f9d : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/06aebb0baa014ad3892d23ad8080662b : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cf1e4a240f0a465ab0a7cdb201be7a50 : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1e22de2f955944a18b22b4ad64c141ab : SUCCESS in 2m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f3b1d39590d14d3a9f6e100abee32227 : SUCCESS in 19m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3c60f29523784884b25dcd08658058c1 : SUCCESS in 6m 33s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/66a2dde5c25a4de78dcf53a84dbd01da : SUCCESS in 16m 54s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/fde3120732794e3a9a5a382ae97ea532 : SUCCESS in 11m 45s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f5ac86fb2bca46e6807329b5217cfd92 : SUCCESS in 3m 47s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/fd99b350c98541bfb91e0fce6c0aee80 : SUCCESS in 3m 46s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/2cc1c97ed2cf48458042bfb9dad398e2 : SUCCESS in 3m 03s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4bd32556922143f890afe36722615959 : SUCCESS in 11m 02s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/2aa9c1115194475d941d2f8cd3555d0f : SUCCESS in 4m 45s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/5d12b23196ef489bb70cc38f41ff88b2 : SUCCESS in 5m 33s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/4055636297de4f59b209063484c62fba : SUCCESS in 7m 23s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/49c9d6c1cd674c9b9fb26ba634dd7a31 : SUCCESS in 12m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/114cc141894d46a7941be0b536ee24a2 : SUCCESS in 6m 37s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/1d31f776d5ea4db9a6d413446f319c77 : SUCCESS in 7m 18s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d1b5ddc373cd438fa049ed6f2f1d1b59 : SUCCESS in 12m 51s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/3b39db4c212e4c3f95f95d746c6adb26 : SUCCESS in 1m 52s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/ce8035eff4804a3099d0b9ddf02f5b7d : SUCCESS in 8m 28s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/ce91951bf6a146e2a29009692204a8c6 : SUCCESS in 4m 16s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/4d26c6286d2a4b708643c92e37002465 : SUCCESS in 5m 10s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/0ecf74f22b514c75888ccf8c88e8db21 : SUCCESS in 11m 26s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/c5db6077cc864d5aaed96d8c84abd74c : SUCCESS in 22m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/273faeed3a8247299b23c3fe33dd4a18 : SUCCESS in 7m 37s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/900aea71513a478d88dd2dec63b8d57a : SUCCESS in 14m 46s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/66bb31e1a16a40e487c1bfc60d82faa8 : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2cbcf849d683492bb83a69bcb8898292 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8f28935c453e4d90a20605c607fe1e53 : SUCCESS in 25m 34s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/5d95d7e61b0c4d61aa1d8e3828dd6b0b : SUCCESS in 5m 55s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/30b5029b8f34476c98e03de496976766 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bc320cf2f2ed4661852f364ad90d7be7 : SUCCESS in 6m 41s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/0dbcfd423c604841b1b15ad0509ef43e : SUCCESS in 13m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a240ca0c3d8461181bb049b67acc375 : SUCCESS in 1h 00m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b217f30611da466082c5af717843523b : SUCCESS in 1h 52m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c8076bcbc0c4b95cbc460b17b4ed9e441d498302","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-11-15 16:43:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"90cfdf5839e93ffec1fe3fa441376efab7f80d0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 16:44:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8a1239387a1ae46a85fbb55a16917ea487cd0014","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 19:01:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8036ee3dda574a528b8de2042cc60bb7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/470154eac5dd4b69941a22489e95b9d9 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d53b68f513194c08a8c927d5ad8f2499 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16b540c912af4ed787bd57ffe6915148 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b02413ff3be94b08b0dcd1c82bd76cd0 : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b01ef22001148cc894cf59f5634d1c3 : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/86b5906a50f442a7a3ee15cab7230733 : SUCCESS in 3m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e467ddfcb037489aac317155ac56170f : SUCCESS in 15m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3357ca03a7a546e88711a16d68d53148 : SUCCESS in 8m 41s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/cdafa9de43954e9196f0b395e60cfe5c : SUCCESS in 16m 41s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/2c9ac3c2928d4cc8956fa9773430a71d : SUCCESS in 30m 22s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/2bdacb5ffa244eea9fc940b6bbc42aaf : SUCCESS in 13m 27s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/867105d4ad5c4fe1a9d3c86dd96e500b : SUCCESS in 5m 04s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/2d2a77698e194456b1e4118d8383edc4 : SUCCESS in 7m 46s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/03b48b7e72884e0a92415f5df61f340a : SUCCESS in 8m 59s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/0c3f90ceafa3402bac3fa5fa768f8ecc : SUCCESS in 15m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f988ad8e03e8446d9dcfca9a288ea467 : SUCCESS in 10m 17s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/06c558de60aa4279a2697893fd3b27a2 : SUCCESS in 10m 40s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/539d18ae2c214c8199b8e9e305df829c : SUCCESS in 12m 16s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/19530d526a9f4254820f284a9423738c : SUCCESS in 3m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/70bc7140526c438e947c874c3450d852 : SUCCESS in 11m 47s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/6e1b01a03cca4cbb8a176bb846f5d0bd : SUCCESS in 22m 13s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2605ddbace9646ed83b84ff627ffdfb0 : SUCCESS in 8m 09s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5576dca0eeec4a7bbf0140e1fd9aa8f2 : SUCCESS in 14m 50s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/61e2b71792d240d0a994db69d95fff4b : SUCCESS in 5m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d462d6f543784e988ae891f3f5a0d144 : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/efea544c118c4e76ae3749e1da8b48c8 : SUCCESS in 25m 24s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/e2d6e79e39bd46e0874b452e94243a4c : SUCCESS in 5m 46s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/f66de848078d43b5b03dbbb55af90644 : SUCCESS in 6m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9f08d2b187ea42928f2975113b5f157f : SUCCESS in 7m 50s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f4980b8069364937a6038f9c928bb545 : SUCCESS in 19m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e55264923daf4a69914458e095336227 : SUCCESS in 58m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b94c3236df5e41be9cdb1e783c8acb67 : SUCCESS in 1h 42m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1e1343b6b77c0605e4559ab834d792aca4a9ec15","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 19:01:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0f211a51c098e729d76770de389c863d53d8b9b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 19:02:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a64fdfb6f3d04740b046b4d45032367d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6f21fa08bdd457fa6a47bb9c37d0a43 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d34d0d3f6be12ef69b8d7664ccd7cae58b1d15e","revisions":{"7d34d0d3f6be12ef69b8d7664ccd7cae58b1d15e":{"kind":"REWORK","_number":1,"created":"2024-11-15 14:21:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/58/935358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/58/935358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/935358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/935358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/935358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/58/935358/1"}}},"commit":{"parents":[{"commit":"5d3f3fcecdc0a48c7640cebcceea5f2fae8afbeb","subject":"Merge \"update constraint for oslo.cache to new release 3.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d3f3fcecdc0a48c7640cebcceea5f2fae8afbeb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-15 14:21:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-15 14:21:13.000000000","tz":0},"subject":"update constraint for python-zaqarclient to new release 2.9.0","message":"update constraint for python-zaqarclient to new release 2.9.0\n\nmeta: version: 2.9.0\nmeta: team: zaqar\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I3d3b26b2f06f413fbb14dc8e1fd84ccd1cb33b10\nmeta: release:Code-Review+1: Hao Wang \u003csxmatch1986@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I4b04da1ea95c2ec52727a9bee4921c2a097c9f1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d34d0d3f6be12ef69b8d7664ccd7cae58b1d15e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d34d0d3f6be12ef69b8d7664ccd7cae58b1d15e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
