)]}'
{"id":"openstack%2Frequirements~935634","triplet_id":"openstack%2Frequirements~master~If1a922e084e521a2c42daeb89e9d34f6fed68879","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If1a922e084e521a2c42daeb89e9d34f6fed68879","subject":"update constraint for oslo.rootwrap to new release 7.4.0","status":"MERGED","created":"2024-11-19 11:35:17.000000000","updated":"2024-11-19 15:51:41.000000000","submitted":"2024-11-19 15:51:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935634-new-release","meta_rev_id":"b148cb317c79b471e8bd2e6a74b73bff64b3f967","_number":935634,"virtual_id_number":935634,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-19 15:51:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-11-19 13:28:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-11-19 13:28:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-19 13:11:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-19 13:28:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"1159824f751b266268d899125f7e37a1eebf0c7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-11-19 11:35:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90a70dd8f45838f7da7967fe078de701ea088651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 13:11:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee437eef171841a9ac9e408b24143023\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8796a78d3db45c2a77efdddfda7f982 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca34eebe1a284092980955d041927845 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f838aab63ae749d7876d1c858e1b9f02 : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9df2bb9d7ee3479baa0babfc9c0e4234 : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8786d8d64c3d4b968b7c94b470033723 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/68ab94392461429b9f548aab9112e676 : SUCCESS in 3m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f251443279014addaf244dcefa0f0490 : SUCCESS in 14m 05s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3c27e2bb96384ee4ae5e5f42acf3cb08 : SUCCESS in 10m 02s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/12287870751f4b3895811efb0ba6fbbd : SUCCESS in 16m 53s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7f5c6a0d3e7f4b9b9a54cb8004a1e07f : SUCCESS in 28m 25s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/876ffa4622de4b348f8483151a795547 : SUCCESS in 6m 46s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/4232fd80c17b41fb90853ae550d150ef : SUCCESS in 5m 20s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/4895068e5857441a8e3c0146be385e60 : SUCCESS in 4m 56s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/480081a8cb18485d9beb37e1c5863c49 : SUCCESS in 13m 38s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/63af0bf4a08b4a548da92398018dfcf6 : SUCCESS in 5m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/54be3fffcb864655bf745cd905a511ff : SUCCESS in 5m 44s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/ec577760adf9469c9a137ea46ff8e4a0 : SUCCESS in 9m 17s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/1d2ca41a0c2a4ae9aeefce25a7753363 : SUCCESS in 13m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/839a3a8d6b3249c4a884165d0687ac7f : SUCCESS in 9m 50s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/17b0d0b6e38a4b07a037b8834edbb52c : SUCCESS in 10m 23s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/627ef1622e9e4b5b896cba0348639cc1 : SUCCESS in 13m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/b72ff39c9d6c4342a1461eb7a45e4b38 : SUCCESS in 2m 30s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f7f5be6fcecf427681294135afaf308a : SUCCESS in 12m 00s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/86b9c8b8c0ec489784b37f371dddc9de : SUCCESS in 4m 53s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/c4aa4cc23a6c4ef393a153467004fcd7 : SUCCESS in 6m 04s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ed566b3c929c4057b97cd9e5b0302ac6 : SUCCESS in 11m 35s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2110f97a08cc4d10a12a3fc5ae972b3f : SUCCESS in 20m 12s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/4ecf1345e7214033ab51fb286b4c0ba1 : SUCCESS in 7m 40s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/cbe06f8b0861480d92c7144416587502 : SUCCESS in 14m 48s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7c0cfede1d304209b63da3c80b03b371 : SUCCESS in 5m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/afd3d4f2a05748439bb54f31dbb9e68a : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4db9047488174144b587375563c06322 : SUCCESS in 28m 34s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6e50495fb00b4eca9edf17141dc1e1c1 : SUCCESS in 6m 27s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/97c25b84faa849cb94a6572a96582860 : SUCCESS in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4d5902200b7e4ea68497937315b02016 : SUCCESS in 7m 29s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/392949f9fc5242e5a5e6e1311aeb0c38 : SUCCESS in 18m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d0491052a964c93a0de156920e5b384 : SUCCESS in 1h 01m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c05c7ff3fe98453db175ce174738fb2a : SUCCESS in 1h 33m 40s","accounts_in_message":[],"_revision_number":1},{"id":"9e49d2d4278bf0e05b6254b86e800de93690c439","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-19 13:28:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ac8086b203a7136111ab54c53e8f3eef67289039","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 13:29:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4bc5786f19f96269c59c2e7ca9c42fa3a8eb1725","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 15:51:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec80d2f0f32243e7bfb7b453fc11c758\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01818de395da4852bc9bd20fb7d9c746 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e50072ddcd04ae09cae0c8842e5f03a : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb03268b75d945dab56e5243cb3c2c5d : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b471bdf0fc04baaaba8a4a7263698bc : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/57c8cbf9398547cd80bda5241cc83905 : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3dc82d40e61147a5889f153edf11e997 : SUCCESS in 3m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8f313f8a13904d8b9c75b64dceefba26 : SUCCESS in 18m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/592736ec29b0401d89cb39b6faead1b9 : SUCCESS in 8m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b897c13ae22a4ab3bcbd4a9d915711c5 : SUCCESS in 16m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/67ac4f8e6aac4a2c881c99ab650da1fa : SUCCESS in 27m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c5d0fd294f4d4b3ab27dbbdbdbbe98c9 : SUCCESS in 13m 28s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/262b18372a37482c9dd5f9039d63eb75 : SUCCESS in 6m 35s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/949b79031bcf4cd49771d17900ebea55 : SUCCESS in 5m 32s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/cd6d81acc89b42b096fbc2543b2984b8 : SUCCESS in 9m 36s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/15802017b1c444fc9ebd31fd6852f95c : SUCCESS in 16m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4a1a12f8a54146ffb47fb51fd3dc8ccb : SUCCESS in 6m 40s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/81f8ac3b8c284f87bb24e21d0d6c99e2 : SUCCESS in 9m 32s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/105a157943ec4d3ea34f703a62a3446c : SUCCESS in 15m 42s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/7f77506370394aa2a45a9f47b3ba404b : SUCCESS in 3m 18s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0b2bb2ba70e6497b9db084bb3a8c8c6f : SUCCESS in 10m 37s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2d6be82343814940a5e18ffef1096d7b : SUCCESS in 22m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6e47d479fcdc4724a3bcfdb2d2e6f92d : SUCCESS in 12m 09s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/e0b223b08fd44492a950ad1af0ae159e : SUCCESS in 14m 29s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/5b5272e35f5f43099c14241f907b25a3 : SUCCESS in 5m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/19b772f523654fb49478749fecdb7bb7 : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/08221abd81cc44ecb8939a029d49f843 : SUCCESS in 27m 26s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/22ce3fd9b9dc449a90fbca66bc2551e7 : SUCCESS in 8m 10s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/4798d7dedadb43618c9f2dfc78722ad0 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/91dc33e373364c43a0ba5fbcdf04eb58 : SUCCESS in 6m 22s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/4738046268054c8a936b48dd0939fb85 : SUCCESS in 15m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a6d449ee9ff400fbfe63b8ba9613c15 : SUCCESS in 1h 00m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7fbd75b17972481cbdf707fc596058e3 : SUCCESS in 1h 42m 13s","accounts_in_message":[],"_revision_number":1},{"id":"b148cb317c79b471e8bd2e6a74b73bff64b3f967","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 15:51:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30b66687fcc319857a62a24ebe14eefd0d524101","revisions":{"30b66687fcc319857a62a24ebe14eefd0d524101":{"kind":"REWORK","_number":1,"created":"2024-11-19 11:35:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/34/935634/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/935634/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/935634/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/935634/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/935634/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/935634/1"}}},"commit":{"parents":[{"commit":"8a8abf5abf6035066f9071f2baec1049a2e250ee","subject":"Merge \"Bump pillow to 11.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a8abf5abf6035066f9071f2baec1049a2e250ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-19 11:35:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-19 11:35:15.000000000","tz":0},"subject":"update constraint for oslo.rootwrap to new release 7.4.0","message":"update constraint for oslo.rootwrap to new release 7.4.0\n\nmeta: version: 7.4.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ie0199585471ca4cdfd12cdefbe372b9d4f560c33\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: If1a922e084e521a2c42daeb89e9d34f6fed68879\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/30b66687fcc319857a62a24ebe14eefd0d524101"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/30b66687fcc319857a62a24ebe14eefd0d524101"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
