)]}'
{"id":"openstack%2Frequirements~935641","triplet_id":"openstack%2Frequirements~master~I4cd3f21b7293dd1890ffda7a14d7f8d7132f71e1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4cd3f21b7293dd1890ffda7a14d7f8d7132f71e1","subject":"update constraint for oslo.serialization to new release 5.6.0","status":"MERGED","created":"2024-11-19 12:03:29.000000000","updated":"2024-11-19 23:14:33.000000000","submitted":"2024-11-19 23:13:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935641-new-release","meta_rev_id":"8f32affbd4b0498c2c4728e7b64322bd9316a111","_number":935641,"virtual_id_number":935641,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-19 23:13:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-11-19 14:10:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-11-19 14:10:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-19 13:47:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-19 14:10:01.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"2e9670fbd332e917a99ed8546c264d0105b81ea0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-11-19 12:03:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b38a88ec876ad3aa07c4f7a1afe1b93c20efe2be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 13:47:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39696fdd440b4e688a3b8f8cda8d32e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/837fc48feafb4da8bcf6b9c7f33de74a : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1207e364b1ea499fb1efce80836b2fe6 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f723a11d530646f7a592357fc6d6bc0e : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bfe7c186e10e48c795f6ae9afcd3aa0d : SUCCESS in 3m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d9ebe73b6da34a489300afe5bc4eee1c : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/69ad8bc2f57e4ea4ab34465cf41bfdf1 : SUCCESS in 3m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/abe6ff9faaa142fea160b14fb15f2aca : SUCCESS in 15m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/f1c6e8dcd53949f08473f752aa2fcbe7 : SUCCESS in 8m 40s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/2cfc567f35d64293a44719c46e994de5 : SUCCESS in 16m 17s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/3ffac6bb3a7e4669807787b5840618e4 : SUCCESS in 20m 30s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/b1cd4522fff14119abb2372f097dd577 : SUCCESS in 5m 28s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/5f262e2667374c2c8f71a05a8bb87634 : SUCCESS in 5m 23s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/308d6b2cd83440809dded5c943d6528f : SUCCESS in 4m 26s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/a8153dc2267a473b937ca4de97cecc19 : SUCCESS in 11m 48s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/0ff78faac3ff4feebd41db9fd2883eb0 : SUCCESS in 5m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/bcb9110d8a6a4345b49aea871f606d1c : SUCCESS in 6m 56s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/9f53b66b92714a43962d7102ef464bb0 : SUCCESS in 10m 19s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4d332c50361c406996fb881caceba8eb : SUCCESS in 13m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5182470048b441fe8a93d3d4c66a9faa : SUCCESS in 8m 16s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6e2184e818a34a3b8f79d14ae1c27903 : SUCCESS in 9m 13s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/2d2d88b4b257405b88bd823a48c992a1 : SUCCESS in 13m 12s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0b866e942e224b5d9ac3e49cad1015a4 : SUCCESS in 2m 04s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f51ed717b51a4e72b9b8dc997713ce63 : SUCCESS in 13m 06s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/f61e19876d6c4e18ae592d67135107ca : SUCCESS in 6m 36s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/c3ab25f2ae1643618843ca34d02955ab : SUCCESS in 7m 29s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/995b57dd8bac4e7c9429e80429ff85d3 : SUCCESS in 11m 51s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3df9e7b8273341b79b79b1c0f106c848 : SUCCESS in 18m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/216150052d1e4abcb4b86f85c695df20 : SUCCESS in 9m 49s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/21b10d30e5a646f489d25245974e67d1 : SUCCESS in 13m 44s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a7f18d5fe4834bc992c55551374df471 : SUCCESS in 8m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/18200d90c49a446795b19781eb7be619 : SUCCESS in 5m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c5a3c7992191461cb327bac271b7e489 : SUCCESS in 25m 11s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/6f40a8afc6b8400189fc78c675623752 : SUCCESS in 6m 30s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/03f4e34198f74955bbab704f846f7e9c : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9681025a48ab4215a93c66e8f09acead : SUCCESS in 6m 13s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c6a2a2fb4afd45828bdce5212bdf18e5 : SUCCESS in 17m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e9cf7d0b2d34271854e3058a273d985 : SUCCESS in 1h 02m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/261af14415814ceaa2c35b2b4aadb337 : SUCCESS in 1h 38m 08s","accounts_in_message":[],"_revision_number":1},{"id":"fc9c980d5c50d756386635d2b4ce406a841d07a5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-19 14:10:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"509ab473a3dea85660dc406624ef07ad80355b5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 14:10:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"295223240ff63c4727eb893db66c484e6946a3e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 16:38:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a1d5b77da5d45259a1117c1bc6b2836\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c9103cae754bd5bcc4f7c5008110c6 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/128c60eef89847babd6bf97b9231fd67 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b8ede43b7cc4ba0bb9ac9ec72114269 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0cbf96474378441094bfe20f7745be54 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/23ad4f2d84b84164a7f17dd66335c1e6 : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3d25324bc6e7463596948f5c9acc5a31 : SUCCESS in 4m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8d95accb9e2e4582a73c12670440e053 : SUCCESS in 28m 11s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/0728aee57e7544448b7c9128af0a73a2 : SUCCESS in 8m 36s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/318d32c11cbb4c2fbc96de2d57da4cf2 : SUCCESS in 15m 45s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ffb2a5dd1a054b52abb89d54e7b072cc : SUCCESS in 23m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/9c3be8bf76c04ec78f93cb14b1bb3597 : SUCCESS in 12m 16s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d18db60ea054484392f1df6d11e12f77 : SUCCESS in 5m 52s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/0e275c42addb43cfb70f69f12758f218 : SUCCESS in 7m 17s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/7e298b9ae4ca4477b682e00e749a68bb : SUCCESS in 13m 23s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/ef2c47ca74d0410b9253b9ea73a9e49e : SUCCESS in 13m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8c80f30f40d9469f9bdc83c141e52e6e : SUCCESS in 15m 06s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/54eb6f9b4ca8448e91313012b549af47 : SUCCESS in 9m 43s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/593b989c4c3b4131b5583b3d825ecc2f : SUCCESS in 12m 34s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/50b1562c77604311a2d639b33da3fcbb : SUCCESS in 2m 49s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/92141e4837a943db8f42436332643e6f : SUCCESS in 11m 04s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/7d50d0ce7ec74df285fb4f3ae2568adc : SUCCESS in 17m 43s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/6b86e58383b743f1bb396ddf2f08ad67 : SUCCESS in 8m 05s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/395e022ccc9e40d0aaad65576c3ad6a2 : SUCCESS in 15m 00s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/88ca7cec1a5a4bec81ca09b9ea0bb74e : SUCCESS in 5m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/28c998a6709240cd9ccf3a621f1d465c : SUCCESS in 5m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6bf7bbe12d82475db049f5e7c110f846 : SUCCESS in 26m 51s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4039d30a4cae4adf9a7f50fb7053823c : SUCCESS in 7m 21s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0a91cc7e40494387af8b9ad67cee58b1 : SUCCESS in 7m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/65e55ce9b70b4a2987cd651fdf9eec5d : SUCCESS in 10m 24s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/213b58765d624903a5154d39e3865a34 : SUCCESS in 19m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/842563ff866342aea62bda1762812ea7 : SUCCESS in 1h 02m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/48d56af5e1b44db2a4d2d4b5ab769208 : FAILURE in 1h 24m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e40658db1eaee0fe0d0d6ceb1285dd0618b6f808","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-19 19:44:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c61317c442e6aeb94a450369da41fa7e9cfdda34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 21:43:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97741574c3874393b7c26aa3cc9c8c6a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f9680d8be81439f94523f86a6620e53 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6afcd07aa1554d0a83e8152952b98dee : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/367e9e12944840ce88a20272ced8b807 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/88e7f3f8849a481ba5492195f8a826f6 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d0140c4ee0a44713b0359b43ee15f7ff : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cb533de0f9de4e66ad602c21e2ebeb04 : SUCCESS in 3m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/88c10368d5b746e6966dd69c7791622e : SUCCESS in 14m 29s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/cfaf55dab0844530bd451b551147e0c5 : SUCCESS in 8m 14s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/b87b900064a84efb9e709e473468ed51 : SUCCESS in 8m 34s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/c274ae5a53d24af89f961054b8d40130 : SUCCESS in 21m 29s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5dd045a70f1f4d4dac20ef600ddff471 : SUCCESS in 5m 42s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/54ee666b627445feb597172292a87a19 : SUCCESS in 4m 51s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/b6621fcb769d477f9d83096e2aff8dbd : SUCCESS in 4m 13s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/1589bf0545a84acb90a2422ff0327b8a : SUCCESS in 8m 40s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7b473a7fd41545e9afe66bc4fb9f95b0 : SUCCESS in 3m 41s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/e3c991be44904fe8bd8426309a882a2c : SUCCESS in 5m 54s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/60198df262b346ebb47e08f48fee3c17 : SUCCESS in 22m 13s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8cfd8666664d4cbb8e750949c61d2c90 : SUCCESS in 16m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1f238a1dcf63421d8c1faab9310b43b6 : SUCCESS in 6m 37s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c0101653482b4144acec8b3d70b601b2 : SUCCESS in 7m 26s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/b13480000614484d8a2ed5bfc3266109 : SUCCESS in 12m 32s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0460435bb8c34ab2bd55cebb879b0098 : SUCCESS in 2m 08s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/d4a2512fea1746c682e67b8bb2e9e8f3 : SUCCESS in 10m 38s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/fa98151518be45429b508c916a7e9824 : SUCCESS in 4m 52s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/368235960d82460fb9aa5ed46ddfc50f : SUCCESS in 4m 52s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/7c1d4c838db94eb48092f9cbe7856a98 : SUCCESS in 14m 05s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/eba833fa31384701a6b3a58398e9dcfc : SUCCESS in 15m 07s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/534431198f0746388ed8c338cec3de70 : SUCCESS in 7m 10s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/7c2fc672ff554fdfa7485d87d4044e54 : SUCCESS in 12m 07s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c9c3cb350dfd4ba5b1a4232d6b2b15cd : SUCCESS in 7m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a60944ce84e94c1190f7fc98d802a372 : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd71ca5ac0a14f91bfd67b024be1e05a : SUCCESS in 24m 24s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8bb57ecfbd434bb88c098e63d7ad1774 : SUCCESS in 5m 41s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/7228541f673c45d0bf94d5c5a4860de8 : SUCCESS in 7m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/96fad86b2b354d71bb23cc1cd3691f49 : SUCCESS in 6m 17s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/41016222bf14474da0783ca9cbe367f4 : SUCCESS in 19m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3993f7b9462c41d7b93d1db415b0c42b : SUCCESS in 1h 01m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/add25463861942e2b886ac2a324f67ad : SUCCESS in 1h 45m 31s","accounts_in_message":[],"_revision_number":1},{"id":"880d0029f7e0785c1b2516bad4c2bf2c47c12030","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 21:44:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9bd6d932a5a36f0581e55531bca3edd25b41f12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 23:13:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5d3024143e3467e988301f19d79ea35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0ff57c4bd648079cef5a21261f9f40 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c287949593fe42b881d5832cbb832cbd : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3f941133bd24969b1a0f82dbe4dd424 : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7b45f471d97d410f81cdc76105501042 : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a7e67613ca14a1cb0ad61f67b224ac0 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/37eb8b8e88e04e48948b4c2f965d07e0 : SUCCESS in 2m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6e4f4b7d9c4b42b9926ac8d50a4065f8 : SUCCESS in 14m 57s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e4c96b5dcc4941d8ab8aa331a272ed79 : SUCCESS in 8m 57s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d53acd26e10d4fcfb71cea99809121b4 : SUCCESS in 16m 41s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/89342358e52341adaca98b35304b14eb : SUCCESS in 23m 29s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/ea46c12cd36340d1835cde7e06d13a90 : SUCCESS in 10m 49s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a027d95896df4eba91ef8d360a8171fe : SUCCESS in 4m 38s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/dfed6d3b668e4cf694bc96dccba12ee3 : SUCCESS in 6m 01s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b8d75a7d92234040a99571afcb3c5a43 : SUCCESS in 6m 53s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/7980d92415a94c9ab4dba70a0c7c9b24 : SUCCESS in 13m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d2b2a88906fb4accb60a1f06fc108576 : SUCCESS in 5m 59s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/dbb9e4edcf304ec8b707bea1be37ca3b : SUCCESS in 8m 27s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/40bf8a2a8ce8421b9d750b34eee46dca : SUCCESS in 10m 53s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/dab512a14c4541088db378a4c0ff80a3 : SUCCESS in 2m 23s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f86e0345644b45898b244a6051473016 : SUCCESS in 10m 10s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1f2d90d3ddb344188acf9872851c6d14 : SUCCESS in 19m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1352dcdb3bbd45d7a820657bdbb69e14 : SUCCESS in 7m 27s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/5d942e3b2f024a87964d5f0a5d203abc : SUCCESS in 12m 17s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/d7ee8a3cba1a4e4087203196c881374a : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a8eef72ee0594e96956d43d19791e9b4 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2ab2a3cbafcd4e99bbc01748d28bfeff : SUCCESS in 25m 06s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/cd475f78d62c4514be663f9f9e27be55 : SUCCESS in 7m 08s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/b6fd3bd1c4e846589a54a25b17a63be2 : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/63f49889f80e4e9283279a5640859ac3 : SUCCESS in 6m 28s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/9f6f61b406a943028073ea6125163bdb : SUCCESS in 14m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d949c467b9da421eb98c282ff8593ed4 : SUCCESS in 1h 01m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75d9da48165a47eaa73dbf5d1dec49e8 : SUCCESS in 1h 27m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a03b48c457b903a4653f68ea4ec361cbc18c0528","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 23:13:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8f32affbd4b0498c2c4728e7b64322bd9316a111","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 23:14:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e33584c37bb43a49c52c373ab16bf05\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60b6e3f35a5a41028a837b52c50c0771 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7c97e824ca23e2ab20c5ba82589ad45aafccbd8","revisions":{"d7c97e824ca23e2ab20c5ba82589ad45aafccbd8":{"kind":"REWORK","_number":1,"created":"2024-11-19 12:03:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/935641/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/935641/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/935641/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/935641/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/935641/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/935641/1"}}},"commit":{"parents":[{"commit":"8a8abf5abf6035066f9071f2baec1049a2e250ee","subject":"Merge \"Bump pillow to 11.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a8abf5abf6035066f9071f2baec1049a2e250ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-19 12:03:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-19 12:03:27.000000000","tz":0},"subject":"update constraint for oslo.serialization to new release 5.6.0","message":"update constraint for oslo.serialization to new release 5.6.0\n\nmeta: version: 5.6.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ifcc8ac20457bb183546a7f75a7415e71c261c194\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: I4cd3f21b7293dd1890ffda7a14d7f8d7132f71e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d7c97e824ca23e2ab20c5ba82589ad45aafccbd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d7c97e824ca23e2ab20c5ba82589ad45aafccbd8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
