)]}'
{"id":"openstack%2Frequirements~935649","triplet_id":"openstack%2Frequirements~master~Ic5674e27760e65c740a5da896e3146e6a3fa56ff","project":"openstack/requirements","branch":"master","topic":"bump-yappi","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-11-19 23:12:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic5674e27760e65c740a5da896e3146e6a3fa56ff","subject":"bump webOB to 1.8.9","status":"MERGED","created":"2024-11-19 12:49:48.000000000","updated":"2024-11-19 23:13:45.000000000","submitted":"2024-11-19 23:12:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935649-bump-yappi","meta_rev_id":"eb211589782e010aa826ca162f6551df138ae5d6","_number":935649,"virtual_id_number":935649,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-19 23:12:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-11-19 14:28:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-11-19 14:28:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-19 14:21:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-19 14:28:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"fad877c0f953d55225dd79b5317d38e432985231","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-19 12:49:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5465115bf12625122b6d6149a847e1a4b0ef0cea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 14:21:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de66355811a44f09965dc24cd0afd353\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c4f0cc14c9c4942bfdc0661a2082305 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fa34adaade842e7a563ef3bb6165226 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5c112c5d8084505a4c8fd0f7e2c1326 : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9dd3e4e518d47fb95382b1101503506 : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/06a1cca061324e70aba4824fc8f95b83 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ada37924ac17434f87ffb4bd1e7a9f6e : SUCCESS in 3m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d1ff972203de49b79c24ced060802e5f : SUCCESS in 14m 45s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/152ed6643aa24a0283783cd538a768c3 : SUCCESS in 10m 09s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/94e7b368acff4076bea4daf4fa7a6a83 : SUCCESS in 17m 50s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/9481ad3386064a9c8e2989b2b1480103 : SUCCESS in 23m 30s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/cc2d8b8ae3b84c85af42c414dfad98f5 : SUCCESS in 7m 40s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3017a985bc07415ca9e228e3ed0fc50d : SUCCESS in 5m 39s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/ec96067e844a40c0ab0b53dc959d17a7 : SUCCESS in 5m 55s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4a37ff9b1f6c4b119155140e39cc3df9 : SUCCESS in 12m 27s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/735ef6b5b6c9428b97261bfaa0cf4e8c : SUCCESS in 6m 09s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8cadb037a7814c1ab4d2973cb5c50053 : SUCCESS in 5m 33s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/b2406bb0f7d4468bb46354fde99e64fa : SUCCESS in 7m 44s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/8ea42c4b4dd14380a63bbad02a46f48e : SUCCESS in 12m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8056d137f5e14631a80e04cedbd7131f : SUCCESS in 7m 43s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0bba8163375a4e83872abd0d6195da38 : SUCCESS in 7m 41s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d1bc643dc8f341bfa8b9354839e655e3 : SUCCESS in 10m 56s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/9b92f0db75bb43e49cb7927f996b1dae : SUCCESS in 3m 14s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/238a71e534f243b4ac7d58d600354918 : SUCCESS in 10m 02s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/d07f55290513446bbe695a6c4c612fdf : SUCCESS in 4m 37s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/9ee265dbf32a46aebcd9e4e9ab4ad951 : SUCCESS in 5m 20s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/304fa5023c194adf99e3ff87b1a60f41 : SUCCESS in 11m 37s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/de7a7e017b5740e09548b9617493c043 : SUCCESS in 16m 37s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/f918e00774d7496eb2c7c2c67a86bee8 : SUCCESS in 8m 32s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/89c12b0aa52b47c598f820b5cdf549e7 : SUCCESS in 13m 12s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/ec4ee3b0e06346c88f32392427707fdc : SUCCESS in 6m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/568b70f63d04416ca10ff10d1cbfe40a : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a5c54cd4ee74cbf89672af4c4af48ef : SUCCESS in 22m 56s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/3f8850cfa58542f5abdadb71446b7004 : SUCCESS in 6m 48s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/09d9403b07b14db3a669ad1ddc01ece7 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b15de7610874481fbda42aef09c10418 : SUCCESS in 8m 29s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/090c08f8b62044aeb9bc806a1cce1274 : SUCCESS in 17m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/682274f3201144bf9a8cfd95b7cf8932 : SUCCESS in 58m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4bee2434c0b14909989c8736db03d77b : SUCCESS in 1h 26m 06s","accounts_in_message":[],"_revision_number":1},{"id":"fbd6b29709375bd836f4bb2e6664fc26924fe5ca","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-19 14:28:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f7fd1b13dcf563df90da833f0410bea04fd3a89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 14:29:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d3b9457ab82f37b88077f888d45c9fe76acd473","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 17:21:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/67f7615b50604fe89088b00833bdb6ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b62b1deef10b40c7b4fc2d2f9c7dc2bd : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5dfe139201d492f9de0e3c20f659b74 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee25af9e478f4e2384a1066c1f0278d0 : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca69000a5c964a4e962e3171771719fc : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/580e7da998384d0e841461cb5da8a339 : SUCCESS in 5m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/63f31474eefd48118fc9b5df4f681c63 : SUCCESS in 3m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/529fd4aa99fd49ef915cc24f7f704e67 : SUCCESS in 15m 58s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/639078662e314869857d2220afd5b874 : SUCCESS in 8m 20s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/989ca06609ac4d39a20d2b5c88cca0ae : SUCCESS in 19m 58s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b43bedf9480e4e41a253cf54de0ebda7 : SUCCESS in 29m 41s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7b649c7ed9144e2f9e0154f757e359d4 : SUCCESS in 12m 10s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/8e72242ad17d4ac696d7a3cee5bbf072 : SUCCESS in 7m 06s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/85cc5921725e46349dad1bb8cdf18638 : SUCCESS in 6m 34s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/6387716d95c648caaddb47f18bbd10e5 : SUCCESS in 11m 35s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/54fd1d65c7734ece8f2da71c4c097cd3 : SUCCESS in 16m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c2652c19a12c49f1bfa9f4a78a421b82 : SUCCESS in 6m 55s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/cc328982725f4624afee03caf3111cbf : FAILURE in 8m 05s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/907972ce986b459aaa10ff16cf7f65fe : SUCCESS in 11m 18s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/e44fcf52cb0840fea053b741b211c248 : SUCCESS in 2m 54s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/86bb62d55a55484a912903f076b06089 : SUCCESS in 15m 48s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/cbacc6fe71104203b480a0f6edaf70e2 : SUCCESS in 22m 20s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2ec9cc84fd7c467790a7e30bd93c3123 : SUCCESS in 9m 00s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/b88c35120d984c58becb7d527a7d2aea : SUCCESS in 13m 16s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/0d6e8728aa4146c1b976a3b3d327e57a : SUCCESS in 7m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2b6cae6fc2eb480aa1577e6241d193a1 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1b96800f9271494ba0c6791d8bd611db : SUCCESS in 24m 41s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/218a1a4a12fa42178aa379bd55397469 : SUCCESS in 6m 30s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/122d0910321746c28a5796de192dcb8d : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fc9af27deb92493382d0f203b202a23c : SUCCESS in 8m 28s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/4f880810423048dfa830ed2dc71cc766 : SUCCESS in 19m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/821d03dffe4b41bd94f5e2c7c78540fb : SUCCESS in 1h 08m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0426bcdb775a48a08e4703586a4da8c3 : SUCCESS in 1h 51m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a345d94dc153d88e939715210dfbc5fc65b1e66d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-19 19:39:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74793212cf2848cbd7ba1a8c91901c6beb9cb540","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 21:29:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c30cf95f08144208d94e2491db1772d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f77b56919cb24a32afcab7efc07fa4d6 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aea499187734d40a2e47bd62acca094 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f96cc005bcd248c699a6cd0e3063a506 : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3bc735aa65ff42a29f93bbed0e2b53e5 : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a210caa203bc47a0be8c1fce46d0d7b4 : SUCCESS in 2m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/da07d3133a0f44719c9f8206846a275c : SUCCESS in 2m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f87c52e97f854b399ea1cce9c03d6b82 : SUCCESS in 15m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/221beb09cc5d4134a6ebcb1bd8a35f46 : SUCCESS in 7m 24s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/99166fcf4d0546f392f43b27adaa353a : SUCCESS in 15m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a5be038ca00c460d8f35bd70e008f2ec : SUCCESS in 32m 02s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5d13d54091124306b4791095260dcea9 : SUCCESS in 3m 53s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/171458e9254b44479f631d3a88e17858 : SUCCESS in 8m 16s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/e9c07b3edb8b47bba7de7346d9d52b2e : SUCCESS in 5m 59s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/17d1ded887eb4ff18cc5aae3d98eeb21 : SUCCESS in 12m 06s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/16cdaa32a5684b6cb8f5008aadb799d5 : SUCCESS in 5m 21s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/109aa82fd9df46858d4876ec7f601bbd : SUCCESS in 6m 03s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5d78d3089d1f4207acc9f454bad14203 : SUCCESS in 5m 16s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/84d58ce2101f4d419786ff4edbee0622 : SUCCESS in 14m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a197b0a66f654132929418eb943b5a93 : SUCCESS in 5m 51s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ae62adc9295d4869a2a2763edfedc6ef : SUCCESS in 7m 33s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f8949421530044328ac9d69d405b2e38 : SUCCESS in 8m 36s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/8e148b300d294c39823060ed8edc0514 : SUCCESS in 2m 29s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/3313041c6fcb45478c189f58a01f5b8e : SUCCESS in 11m 20s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/8f2f2dd5952f4445a7244f158f8d1a48 : SUCCESS in 3m 16s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/3100caf083a5488abf291b4d8ef6270e : SUCCESS in 5m 02s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/ccd2aa1e7028430cbde4ed11f8fb6ed2 : SUCCESS in 10m 37s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/a8c891f651304517bf6c9ffa2aaa6853 : SUCCESS in 20m 00s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/572781811f34492c875edc62d2ead23e : SUCCESS in 7m 22s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/bf64a9f0454c46f99037b9c5aed2f208 : SUCCESS in 9m 02s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/a2ae1d3ef5cd4aa4aa594e5f7b11f588 : SUCCESS in 5m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f3c889160f864871990d16eb2a715d33 : SUCCESS in 5m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/73cefe175305450eb47b99a84dae3fae : SUCCESS in 25m 22s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4b7de526ccfd4ac2a3a286f07e207835 : SUCCESS in 6m 00s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/46b90def2a09452cad89fa9cd8a437fc : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c2b81b8292e84b2d8b8f340e82017222 : SUCCESS in 6m 38s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/f4eea980407e42e19949edf1312ffcbf : SUCCESS in 15m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/762607b3e0414407a6edfe51a712edb4 : SUCCESS in 48m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5eb815444de44dd4ba9e1b08d478efb5 : SUCCESS in 1h 39m 56s","accounts_in_message":[],"_revision_number":1},{"id":"535c1333fa9d652812cff08455481e5ad2ea3110","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 21:30:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b1fa0f58e56b44040f41f668df745fe5aff4aec8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 23:12:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eab3054c4b847f19c2ea15a576cc745\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66fd92fe8bd646d69b08bc20341ce8d4 : SUCCESS in 2m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5589121ede7e4e4f8139e513760e86c1 : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b0fa4e6091c42b5acfd84efa4a95438 : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8a828b7e08d04e0a8718dd1c4238ceb9 : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3542f3b6739049e999e4bfc5df1dbb00 : SUCCESS in 2m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/afd8bf4120564398838b4f910622eb52 : SUCCESS in 2m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/35ca5975e22b4dbeb0b8d4c7881cb031 : SUCCESS in 14m 47s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/7fc8be3e5f95403ea0204b5f1e48b2e0 : SUCCESS in 7m 51s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/9b388862640b41048fa8b5b482542ad3 : SUCCESS in 15m 09s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b7b98182ee984079ad61e600340c7c56 : SUCCESS in 10m 18s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/4a0f4e11886e4bb1a0006f07a271c3c7 : SUCCESS in 9m 50s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/35c921510c7a407ea1505608b1ca113a : SUCCESS in 5m 04s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ebf586a7db5b4c509bc232199b05b224 : SUCCESS in 5m 29s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/cbf84c421a3a44c380219d436d485002 : SUCCESS in 6m 47s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/87bee2f433fc4aa5905930e3f5bf0cfa : SUCCESS in 14m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e037da99935d435aa153f20e63c8409d : SUCCESS in 6m 14s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/6809adf5546b45c2a59a6041e3d25210 : SUCCESS in 8m 10s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c504c76313d0455891abb30e8b7b5bf8 : SUCCESS in 12m 27s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/f78cadab726048729889721f99db3de4 : SUCCESS in 2m 01s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/dca1521ef34e416e8f3772d35b5317b8 : SUCCESS in 11m 13s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/f8f2249640ee4e7b8e1758ede925ab3b : SUCCESS in 16m 51s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/0f1479f407a84d3f9dccf63a7d6ad5e5 : SUCCESS in 8m 29s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/0245fe71cfbf460096db3db55cc41aa0 : SUCCESS in 12m 50s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/c7adfa3488cc4aad931674471fe00ad5 : SUCCESS in 5m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cba93017dc88495c9eca2890b1968589 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eb3c9acc4f77405aaec27d545f423e8d : SUCCESS in 25m 56s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/f5a5eeeda5fd4cb78da9ce65016c6646 : SUCCESS in 7m 56s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e217d8fd63e44075a0dae7cdd15d9f56 : SUCCESS in 6m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f86097999cc04f978775d78e182cb23a : SUCCESS in 6m 58s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/26c02ba711694ce18d9471a5dc87439e : SUCCESS in 17m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db145f4cce084e5a9b620b65773eb6d9 : SUCCESS in 52m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c9920f7cb0342d7975b9f0aa666e4ca : SUCCESS in 1h 33m 29s","accounts_in_message":[],"_revision_number":1},{"id":"47c0a4d7920cee04e8e62b0725a16b0883c7046a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 23:12:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eb211589782e010aa826ca162f6551df138ae5d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 23:13:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4a971602790425fb384b16c7f9fc1ec\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9f36c251108426698a688404e417ab0 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8f496a0bb14fae37b289ea87021644815e4c12d2","revisions":{"8f496a0bb14fae37b289ea87021644815e4c12d2":{"kind":"REWORK","_number":1,"created":"2024-11-19 12:49:48.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/49/935649/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/935649/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/935649/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/935649/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/935649/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/935649/1"}}},"commit":{"parents":[{"commit":"8902d9c6d1398eb10f85833202108259467464e0","subject":"bump yappi to 1.6.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8902d9c6d1398eb10f85833202108259467464e0"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2024-11-19 12:48:46.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2024-11-19 12:48:46.000000000","tz":0},"subject":"bump webOB to 1.8.9","message":"bump webOB to 1.8.9\n\nthis change bumps the webOB release to account\nfor the removal of the cgi module in 3.13\n\nChange-Id: Ic5674e27760e65c740a5da896e3146e6a3fa56ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8f496a0bb14fae37b289ea87021644815e4c12d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8f496a0bb14fae37b289ea87021644815e4c12d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
