)]}'
{"id":"openstack%2Frequirements~935657","triplet_id":"openstack%2Frequirements~master~If2a7f24866edd0ef244b671f8cb9b28f127eb56c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If2a7f24866edd0ef244b671f8cb9b28f127eb56c","subject":"update constraint for oslo.i18n to new release 6.5.0","status":"MERGED","created":"2024-11-19 14:03:26.000000000","updated":"2024-11-20 12:18:29.000000000","submitted":"2024-11-20 12:18:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935657-new-release","meta_rev_id":"d58590b69407a8dae825bae26c773c31b6db8a5b","_number":935657,"virtual_id_number":935657,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-20 12:18:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-11-20 10:19:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-11-20 10:19:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-19 15:52:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-20 10:19:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"0113e2cc21967f6fb8467efe0f5f1355dc762d81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-11-19 14:03:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f6308d2f7d49a9c0983cdd2f61fa56343a3430a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 15:52:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76c6c9ce2b914feeb4b17118b0c33a58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95afcd875e104786b9a7d01a55cb3f99 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/568f805a944d41889fde1d0a0053f41d : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed7c7655c0024efb9746da2e859f735b : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a507f83bb02a41bd80ce88b2aa26f7aa : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f1aa388a2b9145368f4247f6341361bf : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bea600bec78d40c6a2e7221e1e886bc3 : SUCCESS in 3m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dadac1abefb949fbb25490850b4414c1 : SUCCESS in 15m 14s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/45c3920b4ec74940bd43c73045e65ef4 : SUCCESS in 10m 43s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/7af583e40ccc429aa3858668e7566180 : SUCCESS in 18m 08s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/7ba6ca62f23a448193b8fe836c2b4316 : SUCCESS in 35m 06s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/004c30713f8a4cd4baf34edfb223d83f : SUCCESS in 8m 59s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/3053f13586bd474ea186624f3aafa327 : SUCCESS in 6m 11s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/771d1141c07a42b0b17e9556b2e49cf8 : SUCCESS in 5m 57s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7000d419852c4686895b7698ba7bcf62 : SUCCESS in 14m 26s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/30d28ddd1c794a8a8a52cf9acf50c7a7 : SUCCESS in 5m 04s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3c42bdfe72a5402dbe87ab5286bd7abc : SUCCESS in 7m 00s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/0e808d4479e14f0dbdba40ca1f7b5dd2 : SUCCESS in 7m 14s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/597f9e68d92e4813acae46d1682a9abe : SUCCESS in 11m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/599f2276d66347779a438f8efe3aa225 : SUCCESS in 16m 42s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/06f490bbc52b4beebd97f1093313c189 : SUCCESS in 8m 12s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/17fd02f7e0eb4aebb9a57123d3bd1c7d : SUCCESS in 12m 07s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/343d80afe8064369b341f78bad45bbc4 : SUCCESS in 2m 11s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/8761d43523de4a45becebd65a0b374a3 : SUCCESS in 11m 39s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/6ecd88dfe5b04fc98151aac18cbdc087 : SUCCESS in 4m 26s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/c94e0056020d49999751a2d3b99067f3 : SUCCESS in 5m 19s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/29499bb0de8d4f3b8da8165037bfd7b4 : SUCCESS in 10m 10s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/affffc0319ad41f1bfac3eea2bd3d4a6 : SUCCESS in 21m 58s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/bd3da5912af748e48dd5d8ccd96044f9 : SUCCESS in 9m 21s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/0593fa0273284d26819179f45a232a24 : SUCCESS in 12m 44s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7f890ead74c345aa880e6ebb79c5ede8 : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/871198feb95949f8bba2a0959ce20a36 : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e97be31ac0074d989a99530a95f23e9f : SUCCESS in 23m 40s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/fae4a67c58934480be3badf5a67395a2 : SUCCESS in 6m 19s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e1bba4cfbeed42b09876d0ff977371e6 : SUCCESS in 6m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f6f58294182c40ba8fd1c1e217ad3ed4 : SUCCESS in 7m 46s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/015b7fc106b24c7a86d309dc6c62f84b : SUCCESS in 18m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b3d82d35fb94734ad2d3833e9dfb91b : SUCCESS in 1h 06m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66c3b82573814343bc388ab1e228c318 : SUCCESS in 1h 32m 14s","accounts_in_message":[],"_revision_number":1},{"id":"9a2e203645da991c1394965f6e340efef972b595","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-20 10:19:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a2683189c02dd7dca4ca6ef5e4aa06f89039d260","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 10:20:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd369baa9c69bc4cefce653d841321c554cbeb37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 12:18:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad755ef02335441f8bb8270c5b3a0115\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9538ec5c4a849e697ec9dd85166f2f4 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7796f6ea5134b348324abf92503defb : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cd91e2fc36448719522f8d208abc1d1 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6bebc3b154e0439f824fadb57c1ee601 : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5195b2cb6e5e47d082ba4f96a34e7aa3 : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/12b5b6a2db314a099393dddeccfdce66 : SUCCESS in 3m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d94a49a56c7b4c7a9d40b63a6b851178 : SUCCESS in 12m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e341d776d9374728af97ab568ee70376 : SUCCESS in 9m 03s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/6467f05680bd4c2488eeffb8f178abc1 : SUCCESS in 16m 04s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/a872d25da6504273a16226ba11aba349 : SUCCESS in 21m 48s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/c6bd952e705443f0b2582d4510fd0c61 : SUCCESS in 11m 24s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7bf2a79dba0b433c90d563400ee2a5d8 : SUCCESS in 5m 26s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/ae7b0ce47f0b425187e636019d773967 : SUCCESS in 7m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/149bb06d8f484351abc62d1e5d390448 : SUCCESS in 8m 25s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/f533d4494eb0475891f95f84b6d49258 : SUCCESS in 12m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0e3040d37b3149c0855923398ad36010 : SUCCESS in 7m 32s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/fa7b5c85c04e4ae398a6ad7d68b11026 : SUCCESS in 8m 32s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/daa6471556eb43169d89de0dda327157 : SUCCESS in 13m 29s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/623e6667967e447b8a8cc0efec750615 : SUCCESS in 2m 39s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/6191ab8d47b7433eabd00ea087331189 : SUCCESS in 11m 22s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/127eaead5ca144739b76f44f15be238a : SUCCESS in 20m 57s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/3bc3642028b84d86ba680081eef1c6ec : SUCCESS in 8m 59s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/571da4016be344d8b765aaebd2076029 : SUCCESS in 15m 14s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7dabfa570637430392a5948939fc0de2 : SUCCESS in 5m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/009b3d8e4e574933b9220b90abd9f640 : SUCCESS in 4m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/70aa03e2c9ef4994ae176e1f2f7e6bae : SUCCESS in 21m 30s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/b1e835d0ef3a4da2911d515b045a9f26 : SUCCESS in 6m 59s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e89f5bdca242466fb62dbddde69390eb : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a7f684cf728e400d908d124db208b41a : SUCCESS in 7m 36s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c4dcb6b873704563a27b3e69f0f0aceb : SUCCESS in 15m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7734a696b42144758d664ae5291d64a8 : SUCCESS in 49m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b48dae47a40041f684befe4da54c81ae : SUCCESS in 1h 43m 54s","accounts_in_message":[],"_revision_number":1},{"id":"d58590b69407a8dae825bae26c773c31b6db8a5b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 12:18:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f53cf89504c00975b27e7526e7950db451b53eec","revisions":{"f53cf89504c00975b27e7526e7950db451b53eec":{"kind":"REWORK","_number":1,"created":"2024-11-19 14:03:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/935657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/935657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/935657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/935657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/935657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/935657/1"}}},"commit":{"parents":[{"commit":"8a8abf5abf6035066f9071f2baec1049a2e250ee","subject":"Merge \"Bump pillow to 11.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a8abf5abf6035066f9071f2baec1049a2e250ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-19 14:03:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-19 14:03:24.000000000","tz":0},"subject":"update constraint for oslo.i18n to new release 6.5.0","message":"update constraint for oslo.i18n to new release 6.5.0\n\nmeta: version: 6.5.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I21b9a9f1b4b54d44aa7705c2d8345484e344b3bb\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: If2a7f24866edd0ef244b671f8cb9b28f127eb56c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f53cf89504c00975b27e7526e7950db451b53eec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f53cf89504c00975b27e7526e7950db451b53eec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
