)]}'
{"id":"openstack%2Frequirements~935758","triplet_id":"openstack%2Frequirements~master~Iaa5767cf8fbb3b5c2acb530993a77ec1efad066a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iaa5767cf8fbb3b5c2acb530993a77ec1efad066a","subject":"update constraint for stevedore to new release 5.4.0","status":"MERGED","created":"2024-11-20 10:27:44.000000000","updated":"2024-11-21 13:45:54.000000000","submitted":"2024-11-21 13:44:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935758-new-release","meta_rev_id":"f19f8dbfe9afbdeb2e5ca6961a0ada28afa9f229","_number":935758,"virtual_id_number":935758,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-21 13:44:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-11-21 12:18:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-11-21 12:18:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-20 12:16:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-21 07:16:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-11-21 12:18:04.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"f092b27a3e3d9de3791cc4c6f8bccf525249eaa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-11-20 10:27:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee02f0b06ee038d2e147fe65e005783a01e2e963","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 12:16:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36ad5aabe29b4058918d13cf1a437fd7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fd3dca3ca8240ccbacc9673b43d6423 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b238aac054c4599874f8ec9ebeb44cd : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e920416bb9846a4a82c997afaaefc37 : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ad573564226243548c6fb42ba17a3dcc : SUCCESS in 5m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/20f2a711e63746eeaf8ed0fb8c13dc40 : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ab6872b5d1c340c5b4a310c01217cf8c : SUCCESS in 4m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/df8dc082402942f4badf027192c567ec : SUCCESS in 14m 21s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/c1aad8045b0240f2b98eed2b27681c2f : SUCCESS in 11m 15s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/093224497ba645c38ad4bc0ecdd613a0 : SUCCESS in 17m 02s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/b4d09fd14d10491b89b4ce1b5edebcf6 : SUCCESS in 26m 53s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f2235392e56748c3aa1c65bc156f284a : SUCCESS in 5m 39s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b0b1e5caca8a4648befdd77b47bd5111 : SUCCESS in 5m 34s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/35a95f4a394d422290543a6bcee9db6e : SUCCESS in 4m 47s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/36644387f6e94472ac20871f5d981381 : SUCCESS in 13m 54s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/73f227fc0ee549e79817c11b42b99189 : SUCCESS in 5m 03s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/7ca4855ed35d4191ab6f80c0a728bc6c : SUCCESS in 5m 39s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/77b0b5dc7dbe4a9b8d360c8f2eaeedae : SUCCESS in 9m 12s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/c853b74372334beba268c4a5c69b2cc9 : SUCCESS in 13m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/faa8cc1ceb13471596b3f747442f9cfd : SUCCESS in 9m 17s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/9484c5a9530c4a0bbafe2b5eba90b6ed : SUCCESS in 8m 52s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/1746d7077ad74a838d03d9043010ae72 : SUCCESS in 10m 38s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/0fec84f62a434bf5b3af9cf03da25080 : SUCCESS in 2m 41s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/f228a0e9438642819aa5529da03209c0 : SUCCESS in 10m 32s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/872c99630b1d4f43ba3e7c5b9f4a6352 : SUCCESS in 4m 52s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/649123616c64426d96ef9360a1734129 : SUCCESS in 4m 58s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/11bc5529a92a4f42b499a8d237c8abb8 : SUCCESS in 12m 38s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/0a0109aacca84d28b735ca31e41dc99b : TIMED_OUT in 1h 00m 47s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b67630cd2bdf46128561ebf4f650b4f1 : SUCCESS in 7m 45s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/de2488bea14446f48b6f57814ca5208e : SUCCESS in 12m 05s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/6244aa4980eb472d9160fa0078fae506 : SUCCESS in 7m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/051ad4c734754ac5b155bd500a0633bf : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/483a2bcf45164cbd9dcad24639e9d3f3 : SUCCESS in 26m 06s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/4c5896afb0f640ac9394079ec507fb34 : SUCCESS in 6m 22s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/e5ba334558ed4794bde1a5942209abd5 : SUCCESS in 7m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8870127ac31a43d183bd7ed616ec9b61 : SUCCESS in 7m 55s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/36d728af0ef8466bb590ee3d23b94160 : SUCCESS in 18m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66ced8a6d4c4402cb5480cb313c469d5 : SUCCESS in 52m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50f4db3f37ef4bc9a59f5d4bfc693c1e : SUCCESS in 1h 42m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a3790597ff08c6b900636df331d355096758f701","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-21 07:16:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb60edff899fac3768b27574de210388e0770d20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 08:59:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2330d52e9a544d6a8074fcd068254c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/525ed50e3be145dc897e44799c90fbdc : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd3994fbcae24b2594d1826ca1ccadff : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32c37959e48446a3a72be0ccc7ad933c : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bef7941ba4c845a2b5f3fe72ef78987e : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dec1f63ad5f34d19a96be24fb4170262 : SUCCESS in 2m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/45c2d9ec879242f9a5f5ba01e755e6bf : SUCCESS in 2m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/40caa0d56e174a3a8bd17262181d3eb3 : SUCCESS in 9m 24s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/e4043ae3ac38472884ab6639d639352e : SUCCESS in 8m 16s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/920329d6ceb4449b84b8ac2122cf214c : SUCCESS in 14m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4114294dc1db43e88d6cbe8cdc0a596b : SUCCESS in 18m 42s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/06b446bccd2646ec989d985f80724948 : SUCCESS in 7m 23s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/c8a39cf2aa424b3b879a282f11556612 : SUCCESS in 6m 01s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/7177bcebec124482a8cb1f9898b4e19a : SUCCESS in 5m 24s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/f35450f4e8e24bc388213e0df9a5cf11 : SUCCESS in 11m 33s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/e0bf132313d74c0686cc969225c6c512 : SUCCESS in 6m 19s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c89cf15bdc8d46d0b01c7d02f50ff6eb : SUCCESS in 5m 55s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/83e196d75be740829b0fcbf876db5083 : SUCCESS in 6m 59s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/4ba437d3ebf547618f626ff35a89b832 : SUCCESS in 15m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/07ea3df06c4f407e8d703ad56b5138f0 : SUCCESS in 10m 04s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ae81030418eb4128a94318ee5faf9c5f : SUCCESS in 21m 54s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/aee3f1b8a307431faf47a67ba742bba7 : SUCCESS in 12m 15s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/52b511bd78f7431e8d4606128bad6c68 : SUCCESS in 2m 10s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/de89ee30705d489fbde2f12a380edeb6 : SUCCESS in 10m 29s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/560db43c9f1543f8aa777dd83ad36857 : SUCCESS in 4m 35s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/de28f7f86b844ec9be038b0155b3ce7c : SUCCESS in 7m 45s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/119f09ca161e4b5699b7e56f5793bdb0 : SUCCESS in 10m 06s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/ad2439cedcfb48e6b2054061da8b35c6 : SUCCESS in 19m 01s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/627826baf65744b8a40e1fe7187f61ba : SUCCESS in 7m 03s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/823f44ac281148ecb5204286cc6f826b : SUCCESS in 13m 57s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/162a8635403c497db1df44468eeee99c : SUCCESS in 5m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5ebd0b52bc8044a5a512f44a4b270a86 : SUCCESS in 5m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/54bdf8ff96164a0c9a5b8a11df146621 : SUCCESS in 27m 21s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/8f2150c7da7c49749f7dce5cee8810b5 : SUCCESS in 7m 01s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/827e8ecf5bdb4f4dbf035b2f28e02f89 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/07e19e677e4542ee903a3bf047c48685 : SUCCESS in 5m 21s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/08e7b81cea6d48749eb36b6cb7617d64 : SUCCESS in 12m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec8399f3a8584d169513ce45603d52f9 : SUCCESS in 1h 08m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef5a559899d74ac3a5a653e5c8726487 : SUCCESS in 1h 35m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d3b02cb742bd2dbda230dab04ec2139d86c0d10f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-21 12:18:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"227ecdc916f35abdb64076540edf0edba4d566cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 12:18:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a3016f5849a962e9a28a32d446847035b6ec3931","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 13:44:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65229f9c660e4064a78037d14f4b9190\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1092ade382734cd280f4df813e889570 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66466a895e1347b5aff319cf279ebb63 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f753e842a0ce4db78737a66f7612b329 : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b05fc300a5047368085c6b57a983788 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2c8f944cf044243b8eb8936b55aae54 : SUCCESS in 3m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/67aeddd22db045fe96661e4a46e1f407 : SUCCESS in 2m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/551f07b864954bae98679347d6346d15 : SUCCESS in 14m 49s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/32f71300a11a4a07a0892617cc479b03 : SUCCESS in 8m 30s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/d73a282dc1f7463a97a634aec591f528 : SUCCESS in 16m 24s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/487ada2efbdf48bd95d1e70b03704ef0 : SUCCESS in 22m 49s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/e27d4490d02c400687935e6f5b685261 : SUCCESS in 13m 29s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/195355f1c29f4ae196c3f026a4a34cd5 : SUCCESS in 7m 05s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/90a7898978c641b8a4b4e392b37ed30c : SUCCESS in 4m 31s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/71dd6d9cd8fb48dc81fba279160039b3 : SUCCESS in 9m 07s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/de1d0ec3983a43c4b61cbca28fd044d6 : SUCCESS in 9m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2c8d7cb6a95a4255988901334087f4b3 : SUCCESS in 8m 57s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/ddade732830a4e5fa777cb5c3267f6ff : SUCCESS in 7m 34s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f8e8620fed364d9eb9785d32b2e6e51a : SUCCESS in 12m 37s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/c9cb2b03b4d048a39137f136bc696daf : SUCCESS in 2m 21s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/c243697ba8d44619b25253709ce38fbe : SUCCESS in 8m 59s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/4526c6d29f0d4464a6a374eb2aa9e3d6 : SUCCESS in 18m 22s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/1de5aa67a9854e87b993ea35e1d92929 : SUCCESS in 7m 57s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/906f0c5728a143608c2dd20fff811f9a : SUCCESS in 11m 54s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7f578f0b645649838e2bb1fa65cc6826 : SUCCESS in 5m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a5ad2f94b6734d15a62037ebf98cc6a7 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/efeb80f3a4864a579ae4322e630baf28 : SUCCESS in 27m 06s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/d15d39ef09ea4a098e2db6c96bcf277c : SUCCESS in 5m 46s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/a87ec93bf80745feae3c12368e084cb2 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/73f76c776edd401e92cbb2171150b956 : SUCCESS in 5m 43s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/d12aad3c8a624378a4fb7b38ef75c85f : SUCCESS in 16m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7347fdd8ad9943fc9ab0b216752b67a4 : SUCCESS in 1h 08m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f6e9257be63493a86f68a7f989fd239 : SUCCESS in 1h 23m 56s","accounts_in_message":[],"_revision_number":1},{"id":"90e7377dbd55cf39133353091d1cbd352ff99245","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 13:44:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f19f8dbfe9afbdeb2e5ca6961a0ada28afa9f229","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 13:45:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8701da29d5464da9bcd92b0fb373b4a4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fff57e8d4cd04c7e80ed50b557b2a58d : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5595f252475461afa78665e18acbb6eae2ee20f","revisions":{"a5595f252475461afa78665e18acbb6eae2ee20f":{"kind":"REWORK","_number":1,"created":"2024-11-20 10:27:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/58/935758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/58/935758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/935758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/935758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/935758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/58/935758/1"}}},"commit":{"parents":[{"commit":"9e67c176ab1e2aa55b0c8bde0fb94121edd8907b","subject":"Merge \"update constraint for oslo.serialization to new release 5.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e67c176ab1e2aa55b0c8bde0fb94121edd8907b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-20 10:27:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-20 10:27:42.000000000","tz":0},"subject":"update constraint for stevedore to new release 5.4.0","message":"update constraint for stevedore to new release 5.4.0\n\nmeta: version: 5.4.0\nmeta: team: oslo\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I21d3814655fddf444c54ee3c50888cfa6050abe0\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: Iaa5767cf8fbb3b5c2acb530993a77ec1efad066a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5595f252475461afa78665e18acbb6eae2ee20f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5595f252475461afa78665e18acbb6eae2ee20f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
