)]}'
{"id":"openstack%2Frequirements~935759","triplet_id":"openstack%2Frequirements~master~I59c46c35d4435e0ff15cdf7b6a13ba9507c094ff","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I59c46c35d4435e0ff15cdf7b6a13ba9507c094ff","subject":"update constraint for cliff to new release 4.8.0","status":"MERGED","created":"2024-11-20 10:31:42.000000000","updated":"2024-11-20 18:35:04.000000000","submitted":"2024-11-20 18:34:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935759-new-release","meta_rev_id":"0b4f747c105d3e5ef6a49a428b25fe593921c026","_number":935759,"virtual_id_number":935759,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-20 18:34:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-11-20 16:35:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-11-20 16:35:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-20 12:12:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-20 16:35:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e3347179999836161dd00c532a4cbae96e210577","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-11-20 10:31:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6d6eb14e21a5e798b77dc6fb4db6de097b22171","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 12:12:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3434653728da4c7c82a4b31b9d82e493\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/693419c64304433999396a476389da40 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bd0157f88ce41ad92b1a0c9fa27e854 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdfa7697f9d64eb2818d8bd0733a93ab : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/91753c10b0b04163921cf9b6e323e54b : SUCCESS in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce7deb79c4794a5c916bf5c235840d55 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cec9aa54ec744a548d1f4ec0ef4c83d7 : SUCCESS in 2m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e59cdde78c844a51b7a087ffe15b4e13 : SUCCESS in 16m 23s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/3f4842533cd34eee806df5609042324f : SUCCESS in 9m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c662e7bed92d4b37b62736ecd99e211d : SUCCESS in 18m 35s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5a0f4c60ddca4faeb641835f44c09460 : SUCCESS in 22m 45s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/5c6f580d04ba41b1a3c4d4e8ac2bf937 : SUCCESS in 5m 15s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/50735b455b1b4d09939a6be691aab84b : SUCCESS in 5m 18s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/5db28c953cc1432ab2577185fb902d29 : SUCCESS in 4m 34s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/26b2672937454f3ab663505f52c7f6c1 : SUCCESS in 11m 57s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/7258394bee244cbaa63154fd4cfdb98a : SUCCESS in 4m 56s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/3b0f097d8ec543388da7badf424ede4b : SUCCESS in 6m 23s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/1abc65500eae4955bdfb8a0190297b9a : SUCCESS in 7m 42s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d93ede7f8c1d4a788e08119fa70bdf35 : SUCCESS in 13m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9017ce1827d5400a9aee326a05cbbb61 : SUCCESS in 6m 59s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/357f175770d94cb7bde130a0cf02f6ce : SUCCESS in 7m 30s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/f1f3d2e9be9042a1952b695ffa70c453 : SUCCESS in 11m 13s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a2f4a867de154b03a907bed09486703f : SUCCESS in 2m 39s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/b057bdee6ef54c92811f40a1972a85fb : SUCCESS in 9m 11s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/b4be7540e8634521a69afa36fd226db5 : SUCCESS in 4m 23s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/b49a400184074eb99a55e4a524bffbd0 : SUCCESS in 4m 55s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/79598d5d18d84d448362a7a39785bbee : SUCCESS in 11m 09s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/295bdc07707340f0bcc2a47cc264e92c : SUCCESS in 21m 41s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/a922442a90114809a54e8e06d90f616c : SUCCESS in 7m 25s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/85115979fc64426bbcd35d7c60f00e2e : SUCCESS in 12m 53s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/cdd7c76874954e04bbb3e17fe0648f02 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c92a3ebd8cf6456693400a5b3908ff07 : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5079a4fea4174fe1bad089cb90e7ac5e : SUCCESS in 26m 16s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c9d4983338c141c5a31a70a0e8cacc43 : SUCCESS in 5m 29s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/eddba3fd8f75436f80f2967e23e2d425 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55f4f2a7c745420f9978980d73c4fef4 : SUCCESS in 7m 16s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c3cf624a19054744b59b1e435fd6949c : SUCCESS in 14m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c412dcfecf7c477881789e224d384003 : SUCCESS in 57m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/389d4ce56233489ebdfc73faaa76ab2c : SUCCESS in 1h 16m 58s","accounts_in_message":[],"_revision_number":1},{"id":"074595c2d2ade4718c9a7e4ffcb066646c25aa47","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-11-20 16:35:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5be40d7aa59fb3880a152c7ef959fae4aacfd41c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 16:36:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b45372544a0ee3c5ce0e4aa6b3e14f7d4e76df52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 18:34:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/375970186edb44be84c8674169de7650\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73eec39c71b84fb6a0f25ed76ffe3a3d : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb9f9426532f4a9d9d960ee874d40cf1 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3900b1ef0fba4af58bb2eda9cc9c788a : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f204e07c88d9437e8034909d513b4a01 : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bdffbcd1b1884defbce0510ec0a9531c : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/629207f64f124a02be03f5feca261d1b : SUCCESS in 2m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e84e3a92d1714ef8bd71a46dc365e7c7 : SUCCESS in 16m 00s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/8cda59f810ec4a9f8f117fad9b084afa : SUCCESS in 10m 48s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/86857c90f874469090f992939a8cc035 : SUCCESS in 8m 21s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/4cb97e0f6af24e028df6a8c406e809aa : SUCCESS in 23m 23s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/fd9d7b196fa64823a478e4f81f3cc0f6 : SUCCESS in 12m 02s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/83e48ffe16d349dca61b7cc677e33730 : SUCCESS in 5m 45s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/7f6ff0c0a12d4469ae5cbe6be2c12779 : SUCCESS in 6m 43s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/38e8ace22b5d442cb16b82fa88b7b74a : SUCCESS in 9m 00s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/dad237605ab24e219e5ce24551847191 : SUCCESS in 13m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/140a484437d946a3b1c51cbeea17ad1c : SUCCESS in 7m 00s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/03f6d0fc0b404102b5a959e0da4b1f11 : SUCCESS in 8m 16s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/956469a317834778b0b0697589aed199 : SUCCESS in 11m 15s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/957f2fd69bc94fd38fe815273cab00e7 : SUCCESS in 2m 48s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/0b66bda134f8438c80b8cf6f9c52fd3c : SUCCESS in 9m 01s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/86cd710a89a84182b6754a8df36c300d : SUCCESS in 16m 33s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/456aa15d066a4e21a6b7b6af8099ae64 : SUCCESS in 8m 57s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/2896616f09b84e3a9fa42621e5d9e486 : SUCCESS in 13m 47s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/08b57d1eaafe42c58833cf1eba4a5f14 : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bda9c76432f24e9ebe03e2ecf25ce15e : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b0ba2b5f1d6d48a2ad12f155e571c75b : SUCCESS in 22m 11s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/486f8a1394d3441f88ac8f56c41129c0 : SUCCESS in 4m 14s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6e3888dec3a94cf3ae4ad750ec454def : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/263bef6dcba64640a76356d420d5ef68 : SUCCESS in 5m 31s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/6c5978e09fa54753bd6369bc3fbc4b54 : SUCCESS in 15m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b614c5b04d0f40368bb8616e6765fc12 : SUCCESS in 1h 02m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/679c9cd44cf9481d903a335f7344ff1c : SUCCESS in 1h 17m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3e8376a177a0d216ccf5b038b838e50d5c11d930","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 18:34:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0b4f747c105d3e5ef6a49a428b25fe593921c026","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 18:35:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efc61b7f370d4219868533555aa717fd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36cfc7100d094438984e1b80c6d87de3 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc4fa04d9b340f445d2fdc5326b1eb6186b486e6","revisions":{"fc4fa04d9b340f445d2fdc5326b1eb6186b486e6":{"kind":"REWORK","_number":1,"created":"2024-11-20 10:31:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/935759/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/935759/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/935759/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/935759/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/935759/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/935759/1"}}},"commit":{"parents":[{"commit":"9e67c176ab1e2aa55b0c8bde0fb94121edd8907b","subject":"Merge \"update constraint for oslo.serialization to new release 5.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e67c176ab1e2aa55b0c8bde0fb94121edd8907b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-20 10:31:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-20 10:31:41.000000000","tz":0},"subject":"update constraint for cliff to new release 4.8.0","message":"update constraint for cliff to new release 4.8.0\n\nmeta: version: 4.8.0\nmeta: team: OpenStackSDK\nmeta: diff-start: -\nmeta: series: epoxy\nmeta: branch: master\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Iaaf36d379d568ae41d84f33b6938179fcfbc1e35\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nChange-Id: I59c46c35d4435e0ff15cdf7b6a13ba9507c094ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc4fa04d9b340f445d2fdc5326b1eb6186b486e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc4fa04d9b340f445d2fdc5326b1eb6186b486e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
