)]}'
{"id":"openstack%2Frequirements~936076","triplet_id":"openstack%2Frequirements~master~I6a588ecc014b3a46274835dd2cac465b82c62714","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-11-24 04:07:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6a588ecc014b3a46274835dd2cac465b82c62714","subject":"Updated from generate-constraints","status":"MERGED","created":"2024-11-23 08:12:43.000000000","updated":"2024-11-24 04:08:05.000000000","submitted":"2024-11-24 04:07:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936076-openstack/requirements/constraints","meta_rev_id":"ea1f738324a4fef84a964f0d2e9cfa501f26bd3a","_number":936076,"virtual_id_number":936076,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-24 04:07:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2024-11-24 02:23:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2024-11-24 02:23:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-23 09:21:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-24 02:23:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"30ea9d9f0fdab1b9701f05cf83b00a0a4b82db1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-11-23 08:12:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c86efe7b0e2dbd134d492e5b64ef7860c93d56ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 09:21:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed1e2ef586064d7ab2f5bea456a77ba9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/553f748b37df451b86831995f8920860 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1673f36f5bb45738dca738f33e1a195 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95279d9f68c84e11835d8b5f2f2e4ab3 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d6efa812fc4b49ccba7ea137f6128d23 : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f2d2e6627db44c4caa77a87c2a1459bc : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5433d206a4ce4f02b02a7e87185cc0f0 : SUCCESS in 2m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7c1c298cee5e42108abcc0380bdc7d38 : FAILURE in 3m 34s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/21021c1573fa45a99052587203d1c430 : SUCCESS in 11m 23s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/c7e4c103853d47c99af9fcbcb8a4c345 : SUCCESS in 18m 02s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/5e94aa3ed6644740b8428c0e1379bfd5 : SUCCESS in 23m 18s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/ed307f9aa3e6425f9794d209527d0ace : SUCCESS in 6m 03s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/a650a6987a424a328b66f0a04c0b4f32 : SUCCESS in 6m 59s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/1e71675ab6604adfbf3ea36584ec7771 : SUCCESS in 5m 02s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/0e7cc3abf3fc4a3997abfb1130aa51be : SUCCESS in 13m 24s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/9a0f18fc412c4697915e2126ed9b0a56 : SUCCESS in 3m 32s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/872c031a1286423d9c375b974fae6edf : SUCCESS in 7m 09s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/5b1f6be858a8413e8c932f32babf892c : TIMED_OUT in 40m 58s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/d7c7948cd85e4f31914f4cac9bb29ce1 : FAILURE in 7m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/36a1b3ababb14105b4aea0ef173fa9a5 : SUCCESS in 6m 29s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/0f9b42b30fdb49c0b70e079c59ad7850 : FAILURE in 4m 42s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/e94c3346b0ed47f785938b84f2b9a17b : SUCCESS in 8m 15s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/46d109c796044796a59b67cf329b2fe0 : SUCCESS in 2m 28s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/bdfce6ad3faa4de0b4d892013f082004 : SUCCESS in 11m 08s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/bd73ecdd1b534ee6bf1e4d56de140171 : SUCCESS in 5m 02s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/76b350b2cac54a46ad964b4d3045f975 : FAILURE in 5m 40s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/d2f47e654dff43bea5cc129e434c1a9d : FAILURE in 14m 14s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/2c3c517cfb9e4cddb97501213ae9adea : SUCCESS in 17m 36s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/2f5ef260b7944483bb262bc2b28eb0e0 : SUCCESS in 8m 12s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/792ef03f03204b5a9a4b2bae747ec8bd : TIMED_OUT in 45m 42s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/7bb36404933c4089ad0da426bfa311f0 : FAILURE in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/86ebc324fb134cad80a5c8c2799c5c4e : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/716eaa9c45c9432285b19436f1cc5d67 : FAILURE in 25m 41s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a1862d0cae414b2196b0fe2c60577a03 : FAILURE in 5m 18s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/6e6985bbd96a491587d0fba4dab46822 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bf48eabd0b404d08b97391ae193926b4 : SUCCESS in 5m 28s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/56f94ddfb3514d6bb181d8922afe5c17 : FAILURE in 24m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/685f434320704b41943c064988b980e3 : SUCCESS in 58m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a43b68388f34b07b2eebf6fee419a6a : FAILURE in 22m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2e53997d9a5ac8b7ab4724f64a85fea2f3c4a43e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-23 10:41:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5b55cd587eec918782514fa0123879960130eb41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 11:46:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b43b2508aac4b4799bbc2b92e054f52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eff06c05783e4e9e8948c5cc461a6a7c : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97216f63951249808ab18de31705ade6 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0db5fe993505426098774040848510be : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/134f5bbebcc64da39281e3ac464305e7 : SUCCESS in 3m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/12dd8b8b5133401cb3c9b05fcc63388d : SUCCESS in 2m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2975ba51aa1247698592270789dc850c : SUCCESS in 3m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3a66a18061ae4552afe5c4a95debd7ed : FAILURE in 2m 15s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d2451d751e894f788f344b442f124f1f : SUCCESS in 7m 44s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/880abbad31384ea49c93ccb54eb03195 : SUCCESS in 14m 11s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/279eeab50998491aad54d200e15f68c5 : SUCCESS in 22m 45s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/8db6fedb9a73434ba08034794f9bac0d : SUCCESS in 5m 38s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/67b379edd7e7452d8c4d906f01bdfc2d : SUCCESS in 5m 25s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/97ac98b646574ba082a155a7324718a9 : SUCCESS in 3m 24s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/6d9c4e30f8354251984c76b261bb4982 : SUCCESS in 9m 31s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/629f283109b94163883ea8aacd283a31 : SUCCESS in 5m 37s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/df3511dc66ec4a998f1b43fe194fe15e : SUCCESS in 6m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/3d52e356acbc4800a69aedb544f56bfd : TIMED_OUT in 40m 41s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/2d4db7c2cff245ca902fa986aaec061c : FAILURE in 8m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8150d9753cdc45559f97eaaefb0a4e94 : SUCCESS in 7m 03s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/c2c29be236b64e15805853bb48e43cf2 : FAILURE in 4m 10s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/cd8649b2f85f4123a517e3f0f5902e64 : SUCCESS in 12m 57s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/1023d6223ca24ffa91a6c28c9abae792 : SUCCESS in 2m 05s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/4c1ce3f56de444459ff5605dfd090822 : SUCCESS in 8m 14s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/85d031c1b89d4a99b6c322d70047f41e : SUCCESS in 4m 38s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/d6d7763260b2452d83aefc2ee633b999 : FAILURE in 4m 52s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/27793696f80643b286662603ff00b21b : FAILURE in 12m 49s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/3183e6f57b604affa4cfdfc4b5820dc2 : SUCCESS in 16m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/79711f7a64b342938f7a5f09dfe365f6 : SUCCESS in 6m 55s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/73591d9a631f46c984ae8ad1e96bf552 : FAILURE in 14m 05s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/b58955ecea7e4a72af2bb9b50dc36b10 : FAILURE in 5m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ac5fdf94ffb047d2b73ce828a7a78d62 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f985e6aef55e460e9704e4c526dbea5b : FAILURE in 22m 50s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/ecd7c6b6aa454e6b8253a597da83a650 : FAILURE in 3m 40s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/35cbb3157bf244b4864688cb52a4609c : SUCCESS in 6m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c5c694eb1fb94c288d4629e9fa639c6e : SUCCESS in 5m 13s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/460a3f9575a54861aed819b86a04a7ab : FAILURE in 21m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30c2eb7debe84e1ebe9c6221ff5f7f94 : SUCCESS in 1h 01m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc8a443befb94f40bf2f6e5251d10a98 : FAILURE in 13m 38s","accounts_in_message":[],"_revision_number":2},{"id":"a5ca576f968a066e802b97d029385404ea5c2f7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-23 17:52:18.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5a83c2609ba3550597c7c938119ac5aeaabff753","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 19:12:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28f7142162564142a6664a09d35a481d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86383f2954bd4bcc85b1a500a1a6fb02 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33560ac4c3fc4f749651e5231f860c83 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d9b47cbd12f4d23aabe0ff2f7b50529 : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f3ddcc1aa0f47dc85389a2be3a8e542 : SUCCESS in 3m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f811e65a39f74183bcc5b36d5cee7db3 : SUCCESS in 2m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a1bd3702475743dda01c8677795e94c6 : SUCCESS in 2m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1eb358f59a8b41a28f0c10fbdb2db317 : SUCCESS in 14m 48s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/9de1a7fcf8224e5d9c9c496fc3957a33 : SUCCESS in 10m 42s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/a001360d8d144f829899870d908f21ed : SUCCESS in 17m 01s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/ec591d0d45e0457287a7e5fc7a86b463 : SUCCESS in 10m 33s\n- cross-aodh-py312 https://zuul.opendev.org/t/openstack/build/f7eadecd7c57444483838019a65516be : SUCCESS in 4m 35s\n- cross-barbican-py312 https://zuul.opendev.org/t/openstack/build/b65d454c702a4a33bb2441c6a1311f67 : SUCCESS in 4m 55s\n- cross-ceilometer-py312 https://zuul.opendev.org/t/openstack/build/a359a33c4c0d42a0818b9496c81afe2c : SUCCESS in 3m 31s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/dd4cda1bdf794de1840a6e496eb4f829 : SUCCESS in 11m 20s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/d7534b718d7c43dab28f9d382d0b0306 : SUCCESS in 5m 09s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/8880456b812d497299be7df956d13589 : SUCCESS in 3m 30s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/005b81071b1e4c5fa75325b7a6935f72 : SUCCESS in 7m 03s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/6b08dc4787714c988f8c89da56bf02ad : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9e2d3caec4c648eb9af458ce70f0285e : SUCCESS in 4m 29s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/18a0723232c14728b4cb54ec709e742a : SUCCESS in 5m 52s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/c89266f7c885494b828b1474e52b7bae : SUCCESS in 9m 30s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/39b1ae1305b84504b5533a43718849cd : SUCCESS in 3m 01s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/39f9b91ccaf6481a84bacdb4578b98bc : SUCCESS in 10m 46s\n- cross-magnum-py312 https://zuul.opendev.org/t/openstack/build/137afc78048744f78bccfb575f82574c : SUCCESS in 4m 42s\n- cross-masakari-py312 https://zuul.opendev.org/t/openstack/build/bec52410b1554bc0a0d154790760e3e5 : SUCCESS in 6m 07s (non-voting)\n- cross-mistral-py312 https://zuul.opendev.org/t/openstack/build/4cedb90c23fc481c988c438f6b048bfb : SUCCESS in 11m 16s (non-voting)\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/1ff08b2627a64073b9fc5d74b6422f10 : SUCCESS in 18m 14s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/da63f963699f499eb41ca4ab068e4478 : SUCCESS in 7m 54s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/092a6e2ccda24d8f8b2f2145aec789ca : SUCCESS in 14m 35s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/9dcac5b0c49f4794974a0017267deb55 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7c82ee6f9d5a45f38215c0a7fafb6532 : SUCCESS in 2m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4bde264823c84e2dbd6a476b017b04a8 : SUCCESS in 20m 58s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/c2c45037f08e4230bf101c458297a410 : SUCCESS in 5m 34s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/5b4f732bb7e348a7af105a31ff41abb4 : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a2cc54288c1242078ab10dcf332e2457 : SUCCESS in 3m 22s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/c059444a53e94d2eaf0d7973ba323d49 : SUCCESS in 12m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9617ec4035524657822ec1245e43cd42 : SUCCESS in 42m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f1237c5ef6f4cfb96cbd962d7041271 : SUCCESS in 1h 18m 48s","accounts_in_message":[],"_revision_number":3},{"id":"bf45bee82e01c2685c5513fb962498da0d89fc56","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2024-11-24 02:23:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c0565ca4a8c2ee866ad3a0aa4b63d98698213cfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-24 02:24:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0b44074da6ac10f4810de10cd2757138ed5a76c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-24 04:07:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e07aa4a396d3458793b5a65f4f205a01\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eccfa8fe8354425fb5ead03a079ea35f : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06d0b374f4d944f49c392b65e76845d1 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a73daffc45764398978c55ce557aa08c : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b177bfcfbe4e441888abae43610d7d38 : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0cd1618e9982493d91cc24f21a27d787 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e61e455e7ecc41c29324ff1d7c9f55af : SUCCESS in 4m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/11d32b535b974894b18207179490609a : SUCCESS in 17m 03s\n- requirements-tox-py310-check-uc https://zuul.opendev.org/t/openstack/build/d533379f9b2c4428afdf1d465faa6ff0 : SUCCESS in 5m 54s\n- requirements-tox-py311-check-uc https://zuul.opendev.org/t/openstack/build/154d5ea0343243cda5ad0dcf81fdac06 : SUCCESS in 17m 33s\n- requirements-tox-py312-check-uc https://zuul.opendev.org/t/openstack/build/6952fd075cf94cd489824be156be7fd7 : SUCCESS in 25m 50s\n- cross-cinder-py312 https://zuul.opendev.org/t/openstack/build/7c9f9e7a85b7416592e646141a8dd66c : SUCCESS in 12m 35s\n- cross-designate-py312 https://zuul.opendev.org/t/openstack/build/a657208fcb544bddb83406554bbc1bfd : SUCCESS in 6m 18s\n- cross-glance-py312 https://zuul.opendev.org/t/openstack/build/c651d58e1a2f4eae81fdb59c1d023068 : SUCCESS in 6m 26s\n- cross-heat-py312 https://zuul.opendev.org/t/openstack/build/f794eac6626944d5915c6935e56bd921 : SUCCESS in 9m 32s\n- cross-horizon-py312 https://zuul.opendev.org/t/openstack/build/388ef62bc4ae4ceba888e83754b88796 : SUCCESS in 13m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/520f44cf87974d6eb71810224a7acebf : SUCCESS in 8m 15s\n- cross-ironic-py312 https://zuul.opendev.org/t/openstack/build/b160bb63f787479aaf4989458b2de6b6 : SUCCESS in 7m 47s\n- cross-keystone-py312 https://zuul.opendev.org/t/openstack/build/d8f0af866e1145f28255420a26411860 : SUCCESS in 11m 35s\n- cross-kuryr-py312 https://zuul.opendev.org/t/openstack/build/a89df26ccb5c4df68aa71c530c96faa1 : SUCCESS in 3m 44s\n- cross-manila-py312 https://zuul.opendev.org/t/openstack/build/97f1b986a32c4b1697fc14fcdf825e16 : SUCCESS in 11m 13s\n- cross-neutron-py312 https://zuul.opendev.org/t/openstack/build/974d92b1f6694720aba8c688596117be : SUCCESS in 19m 19s\n- cross-nova-pep8 https://zuul.opendev.org/t/openstack/build/b0e5ab4ee39b43f5ba2435b5cee0d157 : SUCCESS in 10m 58s\n- cross-nova-py312 https://zuul.opendev.org/t/openstack/build/ec1c356b9e06455995a30eb90bde52a5 : SUCCESS in 14m 58s\n- cross-placement-py312 https://zuul.opendev.org/t/openstack/build/1e0524ab9b394edbb3892cb7f6b7ba9f : SUCCESS in 6m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9de0f44219bd4a59b89050d37a1d439f : SUCCESS in 5m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aca6343b9f104e6ea46d5b3ba860c4f4 : SUCCESS in 26m 02s\n- cross-placement-functional-py312 https://zuul.opendev.org/t/openstack/build/a6401004dc254ed6a9e704f7e817698a : SUCCESS in 7m 03s\n- cross-octavia-py312 https://zuul.opendev.org/t/openstack/build/0933c088c6824f51bd27cecf894d700e : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/652d6932e91048bfa2bc4bd34f6e3eb9 : SUCCESS in 6m 10s\n- cross-swift-py312 https://zuul.opendev.org/t/openstack/build/3c4fa34f69864e2b8ccdc0f9a4a682f1 : SUCCESS in 20m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b956b3db14d840cf96499932a3941744 : SUCCESS in 57m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c0258e3b13a470faab062ad09c31f2a : SUCCESS in 1h 31m 15s","accounts_in_message":[],"_revision_number":3},{"id":"3826e183fd9909f54b7b166e992ae0ef6a74c003","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-24 04:07:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ea1f738324a4fef84a964f0d2e9cfa501f26bd3a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-24 04:08:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9508750f7a58481e87382f7696263862\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82e26bc62d234e77b379b424101397ff : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"52d27d85156ae6f0b3b4a4cfa0aec2cf66367f78","revisions":{"d4aadcea94173c8097c4dadf50564ee850d67ae7":{"kind":"REWORK","_number":1,"created":"2024-11-23 08:12:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/76/936076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/936076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/936076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/936076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/936076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/936076/1"}}},"commit":{"parents":[{"commit":"e18c029d1ccbe9d3225bc8fa1c35bebdec77f900","subject":"Merge \"update constraint for oslo.db to new release 17.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e18c029d1ccbe9d3225bc8fa1c35bebdec77f900"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-23 08:12:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-23 08:12:42.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I6a588ecc014b3a46274835dd2cac465b82c62714\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d4aadcea94173c8097c4dadf50564ee850d67ae7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d4aadcea94173c8097c4dadf50564ee850d67ae7"}]},"branch":"refs/heads/master"},"9dad7b3b204417aef0c6475aaa99dbf2aee25acb":{"kind":"REWORK","_number":2,"created":"2024-11-23 10:41:50.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/76/936076/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/936076/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/936076/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/936076/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/936076/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/936076/2"}}},"commit":{"parents":[{"commit":"db314ff5e170a79f39fd44cbccce69afcfbd75ac","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/db314ff5e170a79f39fd44cbccce69afcfbd75ac"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-23 08:12:42.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-23 10:35:06.000000000","tz":60},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I6a588ecc014b3a46274835dd2cac465b82c62714\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dad7b3b204417aef0c6475aaa99dbf2aee25acb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dad7b3b204417aef0c6475aaa99dbf2aee25acb"}]},"branch":"refs/heads/master"},"52d27d85156ae6f0b3b4a4cfa0aec2cf66367f78":{"kind":"REWORK","_number":3,"created":"2024-11-23 17:52:18.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/76/936076/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/936076/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/936076/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/936076/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/936076/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/936076/3"}}},"commit":{"parents":[{"commit":"db314ff5e170a79f39fd44cbccce69afcfbd75ac","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/db314ff5e170a79f39fd44cbccce69afcfbd75ac"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-11-23 08:12:42.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-23 17:51:04.000000000","tz":60},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nAmended with these fixes:\n\nKeep version specific pins for:\n\n- alabaster\n- Pint\n- networkx\n- scipy\n- async-timeout\n- exceptiongroup\n- importlib-metadata\n- numpy\n- Sphinx\n- tomli\n\nPin versions for packages that are not ready for updating:\n\n- pytest\n- croniter\n- netaddr\n- eventlet\n- cryptography\n- pysnmp-lextudio\n- keystoneauth1\n- jsonschema\n\nChange-Id: I6a588ecc014b3a46274835dd2cac465b82c62714\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52d27d85156ae6f0b3b4a4cfa0aec2cf66367f78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52d27d85156ae6f0b3b4a4cfa0aec2cf66367f78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
