)]}'
{"id":"openstack%2Frequirements~936090","triplet_id":"openstack%2Frequirements~master~Ia9cc10fed844c06e83cb33f58246e13952a21e3c","project":"openstack/requirements","branch":"master","topic":"migrate-to-noble","hashtags":[],"change_id":"Ia9cc10fed844c06e83cb33f58246e13952a21e3c","subject":"WIP: fix requirements-check-self job for noble","status":"ABANDONED","created":"2024-11-23 18:25:12.000000000","updated":"2024-11-28 19:52:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2b0ecb013050e1e04d98ebdf61a77ad1ee75178c","_number":936090,"virtual_id_number":936090,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-11-23 19:47:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-23 19:47:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8ec5afc0317d11b30c23b32e42d810991e2fb6b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-23 18:25:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47ff2878a32de8e8d3fc8a3bdb614ffd5d1632c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 19:47:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58abd767b6994768bb538d9aae3d0b22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b72ad53cb214b2595c0699ab4f9236b : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04b6d197e78c41fe8b376d67724247c3 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1802a48baf084563bbe26e3578655f96 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a89df5e10c8f4ee78bf395d0e914a1b8 : SUCCESS in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/328e5ae441d74344893088a908f6657c : SUCCESS in 2m 53s\n- requirements-check-self https://zuul.opendev.org/t/openstack/build/ce53298d5aa240e7bbbe3a15c3ca31f4 : SUCCESS in 3m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e926025f9414cc68c8ddcc657002c7a : SUCCESS in 46m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ae4bc0d1fa54490be1e67c3327ac453 : SUCCESS in 1h 20m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2b0ecb013050e1e04d98ebdf61a77ad1ee75178c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-28 19:52:40.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6791339c3c75c44767a6143d7c7be96105093c50","revisions":{"6791339c3c75c44767a6143d7c7be96105093c50":{"kind":"REWORK","_number":1,"created":"2024-11-23 18:25:12.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/90/936090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/936090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/936090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/936090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/936090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/936090/1"}}},"commit":{"parents":[{"commit":"181c93140dd53fb13516668cf5da012e1fcd6c55","subject":"Move requirement-check job to ubuntu noble","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/181c93140dd53fb13516668cf5da012e1fcd6c55"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-23 18:24:13.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-11-23 18:24:13.000000000","tz":60},"subject":"WIP: fix requirements-check-self job for noble","message":"WIP: fix requirements-check-self job for noble\n\nChange-Id: Ia9cc10fed844c06e83cb33f58246e13952a21e3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6791339c3c75c44767a6143d7c7be96105093c50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6791339c3c75c44767a6143d7c7be96105093c50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
